)]}'
{"id":"openstack%2Frequirements~892927","triplet_id":"openstack%2Frequirements~master~I5cf86540d71f91f2b40620500992cb94ed8a147c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5cf86540d71f91f2b40620500992cb94ed8a147c","subject":"update constraint for oslo.metrics to new release 0.7.0","status":"MERGED","created":"2023-08-28 13:05:51.000000000","updated":"2023-08-29 12:23:19.000000000","submitted":"2023-08-28 19:56:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892927-new-release","meta_rev_id":"92071b8d5ab8c3b5bd4a3b5277907e65b3a21940","_number":892927,"virtual_id_number":892927,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 19:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:23:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 15:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:23:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"0ec00861cf03ca6c8937ca18190f4a22ca08a615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:05:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eee67cb520b77d2801865b7f344511031e50d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:28:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2d419fee9bd40288d955ee1e3fd4409\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8e264188694a308755bfab3b8e4515 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fe769c44f27411daf721e07ea799778 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cc3b8331cbb4f1f815f4af0700a8a4c : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d15d933af5c49bdad0e32e471dfe7b6 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/799657076d6e4f709f31b62ffdaa95f5 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/004fd7f73a3649dc8da751d1c8b5370d : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/125e9b1120da47a58701ba56db081f21 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/038b6dfb05664ca3bd98bb77e86fc3c2 : SUCCESS in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e85c22a7d58c4a3bb452487295b6c6fd : SUCCESS in 8m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fbb5df79171a487ab8d4f894c516b6bb : SUCCESS in 12m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/30b915bb0268419ca55c4b27a5b1b535 : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cca6b6798bed409a977d6534cc9146ea : SUCCESS in 11m 59s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3cd5e89af26743198e6ceca3b558469d : SUCCESS in 6m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d2aab770eb704b8caea726111dc89f4f : SUCCESS in 5m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/164537aa527d45db8e5b26ba1639b384 : SUCCESS in 5m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/550622f6f705402881bb945bb1461199 : SUCCESS in 12m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/615f6157122e437abd3c1369b204f345 : SUCCESS in 4m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/73336f52ef1a49f291a1d8e908d6b271 : SUCCESS in 5m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/547e8bfc7271486abceb61b2561e8b95 : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fa31024228174c68b994a811768d4fad : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f018d7aec64a4a659322a20297281d73 : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6d4db29e52074486ad3e3d1d0a0fd7af : SUCCESS in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cf7114f9215844d482026d8cef3e113b : SUCCESS in 12m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/635c54ccdfbb425da61564a08ed74fe3 : SUCCESS in 4m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/10eb2a27f1494d7f88f097d64eed7684 : SUCCESS in 9m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/10082ec3d2b74c768d590482018b23cc : SUCCESS in 4m 55s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6dda7dbcca0045f6bed7721af97387d5 : SUCCESS in 10m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8f6f56af8569452d9a1da7761a3bee76 : SUCCESS in 30m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b7998871be244a90852900a47b58f8c3 : SUCCESS in 14m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0e0c20dd654b4181ab80f7c722fd2d5c : SUCCESS in 17m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3af5f478b52749b6873c693aa420c97c : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2906254de1b64a99b1807d7061da1977 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d533c9e6bc1649648aae634386846a72 : SUCCESS in 25m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b643f6ac1a964b8b9194aad25565efe3 : SUCCESS in 6m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7b18acd9f6fe40a6a83f8743361f0a88 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0d0916737b241fc928c6f178b55f092 : SUCCESS in 7m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f99d2c7002b34a4d92e4b081f31265a7 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec2545aaf1b24dd78d0653e959421615 : SUCCESS in 52m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/517877d82dc14c4d8333807994c30bf1 : SUCCESS in 1h 51m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a689df39e640b4d509e09b74e5cf3c200be73fc2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8af270c0bb09fb81d6b36f38837c0cf34b3411a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:36:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4025a4f6ab5212d2701a81bb201c4a504f5e0559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:56:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7089b1bb1add43b5935a5efa713f7862\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a764f41a18be480eaa009b1956fc09e4 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbf25b50d11f4bf9918e7d8edeffb4d7 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4a6f93d4a954bb5b334917489a86a02 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32e27054215f4c1fbeacc3331bddbd2d : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7936842b1e7f41c5a6b59307989f3498 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ce31bf1fb3946d3b1961805e14154ff : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7c0a1c963c84d5fa16ac1b67e224b3f : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf52911812f6441aa442fc1f930f07c6 : SUCCESS in 9m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d60d8c1fe1ec4202898580642d6704dd : SUCCESS in 12m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/34a805b7ede343a4a6cd364c8208aa2b : SUCCESS in 8m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/359173f040ea4426ba405a77ad1767ba : SUCCESS in 10m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/78fd931339e347c3872c5c05622adee7 : SUCCESS in 5m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7405f8b92c3e4cc49522e643b015379b : SUCCESS in 5m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/953df5c0d198409abad750a96dd3257c : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3073464d26d54b5285bd8d93c8ba0264 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e2e249d09a940d682fd7174824ac9f0 : SUCCESS in 7m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b84d8899bb48440aaa745d0fd78e17cc : SUCCESS in 9m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0493481c77f148a5a0a5df2b4b55f701 : SUCCESS in 12m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/417c6f2b662f4716a29375b373cbe57b : SUCCESS in 4m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a9a82316fc634a6691875153b327e152 : SUCCESS in 9m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/53851036db1d41a6b7e52f9e7b64f38b : SUCCESS in 4m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e5b26e81876544c8a3bc23e093da57a2 : SUCCESS in 27m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/52c6b73e823f4d14967ca0bf525a6fd8 : SUCCESS in 12m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/eb95a76804ff4d68aa2412976cc00f28 : SUCCESS in 18m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5769d36f187f4a048716d9566e850ea3 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f647746f01024c86bf4a3de4a351e7e5 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8c111f6ec9543468baa2d363cf31a1b : SUCCESS in 23m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/769948a1d6114da9b7338df36aa7ac23 : SUCCESS in 6m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/934c2c83fcfe4a98a83ac8ad89956e0c : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e59fd44314e42ebbd20757b27f36cea : SUCCESS in 7m 22s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e1e163589eba4475a5bb80766fc6ed7e : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfd5bc3e417b4ff3bf13a08d4e3881bd : SUCCESS in 1h 04m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cc59569b74a4e09867d43cf5614b72c : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"770fd0807e308cf2ebb0b687cb25fc79864aef34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:56:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8dd03cd6d794eb660923f4d06c951e635eb1269","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:57:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c4fcbcb99d4d3290d9808704f92b25\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75bb48b88a504768a46f68d48dcebd6e : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1},{"id":"92071b8d5ab8c3b5bd4a3b5277907e65b3a21940","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:23:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"410c1ddfcb868e26ccc41e2c29fa29bc1c1c95bf","revisions":{"410c1ddfcb868e26ccc41e2c29fa29bc1c1c95bf":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:05:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/892927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/892927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/892927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/892927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/892927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/892927/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:05:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:05:49.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.7.0","message":"update constraint for oslo.metrics to new release 0.7.0\n\nmeta: version: 0.7.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I64551cbf840d4e9fdaf2675d95aad3e5b51ba5aa\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5cf86540d71f91f2b40620500992cb94ed8a147c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/410c1ddfcb868e26ccc41e2c29fa29bc1c1c95bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/410c1ddfcb868e26ccc41e2c29fa29bc1c1c95bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
