)]}'
{"id":"openstack%2Frequirements~892928","triplet_id":"openstack%2Frequirements~master~I104eab3392db41fc12eb947a6568994bc9a59863","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-09-22 19:27:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I104eab3392db41fc12eb947a6568994bc9a59863","subject":"update constraint for oslo.log to new release 5.3.0","status":"MERGED","created":"2023-08-28 13:06:23.000000000","updated":"2023-09-22 19:28:52.000000000","submitted":"2023-09-22 19:27:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892928-new-release","meta_rev_id":"3982a55841969733c6b1f665e4825f88f7555878","_number":892928,"virtual_id_number":892928,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 19:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-22 17:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2023-09-11 09:32:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2023-09-12 06:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-29 04:50:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-22 17:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-08-29 04:50:15.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-08-29 12:23:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-09-02 01:55:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-11 09:32:11.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-09-11 13:29:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-09-12 06:59:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"501172db32b4db6c32fe7ee340b09295c45db410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e3affd23d2bb4eceef02b10b3500f0485bbbb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:51:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/455d673466a245609ce086698665cd21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93abd93d4194072b271b3241f710e62 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/863cdc1758e54355b08308fee6f4805d : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd423b9d4af94c6798eaa94c8aad3c45 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae828381863e437499cd1ae4a405f639 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa4e17014fa043d0b0dfde48d90f218b : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c2435c10e6b4f62b820306c2cd2e948 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1d0b60bf8df43d19b07820cb064a865 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/404f60051ded4da8bae42c54d31a8844 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cd9053aeba964adda2a20d3199dcb4a9 : SUCCESS in 8m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2b6d8a248f39464da97a6fab4e85ff28 : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/48e8042f98c144c290344ffdd9d51d04 : SUCCESS in 9m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6c0d493571104caab3b95a1658c13462 : SUCCESS in 11m 43s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c1c075bad74849a1869b6371db8c11ac : SUCCESS in 7m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bace6460deae44acbebc68f43bcf3c90 : SUCCESS in 6m 02s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/783d89d63d1446f99768cf0666006f36 : SUCCESS in 6m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9e45e2b686004d88b1fb4fa7aaea390c : SUCCESS in 11m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a0614bfa4c8e4f6995faf4669a13060e : SUCCESS in 5m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/657d7beae7494659a005501714c6365b : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/611cae3b6bb64f17bf5c3ba72a604d46 : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2e76cf890bbf4839950564147a6d4238 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08169e4d3fd64001918a0573e5579329 : SUCCESS in 8m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5ac2cef8f6bc43938fb729cb17495bb2 : SUCCESS in 9m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f310910db1e245bfb20505386a7ad732 : FAILURE in 10m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4cfaf82c3c1444309cd40da0078b4ad1 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4406a46373344267a44abd3b0ebea749 : SUCCESS in 10m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f518e01634d540209d452b28e8823ec3 : SUCCESS in 5m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1fb63bc808164a3d8821ae60d56d108a : FAILURE in 13m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b96e7f91b26848ebae3e6562cc4ac559 : SUCCESS in 28m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/77a21cc8397e4f298d30cdc5af706411 : SUCCESS in 12m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/73cabf45e96f46928c51ada4962c6c78 : FAILURE in 16m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/25d42f44a49a45db942bbc9fa5209f24 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee9d451a843142799d360ac423102039 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97d4947b448c4348a6c797563a5b8fe6 : SUCCESS in 24m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c664eafd471740fb8c06202eeca192ad : SUCCESS in 6m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0b2f04a0fe264190a172d9648b7e5905 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32685a5d95294e339d6a7d3bd92ee83e : SUCCESS in 6m 55s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6694231d648448c69e36dd2c9fb8e1a9 : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82bba22d9e6848f69fb14c8a82dd8eda : SUCCESS in 1h 04m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7092951faac24f968ac0eaaf0b0a1efb : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e823ab738f4f4318b25d2f463e558fb0b557f591","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c89b6206d643f32a27a3d4e68b105d609491f5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:35:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/832ff8c601c04460a650b505ae8cc983\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08bf6483d7514c0e80eed87beb4ea658 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0eb46a977684f16a9b2a1570e7b14df : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb7fb07c57d44e12a951e9cb467305b3 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3c7d5d6ea0c4fa8bb0df928cebd35b4 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3699113dfcc24e0ead503a60fb2efabd : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f78af395939c4fbe9a57a042a74e9a9c : RETRY_LIMIT in 2m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d9f3d3d33bd4deebb8033715ca9fbd5 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3be6c2ac5638468fa4847a730c25757b : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf30f35b2d014c8da84850e4fffeffde : SUCCESS in 9m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ad7859b2f1904a75a9021ff1c7ebd0f4 : SUCCESS in 12m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/845f50e2e22d49d094ceff54a819e6b5 : SUCCESS in 8m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/062ec8e2fb364eb29dc7a51d9b71a722 : SUCCESS in 12m 17s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/75df4d2e6e09451d9f8012899548899b : SUCCESS in 8m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bae9c41d71b944a49214692c121e14c8 : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ac1286f624ea4654b8fccb34f6a42412 : SUCCESS in 5m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cefce5d0f86746119c6a0ad049b01860 : SUCCESS in 12m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/805962bfcb2f4c1d9f6af668931e0015 : SUCCESS in 5m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/924332a6064d428e9cc1e06b2511732b : SUCCESS in 5m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/52c94fd572df4292a3a49898ce38a12f : SUCCESS in 8m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a83ab4369e6840adba6f5dc932d3ca07 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f14921b34717495aa07ea634bace2dc4 : SUCCESS in 10m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d27d2424dd8b4b469b36250d4548668e : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cd36023facdf41d5b08c8d26f1833363 : SUCCESS in 12m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a962f3871ab84709981033f61e5feb8c : SUCCESS in 3m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cb6f8776bf37405abbe17098af2e116a : SUCCESS in 11m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/885918a1c5344a52b662341810611ff4 : SUCCESS in 4m 46s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f617522d397b4ec78a48dcd9677cb9da : SUCCESS in 11m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/588b2110d3e54f189d3aa516875dbb83 : SUCCESS in 28m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d055b67ce5df4bfca0a91b4160524466 : SUCCESS in 9m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8692930e08e4485d8e071afca3f0fbb2 : FAILURE in 15m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0134409b53b94264944c4a227badab75 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f9b237e951194ba4b1bb9cee566850f1 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57c3b453b5b64cb4ac87c2c4b472a229 : SUCCESS in 27m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0154d33ac57b48fcb5d751f6dcaa9359 : SUCCESS in 6m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cbdd0a57c91c4726912cba88c17df120 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f41af70f4ff4d23809ea07b1ada6b76 : SUCCESS in 8m 05s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/01085c623c6c4ec7b2b8197f54a4c325 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab0b6b6e621940daac1b96d5841cc146 : SUCCESS in 1h 13m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b33514b796bd41228d2e4d0b78b491a6 : SUCCESS in 1h 46m 50s","accounts_in_message":[],"_revision_number":1},{"id":"404d0f5c9eb67cc8b0580cae6b0f4f6982aca085","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-29 04:50:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2c1f403b990bfb393a041a7c0d14b153b648301","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:23:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54b1d1904d4643b83a0e309eccdac1b876bae850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:12:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6603bc67d7404a7098ae84be5fb96794\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f765a06f174442618215c23162da6c67 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0185aaec7c1b4cef9f28897a2d7c9215 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04c2bfc3758e41cd85745ca4b41bfb06 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0b851bc1fbc47418e285e0e58625288 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67d3de9405e1431e8bb496dc800b7121 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1db6eb3ca0bc4eca814a0f25f7d278a5 : RETRY_LIMIT in 1m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ef624367c9240d6aebe9de19a246fab : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d42b41ad240040768d2910510216d5b5 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2bce7b05fd4047c29389b2fa8ed23d94 : SUCCESS in 6m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6743209d07de4e01883d9163566d43b1 : SUCCESS in 10m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2744441db6ea4c84ac6e7b146755ecfd : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5f1325f8b02149ddb50810283f353613 : SUCCESS in 9m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fbb8980e8cac4d70b69f6a13670747cb : SUCCESS in 6m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/40cf934924a0473aa2c10d344fe5476a : SUCCESS in 8m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bdc666c083e24b0290d5ad5ee8cc3350 : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d1489e4df59241f69da78f7bdaaa8801 : SUCCESS in 12m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6f0aabc24da9415b840ee3735aab83ba : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/335a386dce4a4f5388d72278580d1dca : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e916e2b7da5a4b4eb090831c09d64dd0 : SUCCESS in 9m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c4444a73a962420db200ab88615da4ef : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb37a0db388d49f8adb03415156edecc : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fad57006fdcb44369c840f9bf63e6e4f : SUCCESS in 9m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/12d8dd14a7b245a2a2b3fe7e80135b85 : SUCCESS in 10m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/14fffa8929f64399b8063229365f7dc8 : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f8ad1ab45e7b43f999141a0eaf4dc5d6 : SUCCESS in 9m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c2b3cd7035484fc5a48b53418379ba0a : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c8df84c9d24541caad086a35052b3dc6 : SUCCESS in 11m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/02414c905c6c4fdeaec685dff804dcab : SUCCESS in 23m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c890661ece134a9e8caca5fa3e152ea7 : SUCCESS in 8m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9297295a51fb4eb38b44992cc272c333 : FAILURE in 14m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b2c2a1f7f9d040fe8983dcd90cb707af : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1985004a2d014033804e00324cb93d86 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d371be71cf984dfcb226e8cf1629379c : SUCCESS in 27m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/03a8678fe5c44622bdfad6286dd966d3 : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7ec4bc67186a4027942756f8b6d8c7a0 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e69b17b3388a49f9a5275f8dd30b5914 : SUCCESS in 6m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4d17435eb6914d8abfb9b317274af66a : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ad68a64e8964252b787920d774935af : SUCCESS in 1h 05m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7094d7bf450465d8e3417b09d399f39 : SUCCESS in 1h 40m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5947f1bc10850f40302b91d3b14b2b4387b5f06f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:32:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74671cf8c52175314fba76df319c35d507eb08e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:20:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4f0f3c53a3e47999be6ddb887ce22b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837befb44cb54c50b3439bf237d90686 : FAILURE in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5b9418c76245c992cc053a4976c071 : FAILURE in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee4cce2a9b2a4565b9de7331f0c29d74 : FAILURE in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce0b217af04b4e23b1f6d96e3ebb33bb : FAILURE in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a1167d8fff04764aef34e02e6b9d78d : FAILURE in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21b4be4c8de84b93928673d487e5a3ef : RETRY_LIMIT in 1m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/288c015fede74f908907cc7c6368152e : FAILURE in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ae5d7f385c543ecb5865ac94b6cf2dc : FAILURE in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a0d28190c12d4e0b822d93e70873a51d : FAILURE in 8m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d7db3e4181f045739340a028f7e940ff : FAILURE in 11m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a035b6790c5748b8a6f2dacef3d3fafb : FAILURE in 7m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66807f3aa7ad4ab7a6256f6f5311c8b2 : FAILURE in 12m 56s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fe32fef4d02e4be2acf56f8cc5ca947a : SUCCESS in 9m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/45c6e03580f24f49b701398994645d24 : SUCCESS in 4m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/16fab96dcf714257814efc531ea80176 : SUCCESS in 5m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8936ae96d7f849368de4a35abe61ef0b : SUCCESS in 11m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/eb0e762a50a14621812cb9c2ffa6c11a : SUCCESS in 5m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/02f9e2ef091542d9b94214be293412ba : SUCCESS in 6m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/16d564c67b344c7197d053e9e6df40b6 : SUCCESS in 8m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/24d1c134bd574f38af2db76b31ac0cb3 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6bf5053acb9459c9de6c921f57e4222 : SUCCESS in 7m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b5ec1efc190f4266b97f804d4285c02b : SUCCESS in 8m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8b4b3a6b30714e6c896bf5f2e5c9743f : SUCCESS in 12m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/187c000e152c43a99edd048c3acadc0e : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/adecae7a03ce440a8df8447aa708f55a : SUCCESS in 9m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e291bf88e58d4ccf965059134281a9eb : SUCCESS in 4m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c4bdc2d52752467cafb66a902d838ca3 : SUCCESS in 12m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/76f3b1fb4bcd4ac5b9d1d48b01065d22 : SUCCESS in 25m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/60877523046f4e6f96d058dab22e50f1 : SUCCESS in 11m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/06f36a2c1e34470bbf388c256772c2e0 : FAILURE in 14m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b72c4be653aa4adfb0513ac0a15b22bd : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a67c6b663ff64ec58973fc1460d852cc : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/009f591d9246454aa7b018acf9bd7809 : SUCCESS in 21m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f4d09a8744b9404b9be11cc5c23d8195 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/58763919188440aea07c29c03728d2b1 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15ae48f9f30147e39e193122a740fadd : SUCCESS in 6m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/84fd13c6dee14ea699399e6b5854fa8f : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce2ea8d41b464d9c89c38772d08c3c16 : SUCCESS in 1h 00m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f525e1d79bf9477ca4802bc1a182fa4b : SUCCESS in 1h 38m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d3254601bf327c1f21753a8bffef5896befed3b5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"402e6dac9b91a1547b47f51bd68e86cc1b5459d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:32:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/466a82a6054a4db7802f24ca60fe3c4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebf33087c3c4f10bfe5975c930f994f : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c111fbcd364d209c9c7c2bc5089056 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26bbc8570d8f4d33847563de0cccd240 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e2ab0b886894b4d924fdf2b1901d354 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89bbdc0e39c241dcb9604dffc7d94286 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/935d4a55ff864fa293a88805157214a1 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b6bf8a1d0ab459d91c52571543be94a : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b43685b86c174f74a06bc8acddfc7a8a : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cdbea2714f5e4f16bf433917818c985e : SUCCESS in 13m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a5b9605dba2d43b4940a7ec0427be39b : SUCCESS in 19m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3a76478e788d4cd2b42592d8f7411586 : SUCCESS in 11m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/051e5a47da09441395eb1cde58b10b86 : SUCCESS in 13m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ff613187eea9409ba37fa24793a75d2f : SUCCESS in 10m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9e6c1a7320f44a46815a39149e983d45 : SUCCESS in 5m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/87e9c9e4c2bd48c58d5669587e605b3d : SUCCESS in 6m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8799c66e232846039d43ed16c9002741 : SUCCESS in 12m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/756a4b42f6f3473a8c37ca63bb8f599d : SUCCESS in 7m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5393c7313581423ebfb990094e298c82 : SUCCESS in 5m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5ba97b6a94a74c32ba9f86e33c97a2e1 : SUCCESS in 9m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fdcdd14f56cf43a0952dd46468e62b68 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db58a6dbbf0147daba96bafd5b8785c0 : SUCCESS in 10m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f2a99e0999724360a1a4f8a800fb6865 : SUCCESS in 7m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/679c12c7073d4516978d9f4b66d20077 : SUCCESS in 15m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c2bf5e0cfaea450493c110e0ae30b3c4 : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4ff46177bf4245f196421aea8109c3da : SUCCESS in 10m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e2edca2d40294718b99024092b21d195 : SUCCESS in 6m 21s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cb7d20914adc474489dcb51d3eb0dd31 : SUCCESS in 10m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/edd4444517e84cbeac9c6075a2af36d9 : SUCCESS in 20m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/70a6bb71901d4ee7b366b1e6e1e38a78 : SUCCESS in 9m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/62d40e70455c4b10bee64571d51ce1a0 : FAILURE in 14m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/913a676b578544838a980e7394ff79a1 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f508656f580a4549b5a31d93c1c2d1a6 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a2c04e5ce1a4a3e86220a782e1a94bb : SUCCESS in 25m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/36a3023cc9634e9782d9d1023fbec338 : SUCCESS in 6m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/097eaff0634f472c9a3378b049770aa1 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1399339c35c64b67b5bf435b39cfcafe : SUCCESS in 7m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0318ea8afffb4a909d9ddbdcd0ac2298 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10c2f8ea57d5486da99e823d36b50a23 : SUCCESS in 1h 12m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5195d55682e443f0bc89efb1c7d93668 : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0eead14bed437c2237f1aa31988081d73e65ca2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:10.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"37122f9f0a35aa8b0a5b57f960ab91b83462906a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 19:23:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f373d2019b1c474392a554eca9d4e38c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad07c4d5ebe042ef9785d8960d334bb9 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6bb02776be0495aaa37171dd8c3f5de : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b681dc4a2b374972ba0730a6d02ddeb5 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbd18ccf173048f0a3b7ae7289537350 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30dd04f09582489bad2337e1905e7ef3 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de8413a33e42458e9cd0bdf6b89b6d88 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e522ebbf03f45518f3288d90eca22c8 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ad645ca8f904351a12f87a7f1259d15 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70b4f4b10cc64c7bba0adcea3566cbaf : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c08ea1c0ded34db19b32f7c235758517 : SUCCESS in 13m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6d25e1bef6624423a9cc058b662fc244 : SUCCESS in 9m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d1129149714f4a73b844dbd5fb218164 : SUCCESS in 11m 32s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b07fcb3e890a423bb26e134e0cf298af : FAILURE in 5m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/799e35a121d4439cab8e4ee91ecccc59 : SUCCESS in 6m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/46027069a2f44d058cb870e744c524e2 : SUCCESS in 4m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2687d67d88e24615aac22a8934032f33 : SUCCESS in 10m 53s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5071e9f045614f1f8280bd9ef0aa0869 : SUCCESS in 5m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/df628a3d9e4a4c14ae5fe2ba50310540 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b83fac40c69d4c9686f4458f63a589fe : SUCCESS in 7m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5524759acf11457fb2a3af6c0a4ed413 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ea7c855fb3749c09e9f9e6290ff232e : SUCCESS in 10m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/630e50490dec493abee108abd21f84e9 : SUCCESS in 7m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/aaf44ca799124476beb703b5eb1b3146 : SUCCESS in 10m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0a7260817cd449c987059d8ac52cf800 : SUCCESS in 4m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e905729ab6f243f586a1aa773ede3df7 : SUCCESS in 11m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9307b3d3e5224adf88071c953add11e4 : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d73bc4ee64404d6cb5b5d892db0f6086 : SUCCESS in 10m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/88bb3e1834c14c56bac62c5905105776 : SUCCESS in 26m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/73a13a5b90eb4167a350d20b418a459d : SUCCESS in 11m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9f158919c3a84f16b376478dc6797b53 : FAILURE in 15m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ba7af7acdfef426883f0a0775b26da90 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5bbd53e5a9c647fbadc1475b4ec7a262 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ff5524f6cd64ba081c5e974bc56790e : SUCCESS in 22m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aaf1601183f24fdc93f42671c0d6a707 : SUCCESS in 6m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fa28f3ec3f3c4ea69aecb08a03afc047 : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b9a315be1284a8aaa61126a0cf45be2 : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/645ee5daacf64fca83e7f03786f06e2e : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31d875a974b44bcdbd983afe8c21ba89 : SUCCESS in 58m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9d47da92be94fce855676c645b9123e : SUCCESS in 1h 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"058c34377d42cedb8bb56d987a755c7d45364dd7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-09-11 09:32:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a215a91d2fe9eb386c84618cb94c882ed88f6855","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-09-11 09:47:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96ef5a0756ca37d96266e4aaa571ae64a753854d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-11 13:29:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfd51832a23643cf88c355b8c6b6a0d892c3f2ce","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-11 17:28:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dd789351e56f86f0fd5d9e046a206d290bfd8b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:49:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ae639e3a9849b485aa861066025c21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f71d1794ca4920999784fcb8ef1d83 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5402f3c5222490285b353de59a3e8b1 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8df9b8ab764f4d8081eebab7f1aa6d47 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/437dc6c690b24f5a8ade6873406a8bfd : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea86d6e179d84824b885971ab6e6312c : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/096f402187e74df49301e62b316fbdd3 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c686237b10194ac5afc553096304a869 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d02b4f1f14b4408b640d34c9d58321d : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c6f76cffa8b450996e04108dd0313dc : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b8a7302dc0db4d889d516885abe0d15c : SUCCESS in 13m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d964a56d6481484e999001836fd9800f : SUCCESS in 7m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/110edf34b2ff48a781e0ce5e99d43152 : SUCCESS in 13m 31s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/446eed0f575045b1802dd64fb84f86c6 : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4fe3cc893b924fbbb4cf61823c38ed9d : SUCCESS in 5m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e0d1c0e7ef874a2289ba48810688d208 : SUCCESS in 4m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/78ce7f97d0f64222bfa8e2c8a34362d1 : SUCCESS in 12m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fa4f1ab298224ddc9c9c91e6557b4f4c : SUCCESS in 6m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/77c0f2a31fc247e7a837b271918e61df : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/69b881ba2761450790d6a70c226f1bd2 : SUCCESS in 8m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8cf6175800a1465ab35a988fd641dc2c : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b666b0385a314029a58ccabb6fa7ca01 : SUCCESS in 7m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/153baa221c514ddda97980c8c2cea442 : SUCCESS in 8m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a1057b4878f5439399bcea49e840abf4 : SUCCESS in 10m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/71237fcb36c7458cbe400596519f10b9 : SUCCESS in 4m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ff84007e2e1845fca71a720c3bfb0bbe : SUCCESS in 11m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c3c0d584f67247489f162aa283748d46 : SUCCESS in 4m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8e305f3c68a24930b13c548f0591a9db : SUCCESS in 10m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/70c15cb7a38247139293c84785f5b212 : SUCCESS in 21m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5b127eb5d7b64d60a419997cbd3168c0 : SUCCESS in 10m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e7fb5703afcc4f8880e3e865546b9ce2 : SUCCESS in 13m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0a9ba05d904f4560a0d3b63012c160ba : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4921d97621bb429cbd96034d7b0a44d5 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f74b95f3ee7490098ff74347ab875c0 : SUCCESS in 25m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fab93b9899604bb1bd555faf6b3fa079 : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c3561bd68b314c79b28adce3a757cbd5 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7413fcb73ef9442385a10d5dd32a59cd : SUCCESS in 7m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/07109ac751fa4754a2889670b83fa4fb : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fc7f7d245c54a068b7f9104846594a2 : SUCCESS in 57m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e95305336f440ca8fda7830d15fdb88 : SUCCESS in 1h 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d2465bdd88a5f79d8e6abdd857fad58c06e77f58","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-12 06:59:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbb3703b65237adcf30bde041bb99322b67bb97f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 17:36:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"974b85161336907b953f274e4b3081ef4778534d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:36:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24c367077e9037d161e5926c712512be6dc48b89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:27:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb844b155b04870bf75780dd6a4bcc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc06207510d44f479d87acaf085020a4 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528f18641a3648ae869958276fcf1f2d : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e31aa5c6d903498cb740edad9a7ef949 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05b6baf991254bdaa186c41336d42b0c : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0b54640211b41958bd6c78871ff145f : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/556a7d96e95141d98de7a2cc02de2993 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e824cd1fd8dd42798001c7fdf8c3f2fe : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f7fd1f6294a40a09e4dd1acb0f4f791 : SUCCESS in 8m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cb87f034e78b4746b4441d66f1d81fbb : SUCCESS in 11m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f10ff99de164409b69d9b67fd7ad779 : SUCCESS in 11m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/45b8d33345fa49c8b879ab372a2af4ad : SUCCESS in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/89b52d3977034ea8b41dfa828eecd451 : SUCCESS in 6m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/355857fc887d4f19893a62f880e752be : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/961369dfcd5f4c6e98169866770667e1 : SUCCESS in 9m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/41c72a32f34341eba03264e332e64416 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dccf3a8380047d1921d5de0df5b26ee : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5fd80901cc3c4d16996d1c2bcc2ee831 : SUCCESS in 7m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/58ecbb6ee5374b9d83bec6b7356c89cf : SUCCESS in 11m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7c33747ec1604291a5e66d8e8d592b31 : SUCCESS in 3m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/efebe11198f94565980eafe9aa20d55f : SUCCESS in 16m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e60fc26382c348d48521626cb839cf01 : SUCCESS in 4m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2aba590332bb45a582a44e0699a0b157 : SUCCESS in 21m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c6ef62310b254bc586d1efe936f4b4fd : SUCCESS in 12m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/137d9ea4a7364690b5a3ca77cd799ca5 : SUCCESS in 13m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/56d27975061b49c38ab11b8460718f48 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67121c22e84c4d25a4aadd8a37653aa9 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/039cffafdaee4902a51b0b6acfeaa20e : SUCCESS in 26m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/95a5f9fccca34d39b92aecc293287405 : SUCCESS in 6m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c28219ba51b243878bf0442b83c7f1af : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53e5463cd44b4452b1c5bd5b664e470f : SUCCESS in 7m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/aae8f2556f6e497d9691793721927326 : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2010fc8e55ef4a569c9b5664d1a82e00 : SUCCESS in 55m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c21519fa2584cb8bd1de6b386d29ebc : SUCCESS in 1h 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"395a38ba02e0a544190c76cd8e7e1ba470ed7915","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:27:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3982a55841969733c6b1f665e4825f88f7555878","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:28:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88054739544847608956a228ed8f1dfa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e594cbe9324a5c882da763ad327d11 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6125144ec67d93186ca04d82139be3da636515e5","revisions":{"6125144ec67d93186ca04d82139be3da636515e5":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:06:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/892928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/892928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/892928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/892928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/892928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/892928/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:06:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:06:22.000000000","tz":0},"subject":"update constraint for oslo.log to new release 5.3.0","message":"update constraint for oslo.log to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I568be1190433321d65e2ccdbfd4488d96966ea43\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I104eab3392db41fc12eb947a6568994bc9a59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6125144ec67d93186ca04d82139be3da636515e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6125144ec67d93186ca04d82139be3da636515e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
