)]}'
{"id":"openstack%2Frequirements~892929","triplet_id":"openstack%2Frequirements~master~Iabe715ca5b3139703a20fca595bfba1f9e59992f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iabe715ca5b3139703a20fca595bfba1f9e59992f","subject":"update constraint for oslo.service to new release 3.2.0","status":"MERGED","created":"2023-08-28 13:06:46.000000000","updated":"2023-08-29 12:22:54.000000000","submitted":"2023-08-28 20:06:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892929-new-release","meta_rev_id":"f71098d793ff33589b6c3af66a6286c74fc7d773","_number":892929,"virtual_id_number":892929,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 20:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:22:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:22:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"8e4a44b9b1bbc279caf3311afd42cd1b5ef26491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:06:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7384be2a705e5612d54391ee0d59431583108467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:54:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf7eaf1b52440f1a2607e6bc61afb10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0bce2d558f34a72966e2107b5d422e0 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6887f164dbd74bcc98eaed5372f6fdc1 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea98af7e716a4cf0971c641b864d4a7c : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b45d3b83cadd41f997d6940527bdf305 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16b354816bc649aa87e643c4120fff49 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bf401fa538b403bb6d6ecae0aa49cd0 : RETRY_LIMIT in 1m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8aa1eabd9c14472a4b686a32d7ddfe6 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/121bf914167c405c9af81c94fe6e5285 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d23184680bd84e8fab98c75664c95fa9 : SUCCESS in 9m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0a117d57ae7e4c93a3ca67fd8b732d84 : SUCCESS in 12m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b85cf34e77494c49b5803d88729470c4 : SUCCESS in 8m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2ec31be43a0044508bcc824582ca023f : SUCCESS in 12m 28s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7002e4f2bba140d49429bfa1c2b80093 : SUCCESS in 8m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/de30cef35ec24eabb982a53c17ae205b : SUCCESS in 5m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f0e7759d9fd94628a7e96f05c8c8c70d : SUCCESS in 5m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0f4d5e738eed4cab9ba7ed9dfe9ccd60 : SUCCESS in 14m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e307f00dc85a4139b2ef41cb0fd2a6c9 : SUCCESS in 5m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c014b446330f47dfb2225a81e5c6b2cf : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/50f09fff7c874f39a35797f3cb675efc : SUCCESS in 10m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/df1ebf13ebd34493affe0d4d0b404b8e : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/247af0115dce46a7998e021a7be9d2f1 : SUCCESS in 8m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/563fa5aa5e0e430eac0802ba34cac402 : SUCCESS in 8m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/11b2114b1a3b42e995569c3f9e30267d : SUCCESS in 15m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2b0217dfe6ec415386bf8d920c4a250a : SUCCESS in 5m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/050501ae0af746e391102b3e0a7a27e7 : SUCCESS in 10m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0626e51fd14049fe8824b12d2b68365c : SUCCESS in 5m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8f2bd9abaed14aee8f1738f174df2485 : SUCCESS in 12m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/11db0b347ce54628afed0e3b762a1787 : SUCCESS in 25m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6c889099d89a40c4b662a4d2e1a3bbd6 : SUCCESS in 9m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a2cae28c8cd341dbb40abf3625f452bd : SUCCESS in 16m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1ee26ec264c14002af085a8dd9a81abc : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc351be1edd3444ebc1a1583499c4b4d : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7d6f395eaee4de5a897c14612000baf : SUCCESS in 26m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/82ae4191c6ad46d1b4c57cbfafa5947d : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/62e68c2b0f6a4ad0a7ee5ccbe1c90584 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a19d891ea40b499fa461e5f97d521e34 : SUCCESS in 6m 41s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/88d8e75198114dd2bd994b4e865f2a69 : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24d04069882641ac8b4b283499b26b72 : SUCCESS in 50m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a285f9c4866e43ce89fee9313067fb1d : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1d618b8fa298a56972117a837eb23e6e41e09860","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"874be6f68688fe79947ce4392b4d5ad6917a3a59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:36:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b5fd61d83eecb73e7a1998a141a8eb42cea4233","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:06:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f91d75f24274312b286df9e59fa7d91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db14b2dac4e349ec9c98d1df4e139881 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047a516dc5274230b8198a5fd0a9b0dc : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20fef40960a44859b907ca72e06e17e2 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7caef49e44ea4d15b0f12a1711d61d57 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c58ee9722bf48ec99c54397a0a4c95b : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a053784d32045d49371ecb84a9e2c06 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/915f246284bf407094866c834fc668a5 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3996930dbdca4dac83c8609fe1090170 : SUCCESS in 8m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f85967be76d94469aed1ecd3111c45f5 : SUCCESS in 13m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ba2b2945f59d47088833f7dfb6412c35 : SUCCESS in 8m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8bae4aaa6daf4fa4a8b43a9f51661647 : SUCCESS in 11m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/34dcfe53e9ef4d889af069a91790934b : SUCCESS in 5m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c6ea70b995094a90ab70176fa3f35cbc : SUCCESS in 5m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/435a53829c494f649c197bec085e0f1d : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b4debb1af8014f658d7e085f03f57ce3 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a8632ddea31420baaf0f148485e0379 : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4e3c4fd73da44693899a69670ad38d7c : SUCCESS in 9m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5d6d415955294e8eb7624850a560ecea : SUCCESS in 9m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/acb6904b872942f29a44b0793f75e7b3 : SUCCESS in 4m 08s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4984891a9db842c39062ca7604706e59 : SUCCESS in 11m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c533ea85a4e24281b92df54192ed1563 : SUCCESS in 3m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/57f7491df146425382db391470dcde80 : SUCCESS in 32m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a4a362228e4243b09f33cc42d43b182b : SUCCESS in 11m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/db27d37e3e1d43c8886558016b419ebf : SUCCESS in 14m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a210ce01ebb44ff5a83f45a61a4812de : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b65930c202834c789288031654573241 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7cc8056a237b4427b4fe29ee58c091c9 : SUCCESS in 24m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/45202b1882714541b20474f4b2515133 : SUCCESS in 6m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/16d5c3bf0fa044e3ab265ce750aec67a : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5847a7d4c3e448cabb284c07beb01523 : SUCCESS in 7m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5dcb7735b5b64308b54ce4a1eae97817 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c65971b33e14432b8e3765e63160f3e9 : SUCCESS in 48m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/972799d8340148a286ac2167a52838e8 : SUCCESS in 1h 21m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c2ed05b5798b9b8fe6fd1ab9ac6c7e7d2b63a769","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:06:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0f4d651c331f79dd2286ba72b73482f8fcb3a7c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:07:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed34a1f1e804cbfafb8c5e2444f6a88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f8c654e38a4195bee9779c02186902 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f71098d793ff33589b6c3af66a6286c74fc7d773","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:22:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ee7c253923cd1fcf3e57ccb53ea22575202681d","revisions":{"7ee7c253923cd1fcf3e57ccb53ea22575202681d":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:06:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/892929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/892929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/892929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/892929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/892929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/892929/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:06:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:06:45.000000000","tz":0},"subject":"update constraint for oslo.service to new release 3.2.0","message":"update constraint for oslo.service to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9045130ed31761eb837c0c66133a6cf9f5c31d21\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Iabe715ca5b3139703a20fca595bfba1f9e59992f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ee7c253923cd1fcf3e57ccb53ea22575202681d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ee7c253923cd1fcf3e57ccb53ea22575202681d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
