)]}'
{"id":"openstack%2Frequirements~892930","triplet_id":"openstack%2Frequirements~master~Ib9eb984d19e8cc0bcba4e41ef80fd4d4daf1765a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib9eb984d19e8cc0bcba4e41ef80fd4d4daf1765a","subject":"update constraint for oslo.i18n to new release 6.1.0","status":"MERGED","created":"2023-08-28 13:07:09.000000000","updated":"2023-08-31 23:54:26.000000000","submitted":"2023-08-31 23:54:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892930-new-release","meta_rev_id":"943c08f065ed369c181c71bba3b0192971aff39e","_number":892930,"virtual_id_number":892930,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 23:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-29 19:33:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:41:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-08-29 12:22:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-29 19:33:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:30:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"1cd5b1aaaa934100b74f6ca2ca8d6f32141a82dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:07:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7a4363a67cffe76551a23f9204c1a92908534d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:41:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d2ffa5c1f974531a7acb03b760b35ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9514aee131e843ffb2155314f310083c : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87f8736f7cc4526ba8a5e763526a7ab : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a408f64c056444a97271c3c8e2222b2 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d86125d9f31c44b490e7b4ca9d391b75 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfda82c2477d400aa361a52321041c4b : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4f2804c7442445c9dd9c03395913c5c : RETRY_LIMIT in 2m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3dc708cc4b148b7800158fc0755172f : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b63ecf5a0d084d8ab32859f55a1fd3ae : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d12c85fd8c6d4f5b9b340cf607297945 : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/06ad07308aea408095a92d31ba0578ce : SUCCESS in 12m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4a9c4657f8154dcd8aebe2d36d788c01 : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2004754a15ee4c5290bc1f71918fc4ac : SUCCESS in 10m 37s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/91ad37619520463c9e66d1628d402b5b : SUCCESS in 9m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9ead63b938c4445f89bb41fcc51440ba : SUCCESS in 4m 59s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1bfb0baa843144248b1c406b555c92a3 : SUCCESS in 5m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ded9e60b67784a889146f59d3be55e18 : SUCCESS in 12m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bc420d9abcb54e43a842ae0d77213dc4 : SUCCESS in 5m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/88832470d98148218b7064e8942ba326 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/aa42505a2f674392b5c60e497ee157fe : SUCCESS in 8m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a50af58aeddb4487b4e0c10bcb9a40ad : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1d3117242f84e7e8e8df25f365fdb27 : SUCCESS in 8m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a44d3d3299ce4e49b4a1cb17a9c74ff3 : SUCCESS in 7m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0684d31f4ede45d4921a5d4c9645ec75 : SUCCESS in 12m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2f3f793f2f3b4dc5b8c20bae16c66d6b : SUCCESS in 5m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/37a51c16f7e744aab973354ad812e1b5 : SUCCESS in 10m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ca4125da800c4acdae136bdcf0743359 : SUCCESS in 5m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/64799a5764f04c39829ffe1b109c8c34 : SUCCESS in 12m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3100dfb128fa4d079387caf7b66f3255 : SUCCESS in 24m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/26c69ce2f99e4ea2a5213b0119ed0f01 : SUCCESS in 9m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bfdc3160978f4fcdb6bfc94b03ca4d05 : SUCCESS in 16m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dffb0f091e1a4715b68dd03c5cde645e : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f4099ad1ca554b83add97267bf77793a : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46d3af6d33314d43ab16cd73998d5e0d : SUCCESS in 24m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/52bf1fb3eff94257a203833b9ab45257 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c7253caaaa574d088c70b328a1a89575 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8c6471ac11c46d9971a358899cb5908 : SUCCESS in 7m 11s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9d32710a500541528096d2b0645fab64 : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2972b9df1e1c41fcbe59ce574ae7290d : SUCCESS in 1h 11m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15440e8b127c45b2896c27ef8b1349a0 : FAILURE in 1h 24m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ffbd5427e7ff74e6a5d8ace3f25977d85f714780","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41c355f859d86ba2bb9f4269bf7ad05e9c9d41da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:30:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c352c0b716bc45f6aa3982420f9ab40b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897ab15ef26341b885127e9e79fa313b : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba87ddd4a18644f49894b994ff5644b5 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89083d70a0914ff6bb9ffdbad0ef5f53 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d86630a2aafa4a5bb04b0f26c3b76577 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e17efe0aee014a849f33c84e9aefc726 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b326efb9b66409da614690a1dcbf1fc : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/713e86ed51514baf83f3e5ed061bea8e : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44b4b0dd821a4ab78d02e9edaa1fbd4d : SUCCESS in 4m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c3aaf141bba1417bbbc3a61d9e2cd24e : SUCCESS in 10m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aac9ef96e4ce4aa283cb9b1330ca4b12 : SUCCESS in 13m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e88210a5b37a4ba59a64a227b5930da1 : SUCCESS in 9m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/45acc91907854a90a2570576e0ee5b4d : SUCCESS in 12m 37s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/dbf393f5968d47b894dcc58209c09bcc : SUCCESS in 8m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7d218a19061944779f919012631c0cee : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0a8e19532a0f4514befa733cb62d792c : SUCCESS in 5m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6fb6927b53e2431a89fa83c7170f9470 : SUCCESS in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/27c7c9da12ea44ea8308a7123baf3ede : SUCCESS in 5m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f1bf5d109fea4cdaba088342432f79db : SUCCESS in 6m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/45434ce618684f8e980f03b8b9d24589 : SUCCESS in 9m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/14f149a0e3e747a6a0e09b126c30e943 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0273930bef274b5e8ff5ea584767783c : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0f9350a14ece406ebb4b13473d280ae8 : SUCCESS in 8m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ae653794049a4238ad77333c5acb8967 : SUCCESS in 11m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7e15798f72b84f3ea567026d3e418b9f : SUCCESS in 4m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bdff71155f50494e8f7b74c904044e2d : SUCCESS in 11m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c96f589d09c94e48b9277b80c91a50b3 : SUCCESS in 5m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d3807dbb145f4a9e871538fe4cfaa7ab : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7b50f325440e4299a016bd4198b4e82d : SUCCESS in 23m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/90aa166de24a495b9c746fb0176f9ffd : SUCCESS in 12m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/55f65e3cd1d348899ae7166d60301cf3 : SUCCESS in 13m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/637f5a6c152f4803a9e057ead9142fb3 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/62d507ccf26d4fa2a2a2fca4275a4ea9 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dedd2bc543244a987501c3641ac41d7 : SUCCESS in 25m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/be98a174fa824fe584bfd18a534d204e : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ff92d3ccd01f4dd494b8d07bf1d5776d : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b03686a9bbec4a838aeb48622e7ceab0 : SUCCESS in 7m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5b30cb61055f47a69fcd2a662f2c5c74 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c90df3ee450c46d9a925a4305b8f6168 : SUCCESS in 47m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d39c3131b3f461ea366b7d0e12ac6a8 : SUCCESS in 1h 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1429c02f7d044efffaa91f2501b308e630397920","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:22:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31c0bbf0b3c2cd91a78b87f895cfc0a328bdcbab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:33:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21199f667e6e8edfcf74bf1a145e7d8e25be247a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:34:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"594e49111d12106f2e9b18fc9ab6f2061fab03c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:12:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/648ed19396a24807a041bf11dc104aae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77db31f3f2c944b1bf4b03126f2ff60e : FAILURE in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b3fc3d9eb444b0b9d93c7ef5102cf6 : FAILURE in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6136fd754a0446b587a197416f879e5c : FAILURE in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c74d8ba3e3c74759bff647e7409bf143 : FAILURE in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30c09a13159745ec85c7492ae810087b : FAILURE in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad33004658944077abb685633e5b4ba7 : FAILURE in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/615b73b39d014b888f1b3fe6ef48d95c : FAILURE in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3544de24a23d444eaa8a678f4841ceb3 : FAILURE in 7m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/99962e2158bb43f5957f7fd0b4bcca10 : FAILURE in 12m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/40c7e45d8734447eaedded28ffa6ceb6 : FAILURE in 8m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/591b3eabe68f483f99edb31df00d1588 : SUCCESS in 11m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/998ce6e21e1247918ab9a2e02586428b : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/723dfe913abb4bf28a111d8ce6b29da8 : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f2c9ed26cafd4b1eaafd69679cf9c5c9 : SUCCESS in 8m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e97a699e1938442faf24446e9e676cad : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea10701a73b24bf49c0df87cb2ab958d : SUCCESS in 8m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c55cb29574b542f096a89399a7cc8fbe : SUCCESS in 8m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2df7d8ea517643ab8005942dc2b16119 : SUCCESS in 11m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c04aaaabf0ae4b969469084f8b52bc5d : SUCCESS in 4m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b71af0de35e34e98bd59f3358119fb59 : SUCCESS in 10m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/209a7ed69f6c48e69b3afa6fc6851c8b : SUCCESS in 4m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/afe615dbe8cb472fb8e82539a663af45 : SUCCESS in 27m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2fd3f704a4b64e6891bd2b757711b03c : SUCCESS in 13m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0b280508fcde46788cbca679e5fac3e8 : SUCCESS in 13m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c469a656b13b4c3b821e8b70d0aabf80 : SUCCESS in 7m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f2ab85d057c64b2896ae5991422020f8 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41e609d623da4969abf578f1e270dde1 : SUCCESS in 31m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2ddc3b3bdc12463f890144c762cdf06c : SUCCESS in 7m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/37d4be0c9a72403b82585cf8104eb22f : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38f63a5f0a694a3ba9b9133f0ec0e229 : SUCCESS in 6m 54s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a9139eddf3594524b00e7118acc5ff49 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/160f9764fbf34406bf4c553c81ac1133 : SUCCESS in 1h 03m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47f617b6c8db4eac96be9a4978cd0ebe : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":1},{"id":"928723d10ea2b0795bf9070d45974711b519e3f3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:30:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"073a383c7370f32fa5715074a618c87389fc16fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:01:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39993283d3d842cdbbf1e5407b4761d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f1e177b14b4cb583123969682875b0 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e86ab6bac8a4cc6add184c0c9c0f01e : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4e364ac1ead4a9aaafacb4776ac4e5a : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e789f04303f947a99091af923fc14323 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a0ac63caa464c0abd2cb232787f62cc : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ee52d94aec0487684c0f0a42a6a53a9 : RETRY_LIMIT in 2m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc0880e071654f51bac0e671d51058ad : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4b7733bfebc403c9d089c82a65dcfea : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c6e132a31af54870a9b1be01d0754924 : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/29d278466bb14931959f47e3465b85dd : SUCCESS in 12m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ab018e7c76b74a39ab2a2d153f15de73 : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/16772381902c446491e2e06b96d6eb91 : SUCCESS in 16m 52s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/06275be7ff6b45a5a6687d75121c99f2 : SUCCESS in 6m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bdb094ff42b44c3fba3c22beddbe3414 : SUCCESS in 5m 02s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2ef1d8abafd74c2a837ba00c7c7c6bb9 : SUCCESS in 5m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0b42bba8d978472890333d8cf79e66cc : SUCCESS in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5fcb6af38aef42619f358ccb0f1a6875 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a98971c842bc4dc79d7964299a97b7b4 : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eda548d23e034d8cbc7ce79e02ddabde : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/eea3a1e80ad844b581b5e5a025715413 : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1471d4bd4c1e4129b166ee1dd94f5141 : SUCCESS in 8m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/18d8442465464b668f1975293e2fe646 : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/44090b5298134ed390f01089f6814da7 : SUCCESS in 10m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f580c692e6844b539bb5266ff6630325 : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9532ed894edd481f9f23ba6057e6d826 : SUCCESS in 8m 55s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/390105ce30504d8f94264e98f39f5d90 : SUCCESS in 5m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9ee6a4d84df74612aceb42de60c4e61d : SUCCESS in 12m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9761659c76064727ad4dbf9473705995 : SUCCESS in 23m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2b5e72b507ed4c1b847594a1fa938c69 : SUCCESS in 14m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/84d1bdd831b2483f90113b2e61ab35d3 : SUCCESS in 13m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5a4e2b4801f7411f952c5767271d71bc : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5e014518f17a4c78851bb22079671b75 : SUCCESS in 8m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5710e7d9f6644c49496b63612152247 : SUCCESS in 22m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f8ee4eca9b97440bac93ec7dd9ba8ae3 : SUCCESS in 5m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4d224561238840afa1fe2c8512e1e6bf : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/832240e443eb4e02abbc99ccbd0824ba : SUCCESS in 6m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/42d710fa33e843039dd066e40f1d279b : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f7c887f51824be4949eadadf2a83979 : SUCCESS in 1h 10m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5fb4659762b45ba8ee42217f96e8f34 : SUCCESS in 1h 23m 59s","accounts_in_message":[],"_revision_number":1},{"id":"eed41405732bcb2b32babbcad9041995b6e02656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:01:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab32874aee4a3891a9a666d4174288768a38434f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:54:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/272109d71f4c45da9e9298f7b24b7998\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b19fbf8065e450d994c1a9ba701bbd2 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4306a55f268d4b179999aa7318b0d4f1 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b214bf8a90274ee3bfd2eab079568feb : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab8149cf177441ac8993a22e3dade076 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/724dec3e540c40ddaea6fd061688e609 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/717d42492b094850b870e6fe932c3fcf : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1366b0e82b8c446398c2f73212108e7a : SUCCESS in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e920e2b4d454e44b1bc2bf0395989fc : SUCCESS in 10m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a27c366ea2a647bf9d8784a7a9fc2643 : SUCCESS in 11m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a6f48577a2c242a58607608009601a8b : SUCCESS in 9m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/632143f5917544af836888ce158f149d : SUCCESS in 13m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/95df4d084ba7428190a47548f71451bf : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/59fe9af49a7e4a0b978f2502cccaee43 : SUCCESS in 6m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/246d116d3a93479fa28fbe2e7bad9c45 : SUCCESS in 8m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1f830433e8b445ed82bdb0e11a6061db : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50ef876f14ae4b4bbcfb9824ff92de07 : SUCCESS in 7m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e3e84762f2cf4341b9d3633d705e8180 : SUCCESS in 7m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9c9bcbbdebd54522ac05a9c8f3a3f64b : SUCCESS in 12m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8334352e1aba429b85b03a1ccd3f2130 : SUCCESS in 5m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/141e4ab594144bf4917e73c9f48094c9 : SUCCESS in 13m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7eda88dff6114247a4d54c94ef46cc1e : SUCCESS in 5m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/75516492d40a4d61b7e3d6be4e90d2fc : SUCCESS in 24m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/062b08004a544ebfa7982a5a348ac693 : SUCCESS in 12m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d3601091e163467fbb7fd243b95811cd : SUCCESS in 16m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7cdaf2112b0f4e629bcdd73bae4770a7 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b2676a149b7435589f1662a71547e88 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7129d1563a0947beb578aa5cf96b1c76 : SUCCESS in 23m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ab61d934914d450fb654ec6518940eb1 : SUCCESS in 5m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3b9060e0da944d828c09da1fb732880f : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0e625855f5a4e6ab9ae2908531807c7 : SUCCESS in 6m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ec63f996b6b74d139785109bcbade64c : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3676ea566fb8453ba4fcc99b432d8e4e : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01bbe3ae3bf14ce797b4b218a5d403e9 : SUCCESS in 1h 18m 53s","accounts_in_message":[],"_revision_number":1},{"id":"943c08f065ed369c181c71bba3b0192971aff39e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:54:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bbdd866a3f1de9e43ef2af6c6d0a56a83371e90","revisions":{"7bbdd866a3f1de9e43ef2af6c6d0a56a83371e90":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:07:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/892930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/892930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/892930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/892930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/892930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/892930/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:07:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:07:08.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 6.1.0","message":"update constraint for oslo.i18n to new release 6.1.0\n\nmeta: version: 6.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I19dd497ca227df83334f702b6d228d5a048d482e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib9eb984d19e8cc0bcba4e41ef80fd4d4daf1765a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bbdd866a3f1de9e43ef2af6c6d0a56a83371e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bbdd866a3f1de9e43ef2af6c6d0a56a83371e90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
