)]}'
{"id":"openstack%2Frequirements~892932","triplet_id":"openstack%2Frequirements~master~Ic38c3facf3b16824f9ef7c466e013008edbddab0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic38c3facf3b16824f9ef7c466e013008edbddab0","subject":"update constraint for oslo.upgradecheck to new release 2.2.0","status":"MERGED","created":"2023-08-28 13:10:47.000000000","updated":"2023-08-29 12:22:41.000000000","submitted":"2023-08-28 20:06:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892932-new-release","meta_rev_id":"86e087a5eb6582210014a6fd86860f5b4b64bd10","_number":892932,"virtual_id_number":892932,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 20:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:22:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:55:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:22:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"bc594fe82d84d6b0f2f778a463b0dba9706cfa09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:10:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318f0adc003d126ae078feb2afca294ac72eaeae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:55:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2780bc7b131e4ffa9f0e82b905168fff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b389ffb47ef94abe84a79409955279d2 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1355621719644321acbda3edeff69f5b : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e352b05f00d4103ac6a56e63f71bab7 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4732dd0b764541b099380818ac5280e6 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddab4588cfb046899d152cdc129f0f99 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/437cb58507d645dd81419247646bc866 : RETRY_LIMIT in 1m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e825832aa9e4b51889f5f78c2c02382 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e775e3ae70be4dc99853e21cc5f0f004 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dea7cd30aaac45cd8dfa317a2ecd3753 : SUCCESS in 9m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d170ff736b084de7904adc87e7fb8218 : SUCCESS in 11m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f342f95df56c4e9cb84521e22c903892 : SUCCESS in 9m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f56be649ee2949b0b6522a444cb1e39a : SUCCESS in 11m 22s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e46d198305844289a0100c73a143af9a : SUCCESS in 6m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ef8d4d3cd9684d068075710f5813c057 : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/31e4317fd8d44b3ab4d8065e2c1ca86b : SUCCESS in 4m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0a9193a179e846a09c70e7d20964e8dd : SUCCESS in 11m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8d336259b85a410badec2830e9c08dd8 : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e27ac75aec324638956a9f2b712bf178 : SUCCESS in 4m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ff1ef321c73747f29603c46e40c7dcce : SUCCESS in 7m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fce339a7a07a466eb0cddd58192a2678 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ce634750554446890da6485ede9073b : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fc95cf4a0a4949d49b71dea34cde1942 : SUCCESS in 9m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/10d29335ea8f40f9afd3c252a79eb476 : SUCCESS in 11m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4d615a95607a42439937183a5dc972a5 : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/72d49583c8434f8d8d15f853dd719bce : SUCCESS in 11m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2e58df259b674654a16ba4e9bb867f54 : SUCCESS in 4m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9ab5eda5bccf40c29be8388fc1f4dbf8 : SUCCESS in 12m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/759a0cff494f4d06a1224331ec8b882d : SUCCESS in 28m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/49763647cc4e45efba6e18819c3e391b : SUCCESS in 12m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/740f9b5625f64663bc84f6000986acdc : SUCCESS in 15m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a9eb140a4b424c6a8d24bd50362621f0 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5020e84d48dc4e7793a51f08b4061343 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3692b3cb91c349d2a5a4688be44ffc78 : SUCCESS in 25m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d1d08f48525b4d01a9436ca3fa236419 : SUCCESS in 6m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b82cc73333304e9fa0c9649de5f54084 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd60649736384f83aa3cc67e5265ec86 : SUCCESS in 7m 12s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/848c364c6474468f9491017d9ad48b6d : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a39587b6bab74378a9d747366bab0e55 : SUCCESS in 1h 03m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bb5e3a68d5d45b0b94bf5087abececb : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"40fa22051c8629667aeb807673672319b4bd139b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a2781e577f26f0f75d6803abb8c57ccbb0a250f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:36:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0715fa4c5a8e0708b759f6b7db4eb921edc6fccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:06:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e4c5421ffb64066be48061cad065da9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91284f061f24ffc9c11d9f273cfa94e : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5fec7ddd5a24600bf14d31e03b7437f : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69e392f852dd44ffb74252a346cc0955 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eecf04f423c4c5589a557d80fc9687e : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d42ddd6b95584b849c3f29e004f7b667 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/604eb107fc4048cd89f0ebdc2494660a : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4d8989bdf5345199aee0fc3da969c8e : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/32c62a33ac254ba888552372bda3619e : SUCCESS in 10m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4ec1785e70f44789a4d485be516a24d6 : SUCCESS in 12m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/24f6e8d56d944b9bbd72179b67c7b0ea : SUCCESS in 9m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/928812d16f194581b03148f1b0f1a1e8 : SUCCESS in 10m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d93b98c154ad4475b91cb82141a27b44 : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cab6ff8027504e28be68dd8e537aefba : SUCCESS in 5m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9dcecd14ca1d431db21accc62556c024 : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5c8a9c9e1c7f411f9e5bb616cdf51b28 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52c2ce0326be4daf9ff45faf3d5b8e75 : SUCCESS in 8m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/10c6494cf20e4e5989402ab3ec8c41d2 : SUCCESS in 8m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/93060fcb096945c3b26388c4b75a21b4 : SUCCESS in 16m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cfc5ee25ceef4f149d8234a4376c3892 : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/84a62bb2e650432680fa93c82f0abceb : SUCCESS in 9m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6900b9e91e9242cf8a4ce72203adca93 : SUCCESS in 4m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/30c84a99020a421f98dc5c920c7d0097 : SUCCESS in 23m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7684ea095c1d4d92bd21096d6474dd5c : SUCCESS in 10m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/28ceac121e1e41c580902ad26777b4b9 : SUCCESS in 13m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/96463c7ac61c4549a91acea2c87664af : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23cd845dee0b4b62bc2f6bce7c211733 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7ae1781bb5d445d9a657431c605294d : SUCCESS in 23m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4ddf5ea74b134d8b9b512cba82250368 : SUCCESS in 5m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/758290b48e5d4ea5a01e15764e86ee72 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95d7fc1f8b204b5abd3ef87ac87fa6fe : SUCCESS in 7m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4ef2677192d6465fae325fdac61af9d6 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/990270adc6fc4e5c92cebfffd091d8cc : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa3cf50d211e4168b08b34ce72cdda74 : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8ed08b44aa7305c7dbbe9b08755068d441f9abe0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:06:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4eac6d3b06f56dd792cd5eaa6b27c38a7dcc6bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:08:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5314df10a044aa4a21919bb230549e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3633514e1f154ef6bf0f12a0cfeee7ac : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1},{"id":"86e087a5eb6582210014a6fd86860f5b4b64bd10","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:22:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"310bd4ac0413a098835c1946d0e9658709e36216","revisions":{"310bd4ac0413a098835c1946d0e9658709e36216":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:10:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/892932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/892932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/892932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/892932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/892932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/892932/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:10:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:10:45.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 2.2.0","message":"update constraint for oslo.upgradecheck to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3ba23a2210ba764ba32bcd93a999716d7a1127ca\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic38c3facf3b16824f9ef7c466e013008edbddab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/310bd4ac0413a098835c1946d0e9658709e36216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/310bd4ac0413a098835c1946d0e9658709e36216"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
