)]}'
{"id":"openstack%2Frequirements~892936","triplet_id":"openstack%2Frequirements~master~Iba0fc098c337c8b18c904eba1d13f417879f1cea","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iba0fc098c337c8b18c904eba1d13f417879f1cea","subject":"update constraint for oslo.rootwrap to new release 7.1.0","status":"MERGED","created":"2023-08-28 13:11:43.000000000","updated":"2023-08-29 12:22:35.000000000","submitted":"2023-08-28 20:29:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892936-new-release","meta_rev_id":"99676188da899353d8de46be69f314d68f9f81f0","_number":892936,"virtual_id_number":892936,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 20:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:22:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 15:12:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:22:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"85f1672d7f31b0c1332d9e6227aac9da9f2b0fd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:11:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7610e2652d863c2cae05f74faaa16f72c3f8369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:12:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af8d9502fcfb40c1bbfdcd7ed8eed074\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56879dbc1d3c4e70acb3bcca0a691e87 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0751b4420a408b9c4f84ece88486ad : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c80e37eb49f4457bc990c7cda5fe5ef : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f212af9e28144cbd9d6b8ee8817b285a : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c5c40c1423b4e688b400d0fc5cc05e2 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/555e4e78d6c54aa2916afa7b2e1c4b20 : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4be6d6b2b454a128b0ab287f97c334a : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd8ba9645b844b8fabaf049cd00a454c : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7dfb6d7d0f2f46e58374dfd72a817cf6 : SUCCESS in 10m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6e2c257b4f54462192ca9f3425f4842a : SUCCESS in 12m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e911fb5b5a82430ca51a51e3c1892ac5 : SUCCESS in 8m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1532e22bfbda4556a1d81aa7bde72dd4 : SUCCESS in 12m 46s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c71412376ef84a32be5af68a23da1e0c : SUCCESS in 7m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/31b78bb41d104a9b815a1bd3dc01e41c : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c5273336abae4b0e9a5622a01ef15c48 : SUCCESS in 4m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/10f0c0b6a30f4cf993ff4e0c46976e7c : SUCCESS in 10m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/97dc080c99664d198972575bdb4c181a : SUCCESS in 6m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b2d04e2520af4c8ba2ef85f16cdfa485 : SUCCESS in 5m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6ea2d055120c4154a04e758b9517859e : SUCCESS in 8m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/955f8e31e94b49c7a6e6c814c516e8d8 : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/172ecf54354542de9b7689f1f7608c99 : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8163d6a3776d4344a65bcfcd8ac27725 : SUCCESS in 8m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1f8766557aca45e3a867d4c57bc5e829 : SUCCESS in 17m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/11cad0288c644e86a19eebffd1b71443 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d424b47e6eba46539c3d554259125b64 : SUCCESS in 11m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b0efa37d7ea64ec4aeb3b973eff94a33 : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ef9444728fd8481b9b64ebc63c641c6b : SUCCESS in 10m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/00c49ae003ef4317ae59a6c102922374 : SUCCESS in 22m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2428b135008e41dd93c4268157b43b32 : SUCCESS in 12m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2b9d4cccb2ae4e3581daa00c94a33f66 : SUCCESS in 14m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/44b6979c125f4a679d9d20e501d4b8db : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5fb3e68b98541b0bce5255a178fd4e5 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5660c2c0202c4446be54151ad98e0fbf : SUCCESS in 25m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/85ecc64c2c8f4852af8e8126f3f147e3 : SUCCESS in 5m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7f46a90fd60846ddb87c5423594803bf : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3db55660e8414fb593c9b5b94f1447bb : SUCCESS in 8m 17s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3ab13a789ec747e9bfd2b683e22017bb : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e17f2a988d9144b1b654ab694f60e9a0 : SUCCESS in 1h 05m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6f2f1b1794e407ba554a79b0ff3e12c : SUCCESS in 1h 50m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ab9415778d7c7c3ef3e77765cc88840a6f931edd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aafc5363e2521ff57c2bce2ac176495d62b60fdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:36:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"258558d9aa7e51e82a8cca7759be2e1d7c90b6c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:29:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c50c377b19df4947a46b339ee6288037\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81b59f6962849018b4a51949112df6b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7607d9bb132b4014b70c8aef0ff7106d : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/843e0c0e3a4940dd9cdab011935e82b3 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb780b80903146639f9bbad8a4693acc : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/772ca793b27e44ed90664b2ebf28d668 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f948f3ca3104e59bf8c908111f4c253 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09851b56cf01477690573228edd94c9e : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/247b3e4648d14c5990957716b7f9a12e : SUCCESS in 9m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c871f8ca79a24125916255a6d91b4691 : SUCCESS in 13m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/527986c135c3403c8ae4861a203d2028 : SUCCESS in 9m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/73357a4fa62b451c97e33cbab4cde240 : SUCCESS in 14m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d67edd91483749fc92eb4f0914338d35 : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6ffc3debedef4b989ec24ff8fcfef636 : SUCCESS in 5m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/280437a6bccc4847a441e46f023d82a2 : SUCCESS in 9m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8223f91252144ec49e0325f90dd5d5ec : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/05f069618c9d49138ea63477fa9131eb : SUCCESS in 8m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cc2bbaf795054e628ee5f13e46590f44 : SUCCESS in 7m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d7a783aa593642cea1e935e49a69345a : SUCCESS in 12m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f3609c9a38954228b0e610b8cfce2ad1 : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cb13453a2f634dbc8054f1609bb5eb4b : SUCCESS in 10m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c62a93eecd8242a785caf365d85859a4 : SUCCESS in 5m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f6ccff4f71d14bc69d5a0860b9af7223 : SUCCESS in 26m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a568e53047eb41aab60d3be64f27b4b3 : SUCCESS in 11m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/37cf511c57a045bcb6d8de6ea6188761 : SUCCESS in 14m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5e825051fd2c4ef9b3370713c9b50013 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e5bc4278ef445a38e4d4a545380daea : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3a4b71fc16f4ebeafb3568f6ef320ca : SUCCESS in 22m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b1e0898f65304eaaa4788ded7f00bd46 : SUCCESS in 6m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/791cf5a49157431eb9c8554c7c0fea98 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f9a51a4093d4176a7f2dfda6a2e2f25 : SUCCESS in 6m 22s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9bc99aed81064789aa528bdef868979f : SUCCESS in 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ca361c134cc48d09b3fdc9a5f28ea1c : SUCCESS in 1h 04m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/297309c256ec4785beea9e0d1956e45c : SUCCESS in 1h 39m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5371a62dd5925916b92e2a835a33250ab1ba7d14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:29:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"65df776eab775251356e3674ebe693d19564b314","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:30:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39cf530b5fd94457ac931a8cef6eb405\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872b71f18c8c4b34959ea8e6face2093 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"99676188da899353d8de46be69f314d68f9f81f0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:22:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd6779e8bb054eae546bafae7cfe5c31114612b2","revisions":{"dd6779e8bb054eae546bafae7cfe5c31114612b2":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:11:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/892936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/892936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/892936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/892936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/892936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/892936/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:11:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:11:42.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.1.0","message":"update constraint for oslo.rootwrap to new release 7.1.0\n\nmeta: version: 7.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If881591bca169d6cb4b0d30cd831d49958f99280\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Iba0fc098c337c8b18c904eba1d13f417879f1cea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd6779e8bb054eae546bafae7cfe5c31114612b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd6779e8bb054eae546bafae7cfe5c31114612b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
