)]}'
{"id":"openstack%2Frequirements~892938","triplet_id":"openstack%2Frequirements~master~I5ae670c821f17aeed91c492b004aed2c52fceca8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5ae670c821f17aeed91c492b004aed2c52fceca8","subject":"update constraint for oslo.concurrency to new release 5.2.0","status":"MERGED","created":"2023-08-28 13:12:04.000000000","updated":"2023-08-29 12:22:29.000000000","submitted":"2023-08-28 20:27:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892938-new-release","meta_rev_id":"1ffbe929716e85c772cba61badeddb9e7f0b2c57","_number":892938,"virtual_id_number":892938,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 20:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:22:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:22:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"142060dcdf94de909b9758338139725156992052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:12:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a145415a8ace8017653a35daad61636eb61c5f66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:54:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa68c10070e45c0847bd097f9410983\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521587c946c84b8b8d911caa02fe6c5e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d79a14fd1e4bafa55c81f03b4dd6e9 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df07b27baef14443820a5547b7a6a1a2 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b04b9ea131b1456a893ca72f23b05e6f : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c42dd3650f0740abb18623694b141b3e : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd88ecb177604c8ebe646281700c15b1 : RETRY_LIMIT in 2m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52b4f3eef31b4104b0b1d90579d2d8e7 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0cfc6302cf804e8984035c93466c3924 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/220ef1d3ea53476990fc11c36ca98e68 : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69f9dd9ded0f4b8b9fb50d9b956caf51 : SUCCESS in 13m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6c7044b81d6749c190386ca11653c06e : SUCCESS in 7m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6255b4592c424f02a74e8f740fa743c9 : SUCCESS in 10m 33s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b439f22df4a2461f8476b5499bbc5b56 : SUCCESS in 6m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d7f79bd27989402fb2973f50e8596a3d : SUCCESS in 4m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8cbb2e7cf1cf43dba9d5d20f18105ac3 : SUCCESS in 6m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9575025f46b84da7a0f24e3199cacb1c : SUCCESS in 11m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1b0c865f99394f4c9e9809f22928e006 : SUCCESS in 5m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/73240c05c65343c898da9ca8cc3f4b6c : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4011d379d06843298928207f3db3e8f2 : SUCCESS in 7m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4ffe980b4ff34d4f880641834bbf4a67 : SUCCESS in 13m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66327987a4144b17819581fad8212e3b : SUCCESS in 7m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/224adad6dea64a4a90d4cccc28f4bfc2 : SUCCESS in 8m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ff6ee107ae7344508823f9c1f0b093e9 : SUCCESS in 13m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0e692af864dd47c4b1b4ab66dc2f848b : SUCCESS in 4m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/abe0df954b7b43b5add808d184437bb7 : SUCCESS in 8m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/27f9629909504e1f8734f8c34ec5ea0e : SUCCESS in 5m 01s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/103b2d307ac5479fa0b35e392d46d077 : SUCCESS in 9m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cd7c9e76186e469db44535c76502697d : SUCCESS in 28m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/63f80cd160084307b7f92d19ce8a17a5 : SUCCESS in 10m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/79855391243f40978b4fb87b8744bd36 : SUCCESS in 12m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/81179a53fb8744879f35bf14d73e662f : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/62a72a4360894508b66f0a46d1d72aed : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc89372d90034dc7a001ccc4f9acc6a0 : SUCCESS in 23m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a786482f82394c6f803d6d749f7c98f9 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b574e47abf8f41abaafe614a6f400135 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27ceae9dce094c5b8e2a21f3153f17b2 : SUCCESS in 7m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/cb8ad1c08f504d1ea698e991a984f676 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05292e78b7e540e59f03ce05e634d3be : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/273452f672924433b03abde6d0c3f60f : SUCCESS in 1h 22m 52s","accounts_in_message":[],"_revision_number":1},{"id":"36d8ced061b03fda037fc236513011beb41aee0d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b380aa875bb9f4e4fcd1933e7d629ffc97151c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:36:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"263cbf8da94d199d86d44f38eb16db3124ba7622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:27:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20fc8b5b2f5f4bc4b07b683003f14ce9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e23bd203d794bc4966fdae9ac910dbb : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2abbe246df481b88766fc09e689b51 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/718a10b94bb343d08db77a65a1dcaf78 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f33384da3fe4da8beb9a70463bb615f : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6586a8100706489a91a325f3ece64bec : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be43105bcb364a0985c62d561e107704 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e9868c8da814800a44ad4ac1e122d5c : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28e6ca47f6a848e5a665421aa75129dc : SUCCESS in 10m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69096cc8a10e43ed8ee327bc9126c644 : SUCCESS in 12m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bc2ac829da074d8894f52b164d368bcd : SUCCESS in 8m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7c62db8432ea42618a9d9a109881f613 : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/faf79c2687104559b3de73f317129675 : SUCCESS in 6m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/112df92d0f704f19aa10c06872411dbe : SUCCESS in 4m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f0b879507e374ffa928d4919d210a099 : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/063f34851ce246a79fe1c07cb5abfd13 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ccd8be2e81a34352a6de25a63892d976 : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/805aed3a98034b0a9dfcfea67a7cbae4 : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f855f3a717dc47a78525210e01d7d5df : SUCCESS in 11m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6b7fc5b27fe54ec3aed2e4dc71fddeaf : SUCCESS in 4m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/09fd6d88be824a77971d27c592e72b39 : SUCCESS in 11m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/202df89c3f5843999873d6404638e473 : SUCCESS in 5m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b810fab4578c4cde8e6a5ac7ec5f2307 : SUCCESS in 24m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bb82e6b82b724a59992dfb6850c91b2f : SUCCESS in 12m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e269b4f8d6434734be9f61f19efe6ab2 : SUCCESS in 14m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8806bdbb5f1a4786b35e7cba3c910f25 : SUCCESS in 6m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7f866ef4d9c4f88b95dda73cfe15fc2 : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/602f1e04d00d4b8180ef6ca8a3e8b194 : SUCCESS in 23m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a4b27ea8a34949fa86c980373d25ba2e : SUCCESS in 7m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c9a02074b9324adeb435c5764335520c : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab67bae8015a4d7191b162e040079ffb : SUCCESS in 7m 54s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/78e7f54729934961b24bf891e2dafbf1 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5857ed6dd2c64e319937c8518adc4a96 : SUCCESS in 50m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6a5143a3be84f97b7d47ac60c936818 : SUCCESS in 1h 38m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0fb17210615c89134266f0a46a2191524c671495","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:27:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cd4221d303406b4327ee27250603c9af7b3a0eba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:28:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88868501b5874723ba72ecfd90059ce3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127f164af2c64cd1970a46c72cea1253 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1},{"id":"1ffbe929716e85c772cba61badeddb9e7f0b2c57","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:22:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"080e54840a17772ad384c94c8a7c41a429bbfd66","revisions":{"080e54840a17772ad384c94c8a7c41a429bbfd66":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:12:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/892938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/892938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/892938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/892938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/892938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/892938/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:12:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:12:03.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 5.2.0","message":"update constraint for oslo.concurrency to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2cdf40fd017fa238b6c7216fea23b9a461751a0e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5ae670c821f17aeed91c492b004aed2c52fceca8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/080e54840a17772ad384c94c8a7c41a429bbfd66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/080e54840a17772ad384c94c8a7c41a429bbfd66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
