)]}'
{"id":"openstack%2Frequirements~892944","triplet_id":"openstack%2Frequirements~master~I695e2085ba58458f95c7806830a0ce6556e35c46","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I695e2085ba58458f95c7806830a0ce6556e35c46","subject":"update constraint for oslo.privsep to new release 3.2.0","status":"MERGED","created":"2023-08-28 13:14:07.000000000","updated":"2023-08-29 12:22:18.000000000","submitted":"2023-08-28 21:12:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892944-new-release","meta_rev_id":"a133732494bce3616da5e04b682d4b86e4e58aa2","_number":892944,"virtual_id_number":892944,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 21:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:22:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:22:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"4c925aee4794c73aa952d71ead4f7d6565256676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:14:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db3756171bd9763a9271c8c6f422e73789874dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:59:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e93eb9d541be45d0a5c694f03ab18e27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9082a5ab6a3462fb7524bdb471a9cf8 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e00993f55f4e29b87355d75d1fb4fe : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0808a5823cf541b4b0e45f08034a13eb : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b06dbc757754d86b078979b3bc29fa4 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a66cc28fa6db419e80f4c6cd06a74083 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ad0e236eb4e4081b341e7b2658a3ca2 : RETRY_LIMIT in 1m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17bda9ff76514b9faf580a3e94354c1b : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/699ab934447945229398ed7a779dd459 : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b800af0534f046808a92189a9c8053df : SUCCESS in 8m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/31c28ceb51aa475c8b96c393be9c566c : SUCCESS in 13m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7dc5e438e184497ca9e149376270a846 : SUCCESS in 8m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1f42fe9b089c40e380a5f3ab1fbe73e8 : SUCCESS in 14m 28s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/790a723802b849c49da22e082b8ae953 : SUCCESS in 7m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3ad9ba18d8a2401cbc3baf746b79dd9e : SUCCESS in 6m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9f0345c3c13045f5aedf93d1e4f85da1 : SUCCESS in 5m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1f90e68fd02347eab553f018ba8036be : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ec23a08139bc4e7aaf787f7f86dfd71d : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5f61ee820e9a47af8b17f51baae9cf13 : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d9962fd2055047c08284f130ff5264b0 : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/443b85470d35442ba4d46b08ab418680 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5af36feae5764463bb151ad243f8c06e : SUCCESS in 10m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/41029a7e5e574676a6d60b7530a847de : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f9ab370c76354dcd8cf0de6de125295a : SUCCESS in 15m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e74b611278234552ba94fbb0171a4680 : SUCCESS in 4m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8e416f7636d94d8d8eb8be30673fca61 : SUCCESS in 15m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4cb1c6ace9c64b1ca7d0f08a4d747eda : SUCCESS in 4m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b23e353fbcea4abc8fa68d0c68fd6cf7 : SUCCESS in 9m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a820651bfa73483387f35e7a9b33d668 : SUCCESS in 30m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c56a457c3d0d46a4b58308e9d67cf543 : SUCCESS in 11m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4d2ad5411d2a400b9e65bd400cf6d84a : SUCCESS in 14m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b8c6881a47e44dd0a1122b34c2b59ec3 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4146c21d8d3d4cfba96c7be3d6594b59 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b69e2c141054e8fb29e5caeb1762627 : SUCCESS in 24m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2ef33fed844e4f949f5c2759b774d49a : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6590d5716c5547c393c845e5bbcfcaab : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f45cf9e4eb74186873a3cc1328b993e : SUCCESS in 6m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c4ff6faf4114495090040a1ed16f329f : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a2ffd24e4864712b5d7bb8ed6577f90 : SUCCESS in 1h 06m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df267c694e5249f1aa0942f4dc93667f : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":1},{"id":"80bbba515f28255db442b2d6b6799e17c9ba898f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a67af9d7fe5c0c4a4d4dbcaf30e05119a51df8d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:37:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d12ccd1d041f95af98d4f7dba822c48fd8b1c7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 21:12:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24a3c6c51df04535ac7726bea228ff92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261e197fcb554eefbcdc8fe7e9afc388 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ededde375145c4aca50cb1db10873f : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93192310c8a745efa0165be8afc8ba9b : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15fe80ae6d6745edb2a232ac8131aa23 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b67103690f6742849d8ce0d72acfdb5c : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbe77a7dff5145e38d25589d3b62c111 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f3fd53454134ab795fe325ca0a99cbe : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/08c5170698c64f719b0b7aa9aab5d5a8 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9f5362b754fe42c483912a6696c56ccd : SUCCESS in 13m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/23203f1be2b940f8b80f53a760ff281f : SUCCESS in 9m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7c78a3e991d344459beb67fdd47c9bb2 : SUCCESS in 18m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2082cb20a30f4a0e8d599dda1bc1a862 : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0a0662cdf9644bcd95f59b3f036fc91f : SUCCESS in 5m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ae7a0c47545d4fba8562d6ee36c164a8 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/45f4d4700df74d42b72042cee3a2d21a : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91f00a3462224f0cb6c1e41f73545076 : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/32c49d7da0a24da7b72225236c0d8d34 : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c6846780d80b46849d282f4b26f1a4c7 : SUCCESS in 17m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f6182fa56c8c4d23ade65d0cf46abe81 : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/515ffdd466eb435084768b1c589f210c : SUCCESS in 10m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/60f1222ea768489d8637d312179d1321 : SUCCESS in 5m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4ee54c0e4fdf487fa7217f567299ed3d : SUCCESS in 24m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6b25e05d096d4fc9b39274ab900d1837 : SUCCESS in 14m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/153782ffda77486fb521a36283e190cf : SUCCESS in 11m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/84263f8d0fc7431f85accef5e8e9e7fe : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53088fd2bddd471e90dd498e65537bcf : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae5304aab7454b20a857d0e15ebcc116 : SUCCESS in 22m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/40972443dc4245cd96abd4a275301631 : SUCCESS in 9m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/77962351678a4a28be1de81de285d354 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/751eda09650f4e348c96c5749355e53a : SUCCESS in 6m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3ce4ba45a62e4d25b21213dce9e0e56b : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa36f9458edf433aaeb9b6481ecf6707 : SUCCESS in 50m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a82da5f65524a6aa24eacd134dd7594 : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":1},{"id":"99b6f54d1907d1c48262885386cff9ae24de4b5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 21:12:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0716ef1ced9e12dd06de4df4a8235bdea04b45af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 21:13:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/761fbad96ee24084a2ebdfe36c45f420\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db07a0ba9b24c678a550388cfa75b75 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"a133732494bce3616da5e04b682d4b86e4e58aa2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:22:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dce96c9246e924137ff633edb4d477123249820d","revisions":{"dce96c9246e924137ff633edb4d477123249820d":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:14:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/892944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/892944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/892944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/892944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/892944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/892944/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:14:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:14:06.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 3.2.0","message":"update constraint for oslo.privsep to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iafd6a831e2202e92d477142ef87d81a227eccaa6\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I695e2085ba58458f95c7806830a0ce6556e35c46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dce96c9246e924137ff633edb4d477123249820d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dce96c9246e924137ff633edb4d477123249820d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
