)]}'
{"id":"openstack%2Frequirements~892946","triplet_id":"openstack%2Frequirements~master~I3bca06b710f811ad3d7237f51b2b24356e7c37db","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3bca06b710f811ad3d7237f51b2b24356e7c37db","subject":"update constraint for oslo.versionedobjects to new release 3.2.0","status":"MERGED","created":"2023-08-28 13:14:50.000000000","updated":"2023-08-29 18:10:35.000000000","submitted":"2023-08-29 18:08:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892946-new-release","meta_rev_id":"281903811cbb5c35514ed534fbf86a8dbf02255c","_number":892946,"virtual_id_number":892946,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 18:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:51:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:22:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"26d872ce7e1da39940c4dee45a5aaea095893f5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:14:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea174feb7381ea8173b56220b9021b5620c79666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:51:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4bc3156ef034b2aa29f54e8f5ac30b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cfd101fef44ab29fb1a03eb4e7701c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645be49706024ce6b664f3ce632001ca : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eddf9aa69e4641e3872fd5d0e832d03f : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0517ced077734efdbbaa65a05a728693 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b43feb53cb604dafbda57745a16ce1c2 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f0c39dc011a44309b78fc933fe8ed62 : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d968e6d6dfc491397eb4319363b6714 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd6a9cf92e4f4f21b7b028131edf96cc : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/19e7928acc2345c6b3fc6783a258d8c8 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/99ccb41b951f484a9eb05b85001e3f34 : SUCCESS in 11m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0dc950db070426c9ca9d83b7921e496 : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/05cf1e707bcd4903aa355494750dfa2e : SUCCESS in 12m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f80a791793994389be15a859df4fe3c8 : SUCCESS in 6m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/89dc54c6384c495a8f54b1c1c2abf799 : SUCCESS in 4m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1e5cab724754447cb4e64e7938f816e3 : SUCCESS in 6m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cb6bdb8d55694baeaa34114c0247cec4 : SUCCESS in 14m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ede3213d6d1c4318a71f53286ce27b68 : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f6dfef1bde264240b442cac50ae3e554 : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4597d86b8de84cf7a7f9d6a09ce208d0 : SUCCESS in 7m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8cff5ea0c7794328a973257520185987 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1bc88f4499294c5cb7fb5b468abb3f36 : SUCCESS in 9m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4deeced020bb46dc8c59b1ba8b593450 : SUCCESS in 7m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/114cef91f2bd46909bd77c9e3879e5e6 : SUCCESS in 13m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4870c2882b4a4d1bad37738ea1617cd5 : SUCCESS in 4m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6a1b00d701234a3db21177237b2f0c5a : SUCCESS in 9m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e73200f09a164c288c40581357602c90 : SUCCESS in 4m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a18ad87e7f2b45a48cc0ad854a900d71 : SUCCESS in 12m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c6a8f5b258804acc871a4de79a40931d : SUCCESS in 27m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/359ee1901d9a4098928eb1c383caaa89 : SUCCESS in 9m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a15be59dc2904baead369f6fa5e01c6d : SUCCESS in 13m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ad258871aff04d18988c64c9cc588c53 : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e6cfe60e90a49c3a4a1c770e4ea48c7 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31990492df574260819e46669aa91e67 : SUCCESS in 25m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8d51361ac209473c8cf18b217edf6a8c : SUCCESS in 6m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0e03ffb98521425db47bc8d224805f5a : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/432ffbbdaf644baeac99ea7ce6272e81 : SUCCESS in 6m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/11dee8debdb0446fa16229ed3d1822e6 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fdecd826d674fc68d6561a720f9c886 : SUCCESS in 57m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d929539ed584cf698eb7342b9073bdf : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3444c383e6feb931f9404adf6b24571e96bc0641","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca5ee79b7bf179426631c93c89989c84b8e5340b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:37:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e94c2903c3d951507576a57a9dd5ab3a98ef1feb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 20:29:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75a55b0659554770ab415fd7b166c191\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bb2b1d91ad4e41afe24f345242be53 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032bb2c4ce2b4ba093de3c4be2c1867f : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02ad87a15fa84616936ee0662a4cbedc : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61f8da8bdef54d4dbd84f4737b290593 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ac922c59e8c4f03851f62e840da6677 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77f4a0af54694d3ba9e91a3b9115813d : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8241017a7ab4f8895f4d4918260064e : SUCCESS in 4m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af23aa7367cf4ac4b5efab5ced376e73 : SUCCESS in 8m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/26649b8758034e0ba8ddeec7383582fc : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8321811550414ffc90f8ffc04682c357 : SUCCESS in 10m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/66fe63d3576f4cdc848fd5003fc00cd0 : SUCCESS in 11m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d4b1f3ef74e744cfa81c28b97aa44c89 : SUCCESS in 6m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ac2b390f7dd747e582ed51dad044d888 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/214957b7b70a4553abbea5324b3dc9d4 : SUCCESS in 8m 23s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b6ee76f1c8ee4715a97be8803d8bd800 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03749c001c684d25bd0b07c8c56d021c : SUCCESS in 10m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/04b143485bba403fb59e0677f62a9488 : SUCCESS in 9m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/83c5ae5c7ea246079b25c8e7f6a0abcd : SUCCESS in 12m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cbfe7888f3884aec9c5833089da6b893 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c36c30c056694aeaa5bbfb7c99459b8e : SUCCESS in 11m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a01da6443750491daf61b16af31afc6c : SUCCESS in 5m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ec6e4d1fe4704c4f8eef81bc35263fad : SUCCESS in 28m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2e6932c8fae34ff3b32a4b4b8f0060ef : SUCCESS in 10m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c9c9ec9e03274f9b912059fac6ee0944 : SUCCESS in 14m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c630ce1db79e44069b98653b36eb13d4 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/519da2a55cb94822973aa242b3decd80 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/681a0154246343f4ab2be90f5266304a : SUCCESS in 22m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c08b5066f3364e188b111d8c120bf01d : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fd73448d37c84cf4bc3319e3aa00e972 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f44dce065b1424f98469a080167fa05 : SUCCESS in 7m 27s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9c66d6b802d047c58cfcb47ddd87a0bf : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c2f835a459143d492cf525c8783261e : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e96563221cd45e29a073dc3157cd0d6 : FAILURE in 1h 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"554dadb1b5f0e847afecb9c34017c122fe0eee78","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:22:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f860c0bbe3bde10986ef78f596e1c2c5aed4e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1921b43f28e4adea0055794efb9e0de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d395d670ab4643c8a35e7138b9459bda : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7adfd8df506e4e76a9b9c3f9498c767c : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0bfcece41174934b390e6f9d3da413a : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bea759fcde740a49201a4a5d61e3c85 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af6c5a3d538944689c8e1c5e542e8be6 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/828652666a294d9f9377126a4fc7361e : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cffd7d3d1db4696bfac206bb4ac91a5 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e9cb2b07547f41daa07cd173ea97fc34 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/13f2928713b74d6286c8878bd352bfb4 : SUCCESS in 12m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fc23405ff2074c709ac9e0f75550cfb3 : SUCCESS in 12m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d2385050ecf94b16a0989e022a3e55dc : SUCCESS in 9m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d8a545c988c443f3a339f68c46442ad1 : SUCCESS in 12m 41s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/948c48c42d514e48a322c56154882297 : SUCCESS in 8m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/13815b5f5a9d47abadb4233377ece695 : SUCCESS in 4m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c720a0874b8d47ca953b37f020a16567 : SUCCESS in 8m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7f32aecf0fb0448191706ee049ebf758 : SUCCESS in 12m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9a3a10dc1e434322913274ac895a1df1 : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/06d248cd28dc4cb59bb2dee0ec8c858b : SUCCESS in 5m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a27c842e03dc44a9b0adec7e6f560a1a : SUCCESS in 9m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/864534f73ce7457cbdcff8d272871725 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/980236b8240542dbacf48a821446e2a6 : SUCCESS in 8m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1e0746a2033b4fd28643f86df5212111 : SUCCESS in 11m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/db805349f07f435ca30d4e16691ac603 : SUCCESS in 11m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/30261c5450ff4d3fbd4fe6caf9aed18e : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d54f41f3423b4c94b24ed155df87e4d7 : SUCCESS in 9m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4c683189213a4a389b54b8a452f9933b : SUCCESS in 8m 01s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/123c41be15b0432b868a62b27c0fd2e2 : SUCCESS in 11m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/50d2e75cea1b4e129491e47c2ca87293 : SUCCESS in 26m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ca27741f2a9741f3898cdcd5ce50892d : SUCCESS in 11m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/156c65b5fad94747acb6f0d02a158173 : SUCCESS in 14m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b831583b88ca434fa7d301f14291704f : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/be39c7126d064ef2819122347b3113fa : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2693f6f8bd69472bb2e52be48c36d5a0 : SUCCESS in 25m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b91d227c0e75427e82882ba2138cedc0 : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/30d60fc61fc8439cbbfee2f0a3bb0f53 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/755e293dff124777883dbe74cdab1b58 : SUCCESS in 6m 09s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/198e82f3fc1346dbb6c6e0bdf20c77ed : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a26ca5294b2e4550bc6717e0886f8c79 : SUCCESS in 43m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98c664fb3b7e409ea9f28403068f89ae : SUCCESS in 1h 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7c7138200706841afb8da8885df5c8e0461b20b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:07:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6df4c9f3f710e0e46d895673b967c7a47f053bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:08:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d9184bafdc448daa75f67e83916e48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd799992f1d4741893660f58bcd88f3 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01cbeea461d5471892ccd68c98a9beec : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c8538686024b94968d7322ddad4b0f : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b336750104e4c3ab0331ff668768340 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3665cd3989043b28599b45f7c4282cb : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f1c9ece7f8d48a0877d83b99940a5a0 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/567316c3fa9c402da8c9eddc1a0ea8dd : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/133081ba42ce4cb693e3489daf50e162 : SUCCESS in 8m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ed89c1a45d6e4b01b879ad5f20169219 : SUCCESS in 12m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/77eb1554aa8346dab492bd81db1a0d0a : SUCCESS in 9m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5507d484670442e3a9adc12a2e488d59 : SUCCESS in 14m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ec4d4f06a8c54a639fe36a88f98516a9 : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e696965247a845dc90f51059e6dca565 : SUCCESS in 5m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/efbac7a92f91438aab7ce26ca4e02b37 : SUCCESS in 7m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5543271230c14e2d8a75b3a3723989e9 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/385465b7d5a34ccaa95990e236a95657 : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2f326ea5f97f4ce4a5da396aa61a44c7 : SUCCESS in 9m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8f6e0d2ef4ae4ec4969aa8b425547853 : SUCCESS in 10m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/23343da8b9394809b7bc49b54e708e48 : SUCCESS in 4m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3be59acdef64460abcce73fc6a2dfcd7 : SUCCESS in 12m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/75a596d6195644c69051033c0919f97a : SUCCESS in 4m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b08cf8375ef24262858d892e695fca7b : SUCCESS in 25m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cf1cacf7d66945859397a06080d67ff7 : SUCCESS in 11m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/de5cd9f3e5144c629dafa33453d5baa0 : SUCCESS in 13m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d96726c71e0c444aaac417a03222248d : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/10d5b7750c3143dc8e421c6f9c28119b : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1d0b65c202b4c1595aca95fe58e0820 : SUCCESS in 21m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5167b3c6ac014326bf028456274c883c : SUCCESS in 5m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/90bb9606eab34a349148a50b2654e857 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ee52381f06b4f4b99fd8f2d23da1464 : SUCCESS in 7m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e011d6856b30499097c5fab262a288a5 : SUCCESS in 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/248522b902de47feaeffdc5116627e12 : SUCCESS in 1h 15m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/add035987e3640fd9a6efab76735e760 : SUCCESS in 1h 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"49b772068e97baf1d2ba6f67f5e2cef3a2859b66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:08:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"281903811cbb5c35514ed534fbf86a8dbf02255c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:10:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46db964276854a50ae4bf44d4095a75b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f7823687ac4dd59eae47bc73f061d5 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5db5eb05a90b1a70af7d6701b4403a37eb3f7586","revisions":{"5db5eb05a90b1a70af7d6701b4403a37eb3f7586":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:14:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/892946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/892946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/892946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/892946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/892946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/892946/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:14:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:14:49.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.2.0","message":"update constraint for oslo.versionedobjects to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic312840b98fe818de4d21384303b5dda5b21d608\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I3bca06b710f811ad3d7237f51b2b24356e7c37db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5db5eb05a90b1a70af7d6701b4403a37eb3f7586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5db5eb05a90b1a70af7d6701b4403a37eb3f7586"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
