)]}'
{"id":"openstack%2Frequirements~892950","triplet_id":"openstack%2Frequirements~master~I6a4f14326b8739036f0230f456f8cb6c7df37868","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6a4f14326b8739036f0230f456f8cb6c7df37868","subject":"update constraint for oslo.vmware to new release 4.2.0","status":"MERGED","created":"2023-08-28 13:16:39.000000000","updated":"2023-08-31 23:35:36.000000000","submitted":"2023-08-31 23:34:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892950-new-release","meta_rev_id":"ced9be15db6ffbdefb43fc894b49e4f4cbaa25ef","_number":892950,"virtual_id_number":892950,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 23:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-29 19:33:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:46:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-08-29 12:21:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-08-29 19:33:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:32:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"c3e28231d76a34f2ece1fd5619cb52bf01cae831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2406b8857c2daedafcbca59a1a96c3e7e9caa8d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:46:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b4aceeec758411c8d47ff902ddeff81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d180f2a10c5463188cd2637e566f684 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b7a1a7c01e450782352890798f4327 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/046c0f910b934137a199d649ffb3dbde : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8f644f924334cf29775cb4408d5b4bc : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/774970a28a21428782c4addd05968608 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/871974613b94409b8d7520c310c4d73f : RETRY_LIMIT in 1m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9075644903e490bab3f874bea6cd5e4 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f3d7243afb9456cb5ecf65b04aa30ac : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bdbdaf65cdfa4721aef5115de91469db : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/badf4198272e45b9bfd608b5184966a4 : SUCCESS in 12m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1c75b64361b45518212f4578f5392a5 : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/345d8f8984c447ef9566c2e60396c15e : SUCCESS in 12m 41s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ec437dd355b443a19283c4bd089e212e : SUCCESS in 8m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/727489c1fa3742d2a52976d4b485d062 : SUCCESS in 4m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/405ce1bd6a304efd8c09d571dc67d9fe : SUCCESS in 4m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d71d8445abe24cc1bf88b6028b521c33 : SUCCESS in 11m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/95180850f6b74d5c9e7b738624640f56 : SUCCESS in 6m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ba113ea4e9854dc08efc6e0c323b7d1c : SUCCESS in 4m 39s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2258c998824f47f68003e2b75c42c6b9 : SUCCESS in 7m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e63872c271a44674ac4ba7c53b3d864a : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3e9e3aed4774b64ac3dc253b0e7517b : SUCCESS in 8m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3bc003d14b544793b4860d82418e6337 : SUCCESS in 8m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4bc107a446e542a6997ed42367c7b683 : SUCCESS in 12m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6fce93fa1cf2486d8953824a5936ddb0 : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7f90fb4a51aa402095b88876f5e01748 : SUCCESS in 10m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ab05529f4fa4414fbee4adb1c47d07d9 : SUCCESS in 4m 34s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3219ae6027d843a78071080321fd11ed : SUCCESS in 11m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fa94df4a3ece4f22b86d991de4f856ea : SUCCESS in 28m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/977f05f964e84f979e8a98366153152e : SUCCESS in 10m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/24a332a4f0734927b3979f02b08b66a1 : SUCCESS in 16m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c36e47b3b22a43b293fdd8d159a4ccb7 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6c8c0dd324042878fa1292f6f8179a2 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9a26b41c9ff4285bcc056742f5acb52 : SUCCESS in 24m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1e3d691cf4674144b221425926a1239c : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4ebeeda8eff942299d8bb7b2db45d239 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f0aba65e28840a595ec36d848ffc78b : POST_FAILURE in 6m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1dc96f1629e24f5a9ccee9a3232a1484 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02d4566013f94d00b197fdb135f1810c : SUCCESS in 52m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02f783e0e5c7409eab9d577e96d5ccb3 : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a411ea54020196deff9df72467b8f9c16777b229","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1da0814db3377cd3e1cd4e1376d5380e174cef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:13:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c0495948bf440cfb71a34e544a2eb3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7f146bb13d84ca6b4bd8a6838037417 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97370e5fb0c2452c95100fdccb20764d : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24a9780d04cf40109b60cbf25bfefbaf : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/582b3e3e865641cbbe78208f02133653 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83c7e36a30c74c9ba260de5011a145ac : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a4fa0747ed947a1b2217ea36aa12057 : RETRY_LIMIT in 2m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa63ca6815ce4e97a8389544721f69a5 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5259ce92d07e4441a5345c315a8da1d4 : SUCCESS in 5m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d80a90c4cbd4bd2a7639ee7166a5391 : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4e582617023c497a8f74d0cf5aedd3f4 : SUCCESS in 13m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/844ee3de23144c88973dc21637f62387 : SUCCESS in 9m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/05c05a18efc94ee8948b9528fedac857 : SUCCESS in 12m 38s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/38365c3893d34babba4a857477ded044 : SUCCESS in 9m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/76b95e30a0824fbd846abb016c0aebc8 : SUCCESS in 6m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/20cbf4f01b3f47669a14affc00c6b47e : FAILURE in 5m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c5656ad4941544f799366402d8d51a0a : SUCCESS in 12m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e24e9bda0ba54ac183c7433a1eed030d : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9545e2cd42ee46b5be9528a1e0a64dd7 : SUCCESS in 6m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/76df14cda857440a96c25273f34f2413 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/48aaccd13b9b4c3f8da6a1ab5f854d5f : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/389d3f4a5c4c4e7097a95a5fc1301fdf : SUCCESS in 7m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ed54a4da6b094f12b05eb1fdedde81fa : SUCCESS in 8m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/99e2379624cc432abd4076e8ab04d45d : SUCCESS in 13m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6c8e12f2f19649099f9eb33743c29a52 : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b2aa2d1a64ff43388b9c365f3da088c2 : SUCCESS in 9m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5bb8dc011c4f4786b40b308c98c3b0c1 : SUCCESS in 5m 11s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/20ca98e48b9c44769fd161159f4e0306 : SUCCESS in 12m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6ad095ff7ed940c980e1807371a2d061 : SUCCESS in 33m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/97fe4d4b92d14ac387b34bee1f491425 : SUCCESS in 10m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/64d362b89fca4dfd83f5bb3c8a1037f4 : SUCCESS in 12m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bcae3cce9ef64b608b20e42837f3df00 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1631fbb6f42340cfb020cdc6fb3525f5 : SUCCESS in 6m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1fb1c430a1d4d988ad907e0b239861c : SUCCESS in 21m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a902aade03ac4525887d22f5660c66bd : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f28210cd8ce54c54bb2099f2dfeb7cb1 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08be2b75295f405086150e99ee709612 : SUCCESS in 9m 58s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/48a6e3b20363465c89320866d1477eb8 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/856ff91013d24cf7a46d88b3fd951397 : SUCCESS in 54m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3452687787df4994857b317ff36e5a16 : SUCCESS in 1h 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"992e2d94fef1ec16f2887d67d092d85632f6d541","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:21:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a617ae299578ee5d823ff43f759871a9a4d71308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:18:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5236edbe348545b28755ec1da5b1e73c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616d59dbc04d43cfa7b2bf5f1bd47a68 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78430d6e4a464afc89f418e383d165cf : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93adbb33bd404aa9bb7e686889f1d26c : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c473dd100044e94be1817ccecfb4f01 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e10956221b32464b8576f4a06ad1a055 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c6a81084b994a7ab4817e6bc79700b5 : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2516ad91149245938e7b0daaa609889c : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d28fb3b947c441f99da1dcadef411a9 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ca61fb93874e4967ba8ab227c6726f1a : SUCCESS in 9m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4b2b31967d044ff6b17779caec9b9560 : SUCCESS in 13m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/23e35c460eb343b2ada28112e1f161d7 : SUCCESS in 9m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e07e2c369f6945fea9403b4dc15f462b : SUCCESS in 13m 14s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d37177aabb01417d8c461df24405e3ee : SUCCESS in 9m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/deb213a6193b4067b2b1b4198d9f2c17 : SUCCESS in 4m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c5df0cf0b6a04ac6a169381dfee7325c : SUCCESS in 5m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3555a70e596446a0936f673e162200a1 : SUCCESS in 10m 53s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/43dc5f23a497451997ed8d9d96a74250 : SUCCESS in 5m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ff6c5f3970f04a22a44ae2b72bab5ed6 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/42123581e6434df383b3ea8d297b6b92 : SUCCESS in 7m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9c3ece720d1e4566a69f582342bdc670 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94bb6ff5f4344f9dbc3508caaa393f5d : SUCCESS in 7m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7af263d00178483f84db75469f46f091 : SUCCESS in 7m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/26ae211c60394611a18ac231abf8dac5 : SUCCESS in 11m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/95dd77b25deb4e7383d85dda94db5c83 : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fb51d12b89434600bfb99728f7fb8840 : SUCCESS in 9m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/afe8f35074f54e689c48911f2d24b380 : SUCCESS in 4m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/11056ed94b7e452aaa3d0f822c1bd66d : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f3dc83c0cb9e49cfa28419bba1444590 : SUCCESS in 31m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2445ccfab70748f89a7902233067e786 : SUCCESS in 11m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a3afb05c633d4d1488946303e1f6ca37 : SUCCESS in 13m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bb5c8c0975464d4bac22dc93647d065e : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ea7c4e5acfc4afdb13db507145a2a80 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b47774fcb564cb3882646b703a26c7b : SUCCESS in 24m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/beb77ed8c17e4ddb94f8da4f41626241 : SUCCESS in 5m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5cef23383b9c4bcd8888ca97724cdb0e : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a8037cb336447aea0ee9e105bbd630b : SUCCESS in 6m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4cc6b54cb21a4b7b985148fa89a6d25c : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3e3ca18c2b047d7ac4a9b84f4f20fec : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ad5fec7a135491984d904d2f1e3d9f5 : SUCCESS in 1h 48m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1b88c0203f2503b7417a73aa7f08fcfd2a3017b5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:33:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"502d04da3e0c064ee87590e6124a196dbc6779a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:34:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2482387cafdceb5c323a763e19937971e88bb495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:16:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/feefb562e31e41a78327c1a3433fd63e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b07ff298b1bd477da7041415a488765f : FAILURE in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/446b1a835b94419aacae6f3a6a359147 : FAILURE in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a0fcd655f7941129050f6307628c530 : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9efdc7498a504a50a07f696a3f101baf : FAILURE in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ba7d72b4f9b493baf440871fc2c7995 : FAILURE in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/368410e4018a46aa844bda5ef8405493 : FAILURE in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8063d196579434480d4c08d5f656264 : FAILURE in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb3a126f4ebc4df5a5cd396bda529d77 : FAILURE in 8m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/282f33ed3ac04898ab34cc5cfa78643a : FAILURE in 11m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d8ff24c333074a75902e836b7f620da6 : FAILURE in 8m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a0f93ceb3f994b278d7bb98a33a47697 : SUCCESS in 12m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e2941d14fe29420ca87a1ad65de7639f : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f3f049f200df4ebfac8ff1ceb0da1210 : SUCCESS in 6m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b64927135c874a979096a2ca75d8bb67 : SUCCESS in 8m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/225e0be2be594972bf3993fcae2d6ad1 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04bea5fb25d8456fac8088d06e7b3602 : SUCCESS in 6m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/58e4708ca58f4c7093a387bde445ff10 : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/26bb6605c794426eaf700b42e1d28138 : SUCCESS in 12m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f7135d252c5940bf93aa0ca953f30778 : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d9b1bc14bd3846f0958d791d25a71577 : SUCCESS in 12m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/584a39ba0eee4f59b9d734e458a26260 : SUCCESS in 5m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4f62aef7ea774881a1ac2ee732f34548 : SUCCESS in 27m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a6313998c8c04bfab417d797f73cf1e3 : SUCCESS in 11m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/48636853ed9f459f871980e8c63f204a : SUCCESS in 13m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d718f1bdd4ce4271a7c054a24d9f7200 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5729b781c9841faba929d894927b098 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c70051bc0d3c45c3b65d30b42ca0200e : SUCCESS in 21m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cd7eeee5f1d24787ab50b13d3ae7c021 : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3af0a5299bf54fe79396c8c9513dfb63 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46fba1836643462face8d1f3931eb51f : SUCCESS in 7m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0ad8f4f1d3394041823a814da42eb8e0 : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49028d08b4c64c518d05147be3a36e13 : SUCCESS in 48m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0aea73b083774d33bf18336c752b9e8d : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"738b96d4eb38bbf86ca43bdfc57ce0923933624b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:32:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2687f02816687f7d30d59e02592b0354a1194348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 21:42:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a5b9b9e7d1c4b7c8bba5b9b9b72f047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f28a6a4c214df5881a3638a8756c40 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed0cd54acae4c32a1a99c7285a60fa3 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77c1a47df6be4cd3a27f3009324a17ea : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8710fe9a3d948f9b0194a72461f5631 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84d5088194e74f17993b7deba5b61986 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ac2abd7927343ba925899872bc3d9fd : RETRY_LIMIT in 1m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b1064f4fe514a41b00e1c195a760f2c : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16a2c1eb730c4590ae43e0a280277275 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46131795d7774041a705026a94f3f663 : SUCCESS in 9m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/db9a90afceea4f26893aa48f5a23e2b2 : SUCCESS in 15m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7c9fc74149cc469c840c929cfa996b59 : SUCCESS in 10m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/72437ff842a547cea4fc051a4990aaee : SUCCESS in 13m 47s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c84d103af88049fc990e1c7deab6ba7e : SUCCESS in 7m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/05a9bcf24c784c0fb87bb8a20ebd0aa5 : SUCCESS in 5m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f32d8af985794fbb855ff3daca55c163 : SUCCESS in 5m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e5f03492183b4289bab258de8594ccaa : SUCCESS in 11m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ccba5965010c42dfa3b89bd54d04a1c9 : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/94ed1ff5c56047538b80796b70f551fc : SUCCESS in 6m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/82d0076f941d4809b5b9b5555705399b : SUCCESS in 8m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fbb3703022814d199ba4ae702f86c3c3 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db9ebf93a7e94b4f8d8ad709fb0c1511 : SUCCESS in 9m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/219061604a64465ebdda554d5ba3f738 : SUCCESS in 9m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e1f80f31e1fa41bb90546eb26e394210 : SUCCESS in 11m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8d856a4371854087accc52498fcbf518 : SUCCESS in 5m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/800034b113cb477b9ad1d6440295e3e0 : SUCCESS in 13m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c2c2ba6eb2744de5a31090eace80ffae : SUCCESS in 5m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0fdbd44e925747a4b3578f4d79e8a517 : SUCCESS in 12m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1872e955d8574fa791c21f6d0623655e : SUCCESS in 25m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d2338d01225a4b20a09a90927421cfd7 : SUCCESS in 13m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d37c0412290840f0b05bbd8fa90ccb6d : SUCCESS in 14m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7b7b9392f10b4fc1b78d1edea4e45a60 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6c8e81f11c54f0ab8d0ecb0c5150d38 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/121df2bc164247b99ef4f9e36710d088 : SUCCESS in 23m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/42f2042bf5ef4f8499a47b05158635dd : SUCCESS in 6m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cd12181dfba44185b02fed433e981e54 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb0ada11a0504f9ba65d04364b468290 : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0e4679c2eb324a058a12ead835c6cf86 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/573c274d504a4097a61105572db59baa : SUCCESS in 51m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c434ee7e3e8b4d1ab8f612af96f56761 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":1},{"id":"67c1bed5aedf82699b16c16df50b3eac81f08789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 21:43:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e74198e1847b8a73acb19ea299f4d1ac49b3044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:34:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894e3fdfdb2341a281bd5d05af2c7fbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167ec2bffe3941fc928e4ba16d4f242d : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d7656604a744f9b7b46db477305b5d : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8346a36dd1c44bb8a0db49510d7e5adf : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e839b9f4822e4c99aa21e55b618c3741 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d240eb04e9874e048a6cc83554d01cf9 : SUCCESS in 3m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca00030651e646b9b22645d65ff695df : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da974d1105d744dd8fc4c63cb9e25a77 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ed9a9543e964c7986fb4c9d9d9869a6 : SUCCESS in 9m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/88109d774b7547318b52508ef8010dc4 : SUCCESS in 13m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/535b7e6662f94a0490867135b9ca7027 : SUCCESS in 8m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b03d2d91301047af9044ff33856a4c1e : SUCCESS in 8m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fe25ed480c3f4952963466fb583cefb9 : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/537d987fc788464ab78210a57ee6a9e6 : SUCCESS in 4m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5639cbb3fcec42809b7833618fecf16c : SUCCESS in 7m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9786fa2f4a6e491aa37434e85487f15a : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77b8da4644d9453ca8f3edb513ecc069 : SUCCESS in 7m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/735074d7ba24488f8282aab20efa798b : SUCCESS in 8m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7ccb8a83ab63458fbc25db13b0f1f109 : SUCCESS in 10m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c94ecaf52343492cbd1f87d0caeccbfe : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/488c1a2bef33400190e257d9f5fa75ab : SUCCESS in 13m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d18ae4d9b9284cd184d21fb7cb72253b : SUCCESS in 4m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/28f140caa88c44288e519de20473eb93 : SUCCESS in 25m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9cb7f9370f0241309d2b7371da40aac6 : SUCCESS in 9m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/34043c744f36402d9b0266ba1f235541 : SUCCESS in 10m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3d9f1f7dfc2e4129819747b0688e5933 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1cdb95499ff448a2abc12af727126a56 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2229d5265cdc464cad94822c3d5ea429 : SUCCESS in 19m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/710340fc2d3241a0b43ea9dfce53d44e : SUCCESS in 6m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c92e3e7105664cc2a5e18126923a5420 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/197e91b1f03749a4a167f8163cf37ab1 : SUCCESS in 6m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/965fb1d162694d2682fbf96920af394f : SUCCESS in 15m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9ddd9876f594cb0a72cff2a16b28b8b : SUCCESS in 1h 03m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1010b8f70205485b94d688ea5ca3d458 : SUCCESS in 1h 44m 11s","accounts_in_message":[],"_revision_number":1},{"id":"70a00daa519ba319ebf3260947de7f353ddf76df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:34:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ced9be15db6ffbdefb43fc894b49e4f4cbaa25ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:35:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d4e3c7439e44f1581153154fa315856\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b622d2a9d9424501b89034f2b3387010 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fae081059a6578737bb22887e941357fcfaa9f6c","revisions":{"fae081059a6578737bb22887e941357fcfaa9f6c":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:16:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/892950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/892950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/892950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/892950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/892950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/892950/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:16:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:16:39.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 4.2.0","message":"update constraint for oslo.vmware to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If63afe5c5e5b77e5dd50ae9b4253d86c0d314f0c\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I6a4f14326b8739036f0230f456f8cb6c7df37868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fae081059a6578737bb22887e941357fcfaa9f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fae081059a6578737bb22887e941357fcfaa9f6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
