)]}'
{"id":"openstack%2Frequirements~893035","triplet_id":"openstack%2Frequirements~master~I89897c387726fc27f042c1d9e5d01f1dcadc74d9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I89897c387726fc27f042c1d9e5d01f1dcadc74d9","subject":"update constraint for taskflow to new release 5.4.0","status":"MERGED","created":"2023-08-29 11:31:11.000000000","updated":"2023-08-31 23:55:19.000000000","submitted":"2023-08-31 23:54:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893035-new-release","meta_rev_id":"dd5b43cf948215d5646f46be4a619b466fbf9d29","_number":893035,"virtual_id_number":893035,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 23:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-29 19:32:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 12:20:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-29 13:14:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:32:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:29:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"148c0938fbe716f20b2e64da8800f535fca0e195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-29 11:31:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3d22de878e042c3bcf1fd9b28923be9232d0435","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:20:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78a9abef7c27129e517393bbeb650ece5e77989e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 13:14:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe68626fcd284366b90dc9ed46e2bf95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3530a143d5354b8093c5606528f3aea2 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4685233f208e413faf621aaa4353566b : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fb01359567840c7a4c0b1d75df4e41a : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6ff5e1a1c7a421c976933a98e5e1967 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47d86297028b40c1bfa28d24603ea5c5 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50d7cddc085d470d97b5bfd2705ae47b : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ce5c817cef3411da543395872b4cc52 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b0517b1431649f08216ef244f8e348b : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2d604a96b97447fe91bc729bc497781b : SUCCESS in 7m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2be2f979ccdb4427bfbf39d64b9f4f44 : SUCCESS in 12m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/16969cdf4c9a4a2f81250a8de5a446d7 : SUCCESS in 9m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0a6aa18e1b4f4c6fbbbdec38b314c2d8 : SUCCESS in 11m 06s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c70d0d4884bf4b1cbac9c2bb02708dec : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1a550a29996c4e32aea9b74607c4e39f : SUCCESS in 4m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5bf1c1c602a6436ba1743cc468aaf10a : SUCCESS in 4m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0d288dc564d34e9b93f21b2b7672b95c : SUCCESS in 12m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e5296d2158764ba98c71d0fa5b45a8d0 : SUCCESS in 5m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a1757fa09f1641c8b46da1d8c6a17e83 : SUCCESS in 5m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cee56e8054c8404591034f1708047dee : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6069e3f2afc748159f98a77c01b64c90 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e583c182c18439f84077630e264a865 : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c8b13d9492c84e43b185956ede2cf53a : SUCCESS in 8m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/35e8b02d35b34b44a586e1e7ae9ed58f : SUCCESS in 15m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f19efa5b172b41e0bc4d5fb7f880da5f : SUCCESS in 3m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6d20d10a0b9641e1812f558c1dd3265c : SUCCESS in 11m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/11b23d3f72824d7aa49910e6523ae427 : SUCCESS in 5m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/893be4b43af34b4695b838f4c6d4612d : SUCCESS in 10m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/79856b292e4943d8868d828df36c6852 : SUCCESS in 26m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/62aa85461a184159a1efa2445297e4ed : SUCCESS in 11m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/282f66d262e346f6a7dd7d750d9aee6f : SUCCESS in 14m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6266974f074d413fbefdf9aef2d5d42d : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0317e5d86e954f7eab0232b7f2dcfba9 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae126566832142afb8e33ac7da2b2e4e : SUCCESS in 24m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b7f5d5548186422896f2c8ff6801e7aa : SUCCESS in 6m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/318cee429c2047d2a847be6cc33511c7 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db4121ded93e402ea4d34ddb85ffe404 : SUCCESS in 7m 32s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9bd62c4c739e4602af50f32c0ae99d35 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9b4de729bdc40299579af5c2c99ae6a : SUCCESS in 1h 01m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4db4de68d3dd4c50bfe0a02c94fd235e : SUCCESS in 1h 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"51eb437b478d9f392809acaa7bc34eeabc3fedc6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:32:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8e3c711cc7b9dd704d87b0a1071fc0422e27951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:32:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f87678a4947361c82aa62a8845803e3ea2f958ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:01:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e20a3ee30fb34cf2912167d1fb17da70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/956a9f0ff00848a28636935103b282fc : FAILURE in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba3fb1e77ee46b683f28a5a20d9d7ad : FAILURE in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0990f86df6d4219bbfc44fb0980ff5d : FAILURE in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a69e13df1df4e6cadb833aa1f18c123 : FAILURE in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f378d664e3046a4a7dee53159501431 : FAILURE in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21e71c908a8840289cc9f738cac475d8 : FAILURE in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d2f316d8ef140ff8058b8d334037483 : FAILURE in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4905d30cacce461bb023e1cc8a8019ed : FAILURE in 7m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9c07ad5cda3f442da77db71b948f205f : FAILURE in 13m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/410dd394b3e44381bf761d6b59b13dd0 : FAILURE in 7m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/842eeb0b3cdf44158f12fe8bd76abed0 : SUCCESS in 13m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9ae438eed0414d55895c97d6e5bcd1c7 : SUCCESS in 6m 35s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/daac95b12495415e80471043665cfbfe : SUCCESS in 6m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ae88653d4eda4684b7a6648dd50ee5b0 : SUCCESS in 8m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/52a4360d26524bf99493debda863d38f : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef24d7a7d3a244559ecd2abf77b4d683 : SUCCESS in 7m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ebb7cb9fd15e41e6ba5995d96b3d5d71 : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c03d8bc06ee44e5eabce23794713d3ed : SUCCESS in 11m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4abc8ac592eb4984986cc96ccabd70ed : SUCCESS in 4m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0cd64433c52042c79998c85fd15f8f92 : SUCCESS in 9m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ca1bb761299b4f13a141468756799d4a : SUCCESS in 4m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/84205c464bbc4d1aaedf20fb4da97fef : SUCCESS in 34m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0f086726ee6841489a0fb5e1ca34d68f : SUCCESS in 11m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3f7ef369ca7a450ab028c23f53661c25 : SUCCESS in 17m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4d193b5e0ee244f5bea696792dc0b749 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2200c4a08c7a4f3980712a52afd25865 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59a176a8318c47e2834552f76e1b8a09 : SUCCESS in 26m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/48a19dcb0a7b4628bfdd712f6360759b : SUCCESS in 6m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/94fc8985bea44fe9ba77a360669f65d0 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49dfb37093814ef09a3c5a4f8b45955c : SUCCESS in 8m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/34a83e3c8e1d4a0b8a5ddf5ccc6c838e : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee5841a01c40408eac8b333ffc962cfe : SUCCESS in 1h 05m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f1a72d8197d4e49bc9b42cdbe579eed : SUCCESS in 1h 26m 10s","accounts_in_message":[],"_revision_number":1},{"id":"22c75742cbad95fff11618ef7f05e6e9113e61b5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:29:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a6e3447ac01afbfe44b39c09dfb8e2b54ab153d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 21:54:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dea4068ba0441ec925899f0831614dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062a50920c5e4c7da9997edddf43f99a : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d195f22fa1346bf818a13612c9081c0 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0d50f42bda14556bed01f7486d05959 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6463efe4be0a4a5e95e342b7c0cce175 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c30d6f59f23e423990aa5568fe7cdd9b : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/595a3e47bc344e2faf2a570c9af878c5 : RETRY_LIMIT in 1m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52b543850846476f8dd9e4365db6b087 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bef4990cd8264922bf36ed09d5f62345 : SUCCESS in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7bf9e00eb1654da28985bf6d69ad9438 : SUCCESS in 10m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b9c802c5b7540448aeb94b7ff09f2ef : SUCCESS in 12m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fa13a031d9694574a714f81275304a86 : SUCCESS in 10m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7897308dc5fc4a518b6f3956a75ae977 : SUCCESS in 13m 34s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5932952fb3cd43ff960ba0164e3d08b5 : SUCCESS in 10m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b01668401a3f4bb4ba05308b7e59124a : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b1edd115fca14a1b8e6241ea41318c8f : SUCCESS in 6m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/79c1310c8cc54e6ebf0673d4270169bc : SUCCESS in 13m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3f2d0063fda4491fac74c97f576cb20c : SUCCESS in 6m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/87c202bc695f434da0412fe7733cf89c : SUCCESS in 6m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7ebc96aad6cd4094a7ca23ceb241bc64 : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/44a6e742bf8c43b28e3f4a9333197019 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f930e061aa04d7aaa61669e60ceddbe : SUCCESS in 10m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1f956052cf694a1fa99c2e477133d70d : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/650e9b28577d484e8cedd8b28bd9da3f : SUCCESS in 12m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e065978aa37047bfb504a16970e9ef00 : SUCCESS in 4m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1930683ae89547fb9a6b6ab8668beef4 : SUCCESS in 9m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f2cf9ef340924b14a93fe833025f7e37 : SUCCESS in 6m 09s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5487d49c92834d179d00badfcc22e661 : SUCCESS in 10m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c1fba8ed976d41a3afccd4c8e20d9766 : SUCCESS in 24m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cb3a3f4941554ef5be231e5f16267253 : SUCCESS in 12m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e01c6c802cbd4f319aa0ba011647b6f0 : SUCCESS in 14m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a7bad67a204a41f082f8ba6743fadc15 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f2f3b7c7f19146b7859a90a46773b6b4 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97214490a8234dacbaa4ffcf1ffbae8b : SUCCESS in 20m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/724e18c8f0c545de9ae6c74c80ecdfa0 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6353dc5b231d42229fa204dc91435af5 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e5916a84f654d309789a727c35af36c : SUCCESS in 6m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/37962f95369243cb95e7afff8d5fbd84 : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7df1cb6097f647febefb0b095cce5c06 : SUCCESS in 48m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0f008b2c4184010ad5337012ced7865 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"600c974304c511979989ae1d3f5df3bfdb947aff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 21:55:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c35546a7d95469ec5fb71075d39ff63cdce9e808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:54:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e0450ccd404af8a37bd612eee591b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deed6537820246c893bc42b62f16dcee : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d2e8e684c0d4039aa5203b74e7b40f9 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4837743696b48a084718e6fa12573b8 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcf5ce049c6e47e78b02e9742face490 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36962926d07e433ab9cc94fd70de5b99 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0355ab1864dd4a158872cc65f38ad379 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b62e117d436444e5addc26cfb81fafd0 : SUCCESS in 3m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a063663b2fec486fb30aa1e7797c785d : SUCCESS in 8m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/596d23514d87421297c861721baa2f97 : SUCCESS in 12m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/758d348d31254b508c590c869e64af95 : SUCCESS in 8m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a7d185fd2d454848b9a7bf73aaf0728f : SUCCESS in 12m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cfaf1cecd7214fe9946ef071c0222629 : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0960ac7c004c4ac78636b89385113d2b : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fcb2a6384a0a462295d34dd5913f1342 : SUCCESS in 8m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b8c1c753cc80406ba5563b15fbf63534 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f63e99f423d94445b1e5dbff6908f95f : SUCCESS in 7m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/26b90e465e10489380551ea20282abf5 : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c9b32c898971426ba418a64088324668 : SUCCESS in 12m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/840d92ebfccb48fa9d500d52d9074771 : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c9618a2b47f44c318a2485b0c4657f12 : SUCCESS in 10m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4b309d074e15452c921f7ea300cc92c9 : SUCCESS in 4m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fd530dd8809e4419aa64cb3dca7474fc : SUCCESS in 28m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3d2677c7970a43cb9effc57e97a8a3ba : SUCCESS in 10m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7a71bd34b1d947ceaa6e33863d2c3abd : SUCCESS in 14m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b750ddec1ff447a685b9600b182d5413 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dbcc96190cd944a38903479930052471 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3425c12e56be4d49bc4fdc9d3065407c : SUCCESS in 24m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bfb4daf052c0417ca6540ab2ffbc2dc4 : SUCCESS in 6m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/94bd0ca44a51400ba512ad3d3de32071 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbb960be57d64ed0af44bf5a1dd93212 : SUCCESS in 6m 56s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3f5eb72b962c418c823e06fa3624e29d : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a54805c7618145d89b437487be69ee6f : SUCCESS in 1h 02m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4029bb5f113d42d1835f8853a76e39c4 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4e4fbb0f8b3e2870479893c4ae3076271ea5d22d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:54:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd5b43cf948215d5646f46be4a619b466fbf9d29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22a1f57009f94e079ca298733c98f356\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c656697d4546039ffcaa3dec38c3e3 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97647e397638e1ace502cfe11b3a0bf789078d3c","revisions":{"97647e397638e1ace502cfe11b3a0bf789078d3c":{"kind":"REWORK","_number":1,"created":"2023-08-29 11:31:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/893035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/893035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/893035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/893035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/893035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/893035/1"}}},"commit":{"parents":[{"commit":"130d068736a6e4ac6039c3d3520d1d0e712ba0ee","subject":"Merge \"update constraint for taskflow to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130d068736a6e4ac6039c3d3520d1d0e712ba0ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 11:31:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 11:31:08.000000000","tz":0},"subject":"update constraint for taskflow to new release 5.4.0","message":"update constraint for taskflow to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I68234f7eb2281729d695984c18ee0d42718c753d\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I89897c387726fc27f042c1d9e5d01f1dcadc74d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97647e397638e1ace502cfe11b3a0bf789078d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97647e397638e1ace502cfe11b3a0bf789078d3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
