)]}'
{"id":"openstack%2Frequirements~893036","triplet_id":"openstack%2Frequirements~master~Id68179c3d725b7eba7c4e68efdbc2a1406ce6d7e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id68179c3d725b7eba7c4e68efdbc2a1406ce6d7e","subject":"update constraint for glance_store to new release 4.6.1","status":"MERGED","created":"2023-08-29 11:44:05.000000000","updated":"2023-09-02 04:31:40.000000000","submitted":"2023-09-02 04:30:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893036-new-release","meta_rev_id":"12a370939f3f260a0e80cb27814cb4d331eae252","_number":893036,"virtual_id_number":893036,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-02 04:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-02 01:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-02 01:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 12:20:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-29 13:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:32:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-02 01:55:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1b0bcba74775e917d3cf377eb375d8a44a7e1617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-29 11:44:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9aaa5ca7f3836aece8f036723f35f26fb482a81c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:20:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8fad1913bb26c2881637736143c1a4e6cf4439f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 13:14:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01ac5d1931954139bb6c61b5cf492742\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b358f1f53042709a81ded22387a264 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3a5a271dd9419e95c5a59eec6fd652 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ad35b35eb82407da97b6a52ee296db6 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fa23fc085f14d148ae51b6b58d989b3 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a5ddf6299fc4aad882ab0d523396c6d : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9cca5f45830f46aca2bf75c4d291a6dd : RETRY_LIMIT in 2m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38e48cc39d9247c2bc24973492d6fba5 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94788b03495a45d0aa62c95a2cca9450 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28db2e52e93046fe90efbde4b1d9b62a : SUCCESS in 7m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/db851adc4ca2436a89cac98db607358a : SUCCESS in 12m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/36c050e5d00f45e69504a866e0f01e4f : SUCCESS in 10m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/499ec31dc4dc46a2bac8825490e3a0d8 : SUCCESS in 10m 01s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ee2b3835d3c64e49a7728ceeaf333bea : SUCCESS in 8m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/292db7f460b04e438df5980aca809b61 : SUCCESS in 4m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5715d2543f18405d8350207d530050e3 : SUCCESS in 6m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3032025ce3754d3f946f375eb2fd10a9 : SUCCESS in 11m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6e19dcfdc21e4cdd8ec43ca7ba0eaf15 : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a441489d6bc443c895fe96aedc9494d3 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1f5755c462044fd1babd0fa971f88ec7 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7ef942376e3c446187213ac8d62ca72e : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83f7adae2dfa432dae15f939b14fb362 : SUCCESS in 7m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0b0b111e963e44d8ad282c2055064445 : SUCCESS in 8m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/45237d153f6046ae9498064202884e8c : SUCCESS in 14m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/21adc4b47dd043c9bfd1a7584944e9fe : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/343a37de02034395929c2078f854b918 : SUCCESS in 8m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/194d096b2f144befa975ce36e49399d7 : SUCCESS in 4m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9a9ba434f8b04c88b33c1e0035853692 : SUCCESS in 12m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1c9c7ab2d6c44e30bcf4a94347482394 : SUCCESS in 28m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/094ae17631c645709fc6c69c078950ef : SUCCESS in 10m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5c89d4325feb4a45961b535b6792906c : SUCCESS in 13m 49s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2a30fafdf27b4925896446f8d893bdc1 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f3f61b61db4f40259e7540853a3f5db7 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f767979f8e074e0d9e5d92de54fc499b : SUCCESS in 24m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/011226b5bcce40068dc6d9d1d3edac33 : SUCCESS in 7m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6ea06b0e97244256baae05030a339858 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a90d422a78e64dbc8082c3b7b909ce0b : SUCCESS in 8m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7b8a20f8726d4814bba0020276e2846f : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45bc6b2dff48447c94a333124e1d8104 : POST_FAILURE in 1h 00m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25470c71b5b849f980ad357f04c5178b : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"75a0f56efa074d629a894be3f3fbc778be083c53","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:32:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"186cc6e38fe46eebb2172aca7c1e1be9f5a5e183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:02:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c781fddb4a44997b39d03f8a01d0b46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db21ad05c3944abe9685f0284fd7fa39 : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97729b878a0f4fde971c275671bd87ec : FAILURE in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/534462e1ceec4504822596c0fbb2e982 : FAILURE in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92d85f30aa2e42049aac87c455a2f8c7 : FAILURE in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a31d07ea6774ba897725ee24ba58969 : FAILURE in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cb8d378eb1946a08a16656e07693d79 : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6523beb7885c4ba0824bfea1f234932f : FAILURE in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdcec96ce0294e1798fe07ccaefbcdf8 : FAILURE in 4m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/611d2be8f5da42f9af0303b3b598a4c6 : FAILURE in 7m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3a76210a2ff8484695952fd783214482 : FAILURE in 10m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dea11c4402354976b05818b87d58146b : FAILURE in 8m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/efce538b5ffe43ed9723ba361e58a426 : FAILURE in 11m 15s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6a6c8116816041748832d089cd8ac1c2 : SUCCESS in 7m 16s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d0f65c8030b24417bba8e0b63bea9e1a : SUCCESS in 5m 39s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c08ad7e75bcf4d5d88d2cdbff489a96e : SUCCESS in 4m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/832223d705e046578f319c695c5588bf : SUCCESS in 10m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9e351f79aff7431db58b8469c05e1a23 : SUCCESS in 5m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1d8077d4b56f46f790bc8d181fe5f67f : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3343c14f8c7f4ce68c73e0a2293778ec : SUCCESS in 7m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ec06cd8569eb476d88dacdad8fb9bdf9 : SUCCESS in 13m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cecc5da581f44d039be4c9be13962331 : SUCCESS in 7m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5d79fdbb06a04d699f2bf9c6dc6e785a : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0926f828042d47e0a6a2774fbad0ab72 : SUCCESS in 15m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dd92a1c4fba54e1984d8be2a65da102b : SUCCESS in 3m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b325f20dfe72487e8678d2c608f2a9a9 : SUCCESS in 9m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/66cd6c356fd2448c84fa1a3015ebf4ce : SUCCESS in 5m 08s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4ebd4e071a604929b955bc56b17543d4 : SUCCESS in 10m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b384468b5aee4612a3eb3db528ed83c9 : SUCCESS in 26m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1ac7335c5192443b9ae9db7308309ce6 : SUCCESS in 11m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/53f9aa4690d94d76beb4fd72052f944b : SUCCESS in 12m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/059f5a13d7fd46a59dbe17d1c08a6fb2 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9498186a05a471695dadf79c3bc31d9 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38fafff5214c47059859b3bdeda36079 : SUCCESS in 22m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d29559f5e13f420c876b4bd5fa91a9a6 : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/68fd81f66e1b47cf83f1f4cb33b548a4 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cce66f03968e46389360b3349da38aab : SUCCESS in 9m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e3093a49921a4992abf0e9fdb1259c29 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d380dbb0b5d64605bca248fd65ba8faf : SUCCESS in 1h 01m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5708cd7516a54359bbe3017ab0d19cac : SUCCESS in 1h 23m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f6077f8ccb5f967d78366ccfd2c3fba9c624b3aa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9171e07be1f8431329c7df9583f4846c7ec4a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:13:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/983dcbedd1854aed8f9fae130e547a2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23bed2a62679429c821ddd0c9474c997 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998a8b91562c4ccfb7187b711ee1fa6c : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79b537585c4748f8a9eb3b391c88b53e : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4de25d6fb7645608ea2d265ea8812d5 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/add32fc68fe84cd7b683aa974218b6f7 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/500ad92f232048a98cca36d88d41e98e : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b29f70cfd2ce448fb4ebbdc7c755aed6 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f62ed22c55dc4b2d82dc38b61176ddb0 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b910017721b469d8e5d3127ebdd3c0b : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce51b446c4ec47ac9d798e0c4f07347b : SUCCESS in 13m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a98cbbc3c00e40e4be17458fe228cb76 : SUCCESS in 10m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31541e5637be4e5aa0f157d56d38a75b : SUCCESS in 12m 44s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b3494813d4b24c1796f56ffa7d7ef379 : SUCCESS in 8m 16s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5d526977accf44bf8302d2c9a39f529a : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2dfa5c3a31a4465b8f68f7d702bfa842 : SUCCESS in 5m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/46e1baadf6ce408ab5aa50c6fde3a8f1 : SUCCESS in 12m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/88d152c0b9d542c08496bba979db613c : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/69f9fdfcc2b642308185f17546bd4606 : SUCCESS in 5m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bde6e30e10084c23a1670ac6ff9b244c : SUCCESS in 8m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0a09325c875a4c088b4673a23ad81d58 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/647dae925c0b4389bdebac3e281bfd01 : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5b236578c9824aa59e772592a368353a : SUCCESS in 8m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d3fe7005fb9d41af9c01ecb7a177cbe9 : SUCCESS in 12m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/52c460ad7a244ff58e7c696d5750fe5c : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ec9bd0ba0023468099e6013d0a420b0d : SUCCESS in 10m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7face845f0cd4de1a8af41e80df863e1 : SUCCESS in 4m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/38e23bf891b0403baa6a6e752d45540d : SUCCESS in 11m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ff6d499b8e4b4c679539eb7d567abf87 : SUCCESS in 22m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b9b47ed612b1467e8761ea545a8bc097 : SUCCESS in 11m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ed4074074cd94760ba42abc6073dec69 : SUCCESS in 13m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/24e12409fb4f414baaafaf11bed07df1 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edd5ec27dd7a4c05a1edc81bd85fdc7d : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26d11fe60dac4aaabab50f82ae1b1c63 : SUCCESS in 19m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/46cdf4f641c74602858d6bb37610fd02 : SUCCESS in 6m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b4f6c069a44a4d9ba2847cd711b097cb : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68ac12cc414a4a85bcce25b1bf3fd42e : SUCCESS in 5m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7065e370e0c94f75be1cd411d461c888 : SUCCESS in 15m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e3ede219a40480a8493a35470391cce : SUCCESS in 51m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/300eb3e8f1064966b7c6d4b88ce542b1 : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4a2cf4e5bf818eb19722994ad7d083a33eb85846","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:14:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b1cf5e83f7b6f10b0a535a8a2521a7a74b9f0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 04:30:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1abc6bd29bd64d05aa0b4cb16397b80d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc732c07e2b4a44ae77d21abeaaa0b8 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5209f37f564e40960f46e46082797a : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7927cd8e6e844311b443f50ed7437b4d : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4efb598e0e72491fb511fb0cafd85f56 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/917d5ec0703b4baa88cdae0ea1cec13e : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c952c08e641b47c0a85bcc294b87ae38 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab0b0a28775a457d95dfd71e629e281c : SUCCESS in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9aa11ede34d44bc5af7374573c205648 : SUCCESS in 15m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b3ba63fa73bc496ba92c36ba707ce54d : SUCCESS in 11m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3ed8b212fc1441309ac7999248c9b286 : SUCCESS in 9m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/76b4a54ef1924a52a723d7185b7fcafa : SUCCESS in 12m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e50addcd1e8d405cab09e3401a2d2d28 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b91df15aa0a04098a5e729eb6a6dff8c : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/867c43291ffa44929368aff281e1be68 : SUCCESS in 8m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a34210a5f60b4b1a80a2d12c02088546 : SUCCESS in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ee9b4e9cbb94e819857fef2a8d6f828 : SUCCESS in 8m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d9234c1be7c048d1a932a9a13609948c : SUCCESS in 7m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d665f6dfa37f4e148a055697b85b9167 : SUCCESS in 10m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e38d2600c6e54e6dbfb230201654edab : SUCCESS in 3m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/94cf96cab3bd414a9dbb8b9599858bfd : SUCCESS in 11m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3e1045c3145b4c7d971f93d7f4d263f0 : SUCCESS in 4m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5447f1e4ba5b4d8ca4b91c80ece1aaec : SUCCESS in 22m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5198e014b8314288be3c405f40d5ffae : SUCCESS in 9m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7c8ca68416a9452cb2618454ab524bea : SUCCESS in 14m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d88d85b7e95643e7a5971f909f7fbba8 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d3d14253cd84d659440c96f58a1d23d : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/406970e420e04a33afe21caf3d9eff3e : SUCCESS in 24m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1fd18c19a15845b596be0305098b7f8e : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/da380cd7f46e4bcdb547a869ca1e1636 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c504d77a77d4e6d9ce9a0145952846e : SUCCESS in 5m 55s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5d9a52c145d4483c997e93a1833af6a2 : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0ed0b88619c4680b2250c528e19ad9f : SUCCESS in 44m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/726565d7048447b5ab77a650fbd33a21 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"079758b6b22c9475885a6b841f239c4acbbd8511","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 04:30:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"12a370939f3f260a0e80cb27814cb4d331eae252","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 04:31:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48224baf1fb14c47bb063f172bf46a2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6b7282e1b942efba4473f73d41f8af : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05ca7b5bcae8e18e87b114bdaf8182226bb156b5","revisions":{"05ca7b5bcae8e18e87b114bdaf8182226bb156b5":{"kind":"REWORK","_number":1,"created":"2023-08-29 11:44:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/893036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/893036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/893036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/893036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/893036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/893036/1"}}},"commit":{"parents":[{"commit":"130d068736a6e4ac6039c3d3520d1d0e712ba0ee","subject":"Merge \"update constraint for taskflow to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130d068736a6e4ac6039c3d3520d1d0e712ba0ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 11:44:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 11:44:04.000000000","tz":0},"subject":"update constraint for glance_store to new release 4.6.1","message":"update constraint for glance_store to new release 4.6.1\n\nmeta: version: 4.6.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Pranali Deore \u003cpdeore@redhat.com\u003e\nmeta: release:Change-Id: I47f2869de1bd238bed1deb13f7586b596a7b29cb\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id68179c3d725b7eba7c4e68efdbc2a1406ce6d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ca7b5bcae8e18e87b114bdaf8182226bb156b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ca7b5bcae8e18e87b114bdaf8182226bb156b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
