)]}'
{"id":"openstack%2Frequirements~893040","triplet_id":"openstack%2Frequirements~master~I54a88149f9ce1ff924c1bd19dc82e1d5940fffe8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I54a88149f9ce1ff924c1bd19dc82e1d5940fffe8","subject":"update constraint for monasca-common to new release 3.8.0","status":"MERGED","created":"2023-08-29 12:21:04.000000000","updated":"2023-08-31 23:55:57.000000000","submitted":"2023-08-31 23:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893040-new-release","meta_rev_id":"59282cdce6d6ca0eedeee8fb26f0db052e2eaaa7","_number":893040,"virtual_id_number":893040,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 23:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-29 19:32:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 14:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:32:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:31:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"1146b2ff492210f65b649f3d807a618aed24b6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-29 12:21:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c61896f04d4f24a9badb36d7390551b289736e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:03:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e18f98cbc9a748bbaff71b3bfaa6030f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50831294851d4d34aa62ffd463852e14 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b3b8d630de2434e824cf556350bf742 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7632d7d028343e68a77a149ebb25c7a : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d6431fb912f4dd784a3af6141700993 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29d4410db07f4dec96c1126c44888787 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6f15f2af10749ebba5e640211bf5d7a : RETRY_LIMIT in 2m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/964402cfd0cf4571810df4222318e4b6 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f87416884d84f8f83aba9682eeb4ed1 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/13f8f5efdde94704be8b5c3276ed359c : SUCCESS in 9m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1ebe3e73a50748db8b355cb580c8e51f : SUCCESS in 13m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d461b2962b2548ebaebee71432b82db0 : SUCCESS in 9m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e637864bca104c56973f38c28985e224 : SUCCESS in 11m 01s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/73a5b048fa404f95aa93e3cf23177cdc : SUCCESS in 6m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/044e73bc962e4dc29fb7bc9d30c86feb : SUCCESS in 4m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/314886d0e774481385118494e8ab6af1 : SUCCESS in 5m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/14d60ed3c1f24abca4e0d802dc8b3502 : SUCCESS in 12m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0900b702c5de4f039a64ee11ad21b460 : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4ab201dbcb234ce8adef014e239cece8 : SUCCESS in 6m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f8b46183f2a843eebb923533dd678ebb : SUCCESS in 7m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4bc2686371424c05bc7226061bdf2bec : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04c9064545d64a6c8a1a6850565c121a : SUCCESS in 8m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ceaeadc164514d7d9b7bc3b52f177114 : SUCCESS in 7m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1e49279e7f964f52ab97cc0ad673b5e9 : SUCCESS in 10m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7ff78773e474467eaebc8dfc81d8c2ec : SUCCESS in 4m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8c0f04a7f8c044db84c07c22e9f11b60 : SUCCESS in 10m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/427b40e1741f4f829dcb751c42e50ca8 : SUCCESS in 4m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/de92593472454fa6bddf88370a947b9d : SUCCESS in 10m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5a4ce53f872c41899650e9610e34eeea : SUCCESS in 27m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8322c289caa1405ca7c9c7daf981cb3e : SUCCESS in 9m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7d7bfcd506794d9dbbfe80f76a3ab9e8 : SUCCESS in 15m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b9ad6f90d9984854867696f8e8a96eb0 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c8eb7f5eea5d46b7a02407320b1ec7ec : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78da7dc6d2514497a7109c833075bf3c : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/74efe9aea82a4120b5b5406fd17ddc8a : SUCCESS in 5m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/88b8fdfbac584f7eba920a7f46d5c035 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85e4c1c2fdc2418c9440905e1263b331 : SUCCESS in 5m 54s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7ddf338106ed4d50a0d3dd46c967991a : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe4544f7fbea475a81350cc929be15e9 : SUCCESS in 1h 01m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d368e08aa5824b77bbc613f058045c67 : SUCCESS in 1h 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f21026b0c629f47e8b83fd063cfc8902902fd8cd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:32:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6af9d1d9bec722d83d2720528167cade308331e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:32:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07a07c0d6b74c4657cc2e548ebc45bc9b8e259f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:15:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ecf415ba2ae46c4936056a213ce232c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9d347baeb0453398e28883b97252f0 : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4032faa7598b436b96c3602ba3b85006 : FAILURE in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd55d7fa1da74a0da949b13115e2a33f : FAILURE in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/250899786e384b9cb27bda5204495646 : FAILURE in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/858ece67fb474ce0a3b15bc48af9e99b : FAILURE in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffde7548bf114be6976f05353230d4e0 : FAILURE in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/628f7f8661044eb5bc7325c089a1d674 : FAILURE in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5ece46c880774a1881d5b35b26163d85 : FAILURE in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1a1f7324c5864b8bbe335bd28e58bbfb : FAILURE in 13m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dacca98d3d0c476e991acd179faa9fb8 : FAILURE in 9m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f317114727004c07b521072e6b2b016c : SUCCESS in 11m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/72811eb265094236aef2c81f4036e9cf : SUCCESS in 5m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/13f2e3229d484471a883d8b58cc93b87 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/976630e1e6e74014ad9dd60ba611dc66 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/79852570a3084989b15d2989bc1916f4 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f0a6cc3c23a49deb8c3c0418a606d4f : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/51c6fbb46b2b42769f73f06baa0b52f7 : SUCCESS in 8m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a0b00d9a33904a7caba33980c838cbd3 : SUCCESS in 12m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f4cd110843f54b9eadaa2f8dcc33a493 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/303dd883e0a7476d8c6dd55a1b13fc49 : SUCCESS in 11m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cf244c2f856945ffb511c8366334b447 : SUCCESS in 4m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e6a840492d294eed9d911c9477c644e0 : SUCCESS in 24m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c911be348f5047f3bd6c2c780db662d7 : SUCCESS in 11m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/debdc81a7ae84aef94a56046e2dba786 : SUCCESS in 16m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/53340835af1749378b93b930f0128490 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49bdaf134ef54e4f855b86943aa54e6d : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d985d3149dcd49c1ac973d22d98b0f84 : SUCCESS in 24m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aebbb5811e554ff2ba9efcb8d2acb9ae : SUCCESS in 5m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3dddbc09ffcc45e6855b59e5589c11b6 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab53ce1e159e4a44a170757ca6462339 : SUCCESS in 7m 18s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/77c689cfe1d644769ef47c15083f2aba : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/895750768fe741d89805e497106d21d3 : SUCCESS in 1h 06m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b72341552574031b1113cfb161eca0e : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bcfa29d437abb48a9dfc4bc6d1f5b81f10e2784f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:31:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9f30fd4d2140425a24aa426f2bbe2fd9f717751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:03:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd9c30195b4a432c81843ef285fea3f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47607015ec5e446da33d4b2fbfc5ce6a : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd7ba82d3d84f569e7f0a0b9e0047ac : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ec589c3c9404c5d804465f40cd63e6b : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd506c52753842508e31946f5da5e182 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eec4eb3c33f641c5af744e52fabecbe3 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/329c869153134ea08841ebce14ce7613 : RETRY_LIMIT in 1m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aeea9ccdcffb4a34900ca17fb58686fe : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cb7ea04995b46a69c0a1db06f362ef8 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/12ec0bcb10714bf8bc255a0d9af5c62a : SUCCESS in 9m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ec007556ad14406f8f87469f936d3975 : SUCCESS in 12m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4fa575de64bf4047896ef4bcbe3f0fc5 : SUCCESS in 9m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/364954ba83a4413ba189f8d02def6c38 : SUCCESS in 12m 03s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/74322c3909f84b4fb5e6fa0418225ad4 : SUCCESS in 7m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6503ea22469342df81281ab46f84b9fc : SUCCESS in 7m 02s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6439d64a9db3450998c236a3d845a1d7 : SUCCESS in 6m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bab1fd8aae134366935f5939ccd667f2 : SUCCESS in 14m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f6339e50d3624fd8a643866f3a612261 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e64cef76d4be47e1b977ef22c37c30cb : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/906f63a0849a4752b159cba4a1a184ae : SUCCESS in 8m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/24507fa8f5d14e2c82b4e7f52d959001 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94cc85433a1141aab24f21f0174091c9 : SUCCESS in 7m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e162e17fd39b4d4da685df2b8a46823b : SUCCESS in 10m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/15dbc94c91684e1e805e3fb5180078fd : SUCCESS in 11m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e2f1c96f452c4b1385dfc9b75338c5dc : SUCCESS in 4m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/20f778bdad8442c3bac4f5a99bceb79a : SUCCESS in 11m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7a1e5df28b5a412cb625fe491972a14b : SUCCESS in 6m 02s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7f72a8c34b9b470a85b67a81c8cf3485 : SUCCESS in 11m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3e74fa5c63154f08b60393f3b7da3148 : SUCCESS in 31m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5c09e10bdc354cf2b35ed2e3eb3c67a0 : SUCCESS in 14m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c94d17c328c247528cd9793a7595460b : SUCCESS in 15m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/faece8d62323456d8d78c2f6c9ee63c0 : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/76ad2e7005234e30a6325b485efed42c : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34c8854e2feb44b3a03e22b6a53d38f2 : SUCCESS in 31m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9495fb3fd1eb4995a2ea5e2cd714634b : SUCCESS in 5m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5cc629b18c074815bbcbe54c854f6217 : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1f73fcec0d64df599a848e72b8cbe80 : SUCCESS in 7m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/68e0f6cb61f8401b931c12d542e9a469 : SUCCESS in 13m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac97b62f1e2847f8a19718945240a41a : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2c01e3bdedf4569ac1d90d12bb9e84f : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2c3f5396f6d433e737550677a2fe9c697bccd0ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:04:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3dcccc58641cda64d8af64f9f20048f54672d4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:54:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88157b650aad40efa9e4dbe726a52849\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dabe247a3f26461ea189b8b5109ff374 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be60b0b18f44fbbae2476abb1e14013 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/349dc63d439645399ee33f7e0d82875b : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05b1613e62694eb78560a8155332124c : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5042668d1234189a1414c2bed1fca9d : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/426a5291fb8a487e9230867515718d6a : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/58dc2b6a4b4340c0a4de585ef67a4365 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2036766526a1471f929e556a7eba376f : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2e1c7a118c4d4cfeb953e0d4e234eef6 : SUCCESS in 11m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/32b4b45313c741919494ed50d9d4cd71 : SUCCESS in 9m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0a62b950396147a6b52ef4b99c21b98f : SUCCESS in 11m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fa70d9f3b221400f86f59dbd3f72292f : SUCCESS in 5m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a0b3e44a8611417f9abf1c862f86d5e2 : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/353c1aa8e5914875aaa3cf686044fd2c : SUCCESS in 7m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a621bd15db5b4b3caee0c1fc073ad4ca : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/308e2c3c5a3e478197ec24a589c5cd09 : SUCCESS in 7m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/08b89ebc36d346869b2fd2a0adccb4c8 : SUCCESS in 8m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1f5a451ab68b4bd0b398ab7bf633b62e : SUCCESS in 11m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bc86660d048d4f30b13b3c65d4bdfa22 : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6cb1320203eb46399520dc2db16cd37c : SUCCESS in 10m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3ce2a7333eaa421e8f82d3a51756e2c7 : SUCCESS in 6m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/016bc67a5bcd4f51bb83b6d52d522894 : SUCCESS in 24m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/40e850af7e5742798f81c48ed9faec75 : SUCCESS in 11m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ff61c015ec694c1a850a32747a96382d : SUCCESS in 13m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5aad8c93b8924f89ad7decd650fccf8a : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b783fd1551e47c4af8cdb0506bd6fc2 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9cc7191b023456bb224bd03688691d2 : SUCCESS in 24m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/80871a8cf61745d989c94a1010135e7e : SUCCESS in 6m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/68850b33389c4699868048a3659b1c1d : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/235e32a354b84e51935ced06f5b0f0e6 : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/16a4e42ec93548a3b26b32e6a20fd9e8 : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4863411dd7946c0a7de7edc8f644a6f : SUCCESS in 48m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35d260b06bb64296814bd933ecac346d : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8c769f0a103ce889b945cfdb8a1def929807b481","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:54:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59282cdce6d6ca0eedeee8fb26f0db052e2eaaa7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 23:55:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19087c531e504d05bd823a811d65de17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da24b37118542cc831b5c6c1573c0e8 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0b3f70bae9e0e79919492f18f2a090f24079622","revisions":{"f0b3f70bae9e0e79919492f18f2a090f24079622":{"kind":"REWORK","_number":1,"created":"2023-08-29 12:21:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/893040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/893040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/893040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/893040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/893040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/893040/1"}}},"commit":{"parents":[{"commit":"130d068736a6e4ac6039c3d3520d1d0e712ba0ee","subject":"Merge \"update constraint for taskflow to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130d068736a6e4ac6039c3d3520d1d0e712ba0ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 12:21:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 12:21:02.000000000","tz":0},"subject":"update constraint for monasca-common to new release 3.8.0","message":"update constraint for monasca-common to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ifa678c016fb37c3bd5059af87fea4c20ae4ab1cf\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Hasan Acar \u003chasan.acar@tubitak.gov.tr\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I54a88149f9ce1ff924c1bd19dc82e1d5940fffe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0b3f70bae9e0e79919492f18f2a090f24079622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0b3f70bae9e0e79919492f18f2a090f24079622"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
