)]}'
{"id":"openstack%2Frequirements~893043","triplet_id":"openstack%2Frequirements~master~I3ad2c40f2e53ac33afb4400312cc606cf58d24d5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3ad2c40f2e53ac33afb4400312cc606cf58d24d5","subject":"update constraint for oslo.db to new release 14.0.0","status":"ABANDONED","created":"2023-08-29 12:36:56.000000000","updated":"2023-09-22 22:39:45.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b3098047f5eeafd7785865855e8c177520c93d38","_number":893043,"virtual_id_number":893043,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-29 14:17:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2023-09-22 22:35:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2023-09-22 22:35:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 14:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:32:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-08-31 20:40:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-02 01:55:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7bd56acac476a0fae091775e89dd4fab645d82e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-29 12:36:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab908f1ae2e409a213cb34e2ae1a6cbdc4ec62ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:17:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22146aee850f41248e3f897c2a11115a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/041fd64e89d4419da2e7c273c3befc26 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e777cb57ff7a49e7b959fde5a074cf24 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c99bdfd1a68b4575885f499d81a100e9 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6731f4824e8b4dc985c7e5d836f5ed55 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94ed41f3d57e4a32abc9cf94c7deffe4 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ca0dd5952314c7c9c8d2b9b86540f8c : RETRY_LIMIT in 1m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab9c64a529144d0296e2d73c5edab152 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/927f546857e44679810663948b00dc7c : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/36bf56d3cf8d45dba822ad4e332fbbae : SUCCESS in 7m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a3c7fe69586245979430e43dd9ee4ac6 : SUCCESS in 13m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/de7e17c9017c448795fe41d7a24af0c1 : SUCCESS in 8m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/00be450b1dbd41bbaf5055db87b28b18 : FAILURE in 6m 37s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ec6d89b8c36442f5944c1bb392618d05 : FAILURE in 5m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3bcfde01676d414ea25b6df37a7a6634 : SUCCESS in 4m 17s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b3d97511731a46109af6fd6bc2963892 : SUCCESS in 6m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/07cfcbbe7f6d42f8999b42201464184c : SUCCESS in 12m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0b61e42841b14a689b7cbb4cdf0ffaa7 : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f6eba59a9c704601aaaf7a38b63cd34d : FAILURE in 4m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/acb0a71b4dc84146b28a97b7557192b9 : FAILURE in 4m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/91327fb3826d493eb88d3200f63515ed : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aeaae5a30bba4add88041675943af53f : SUCCESS in 8m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a9ce5a21a63944cba7e5361640e701bf : SUCCESS in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/67c8d39ae2c846ecafa5a4f06fa4e88e : SUCCESS in 11m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/60b47db64a9541d3b4bbf3edf8a29f11 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/74d5b10456cc433091ee683bcfd9f57b : FAILURE in 4m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e252f7d943df4546a746f4894eb237a9 : FAILURE in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0c684550c66146b6ae66de078a4cbb0e : SUCCESS in 10m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8caf1db936d64f399dbcdeaff0a8c659 : SUCCESS in 29m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/20c06a0ee72e4df7a1cfc4ae95fb0618 : SUCCESS in 12m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3d65f62cd1eb44ae9a3eb0dd31feacbc : SUCCESS in 13m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b95c831512894f66aacf6b92cc73b1cd : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/752968702f3d40199289f12e4ba7bcfd : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/110413236c7545e4b46d53e5ebefd3fc : SUCCESS in 20m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/edccd29bf39b4830b52f7acdccaf2e36 : SUCCESS in 4m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/85166bc43bc9406994bc62441cef7343 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f2e8acbdcb34ac3a7fdbcde10f51c83 : SUCCESS in 5m 32s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ba973ea7e9ca4dc4ac944dec27e5b216 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2ecebff805c461c9dd270920ce24615 : FAILURE in 1h 06m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c1a6c9cb87749c2b64329e239ae02b8 : SUCCESS in 1h 35m 12s","accounts_in_message":[],"_revision_number":1},{"id":"73b3fb6fc743a1e4ec154343590a8658d84ad05a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:32:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4bd25df261cf78acc5033e944260a4f6f0e20e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:02:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49cb8b1f376c4594a8edc87e7b6fe817\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebfdfc230381474e95160e65751c8ce1 : FAILURE in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98789f2d8c77456999ac20f0c8598ab2 : FAILURE in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79a2eb5209c841ea8723ebfec2c1320c : FAILURE in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84190c09fb6b424cb116c97305a37252 : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/288e0f757058413180c6a290184abf8b : FAILURE in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36e5c2d8b4b8451eb4ac8012b783f835 : RETRY_LIMIT in 1m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/091beed77c4240fcb4bb36e32da89c36 : FAILURE in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5232b1e5eb1a41cfa196757ab6e0b789 : FAILURE in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1261f56de9054d209fba5b8a8f30e1af : FAILURE in 7m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1412d86e5ccd41158ddc24c19d059c9a : FAILURE in 11m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/04f2f5bc9f0d4ddca22a835fdf99a287 : FAILURE in 8m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/06dda3636bd847599ca545bbf7a0d5a2 : FAILURE in 11m 17s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/93cea9ec7f1e469b84bea135c0c68488 : FAILURE in 5m 28s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ba207148ea80462eb5f09f286fdc681e : SUCCESS in 5m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6fc85acc60154af29fae40888db58f5c : SUCCESS in 5m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6b5e60eb1e564db393b3caba8a211aef : SUCCESS in 11m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c237f5bab90244919f4fcde7c303b8a4 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d3900243ffb44c9399659f1d03b0268b : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6d7b30b377534ca1a6a4ef3e1d9df144 : FAILURE in 8m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1da5cedf712a461fab1d1b56065b2f47 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a98d2e6daba241c1a6782379b51e0946 : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ba11df4824324b208471cdd29854ad5b : SUCCESS in 8m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b2221f5cb2d54774973505e011d376b4 : SUCCESS in 11m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3682724c13554e14a8ce446826c507dc : SUCCESS in 4m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/40bf5e002604457297510b53cb2e2bd6 : FAILURE in 4m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b046ac4e891d4ae9996fe6bf6a59bf31 : FAILURE in 4m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f3f381cdfd214c28a720afebf4b3b9f4 : SUCCESS in 9m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3d271646eab7445e867524177350d69a : SUCCESS in 29m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c2d703af4ad44c09bfddb0111a20a038 : SUCCESS in 14m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cc57a5683bb24b0ebf6c472574c8f1f2 : SUCCESS in 15m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8c7a8f8ca524433785c51c7cac99e27d : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f65ab6196a6d4c14b2886b7dfb6c34e2 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc1520999b2a4a618d6a53d2dde77c91 : SUCCESS in 28m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/39113c4c8851490fa507c9c2b5beb574 : SUCCESS in 4m 51s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5c66cc41139e42b3be02dc2a2814b09e : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65a31fa0a2104684af3eb7e8c2fed575 : SUCCESS in 6m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3cbc2befc6124a39a123c983483dab92 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04efa2bf665e4a1da00bdc420ed6b053 : FAILURE in 1h 18m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd2e7fab32b343b59d17e38a43507ea3 : SUCCESS in 1h 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7debd5c124fd1f9dca36492efa4d1825b303a4bc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:40:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f20a16f64e4ad85feb391c1de830ac34d7f2ab77","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d8966329d12aa15bacfd232fef18df0bfea5a15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:26:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0f74dfec5ff461a92e060880355db6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defc7570d77b4a65bacda1500b4a5df1 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2565fe258b8b470ea952345587959ff0 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b881f62e41054f66b9ea0c64e6411b38 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54473f98ea374627bb77f3c3a309b60f : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb149b326c1246c987d37029cc2e5263 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4387ef5603c450296a5e20f713cef52 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f4e269be59142fa83794a448100c2f0 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/334ae30819914eb9bbcb227586270b61 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14959bb8fe554ae39206b3fc61a19fdb : SUCCESS in 9m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6424ba2fa64741388b811eeb908db062 : SUCCESS in 13m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/57a1f1693331453fbbcf516ad6d2e8a1 : SUCCESS in 9m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/222ddb7977dc473d83e8f56fef480902 : SUCCESS in 12m 12s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/334e4c72e1444ef49fb8286423a62341 : FAILURE in 5m 39s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/04454b262d5f4a578e002554d1660617 : SUCCESS in 5m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/42e05fb89afd42b9a76c3f484f9ec45a : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/837ee9c388fb444fb601fbf7ebceadfc : SUCCESS in 11m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6ade05a0111a4cd99ee446484a293bf6 : SUCCESS in 5m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a93b1d504728468f9973e52d60d93967 : SUCCESS in 4m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6b6d9ddf25814ebfad036fc13b227f60 : FAILURE in 8m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e63c08b2c24e4c6ead231fe3eb551e1c : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c1f4a3fbdd344b0b1faaab4116e4294 : SUCCESS in 8m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/38b03f34bda5492ba016f02ba1cae430 : SUCCESS in 8m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2f530db0cd374cf188964230e838c5cb : SUCCESS in 14m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3e8c64ab878d48ab8168bd88384ff959 : SUCCESS in 4m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a55c376992e5477f97bc3c6b7dea42f3 : FAILURE in 4m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/09a8ea661b07431fb8f41f9bbe63b23a : FAILURE in 3m 50s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/59ced47aa4ad403481fc37e51bb90795 : SUCCESS in 11m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/768068b856bb4e7dbb09f8b249d630dc : SUCCESS in 20m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3cfa3b5807d846eebae7fedec74b3436 : SUCCESS in 12m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/018df3e71aed417ca546cdf96884a58b : SUCCESS in 13m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/77f0c5745af144e6aaeca6adf9dd0c66 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c4b0725f184a4d97a447348f751ac723 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/648ba24d51064336b19cfd9f241958c1 : SUCCESS in 28m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c0d186d367de4cf9b4a42447c61c399e : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/188a5d3136044a6e90e0926d57bbf14d : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4914c48cbc9b4a5e860f7fbb93262a19 : SUCCESS in 7m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c78897b009c44455bbec6e2c4f89cff4 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/292eb74a5e774f5d9de886d7df68cd23 : FAILURE in 59m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a1db7c286394d26a29950dca534178a : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8227decae6c66c38a410a5cdd862741e0caad264","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:12.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0140a0bf7c5e5b3306325ce534d6993fe08f52ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 19:32:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd6e4f33d9cc4a1aa0f9e51db72e2c21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d515b45186a4b64a439cd40b79c8627 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70cec9f2b0c0463592072bd0b7036741 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a5fbb62b9844c85932225fd4baf536f : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f013ec031c984d7ea31fa4098f1c8b34 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3c7ed2d7ea34ec48cdc8aa8a45b765b : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c89722a56b00432585f27f3b8114f5c3 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f1c3766271145539d33b42ac9264cc0 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbbb14b9f5634d8c8c2b4150628e3938 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28b81d546288486c8b227e6c75fb7000 : SUCCESS in 8m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f52281c343144500bfc091049c9d4e0a : SUCCESS in 12m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/26136a81a6d04de79b31d6d41beaf258 : SUCCESS in 9m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/333164b995d54202a453929702be1fd5 : SUCCESS in 11m 57s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2318c738588947b1b9e8c3ba059895fe : FAILURE in 5m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/386a6eb097684b9d8d7afa47a6ec10c0 : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/71487813af73444baedb1b3c2ae1d910 : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c37ff3580c144e68b63a6647c56fd31e : SUCCESS in 11m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/54b9da0d152749e281cdee16e62d8ea5 : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/45f9d59e5dba4143b3314157e877fef2 : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b07dead61b974e69a32c3337ac0c0472 : FAILURE in 9m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f691609f5f074f0e80813707db411aea : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e034237d9e54e9da72478023bf0d5c3 : SUCCESS in 9m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0becd10c65334b16adc9b6e6f8832641 : SUCCESS in 7m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ce286dfd5df94e49a57a206c873b65c0 : SUCCESS in 11m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/10f7272283ee4944b5ca1570dd7c1d38 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4530cef2673f43638cac37133e4c91d9 : FAILURE in 5m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bbf33a770abd4ffe912197759c04c318 : FAILURE in 4m 21s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/21344e3d8b3f48419a97c8803adb3dd2 : SUCCESS in 9m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a81ace22ab0a47b1a0d8acf5e30effcf : SUCCESS in 24m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c6531693a84341cbb38b897f72f45e1c : SUCCESS in 11m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ea0bdfb411ea4b29a72dab2b007b6b96 : SUCCESS in 12m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/257eba6b12ef4469b511cb5b5051d99e : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0e2346d556c4d79bff03fc1efb79aed : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0473ceed57b41bc9e22e6433a8baf8d : SUCCESS in 23m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/38d69a9cfc4643ab8a612ac78e703979 : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/140fa752bc314a91912e036c510d5e5d : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccf57a846dcc4482ba9dadaf58cca630 : SUCCESS in 6m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1cc4f3630e09451c9d8f91602375bdbb : SUCCESS in 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d006405e64a54f1eaca6a8edf50ec1d7 : FAILURE in 57m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a9ed5241b1e4d06a3523262149bbf12 : SUCCESS in 1h 29m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1b6caa6ffbf3efc9e43e0670a90bcf2a0ef4361c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-13 07:38:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbd3ca001ba1393f074619f44df2b5e51990c78c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 09:15:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91c6a76b09354259ac821ab5c95ea6e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d13ae3ccccf248e6a0cd0ab4001b9217 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ede56f53034607a7dee17d5d5b0f28 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff8fbeae22ac485ea3451c4624e02b8e : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b734b8beeba4fcfaaa79c1211ab4907 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/193ef5207b51461e9a84909a58d072ad : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60fd3c766a7141108363324bfbfb6224 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e14b20250adb44dcb03910ee88072582 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53361a60644849f2b3324cd12dbfc2ea : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70ed71fd8a904e2aa593da5490d9473c : SUCCESS in 7m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/40f7e4d7cd42480dae94d6a13597ba61 : SUCCESS in 12m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/34eb056becf54667bb1bd3954a85ba7f : SUCCESS in 6m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dc54c5358c5f4cabb6c31fb392d38f13 : SUCCESS in 12m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2c6ea59188c74f2180707841a3602840 : FAILURE in 5m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7f0fbbf9c2cd4b2e9b3dcb52d0905e65 : SUCCESS in 4m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1c4ee3736f7844139e54aae83ed36d48 : FAILURE in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/27e47f7d24d34534bf6923e3068d889c : SUCCESS in 11m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6cf7373205d3464b927bd09b56e6dedf : SUCCESS in 6m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/86afd991e8104bd094fae886a8b1f712 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/edfdeb9c8181451eabc7a6c78df2bccb : FAILURE in 7m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/97c67e903338478895a83f5c7470c1d1 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f292257e60444ff7b5687d6b711a4f29 : SUCCESS in 9m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/308e75577f4346e89ed2d2e8a68a2964 : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fc9c45c004bc4ebc837ce4782766faed : SUCCESS in 11m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/17fe7e82c577452cb0555e6e40a75dc9 : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cf97e9e6b47d43df84f5e19f64a7e21e : FAILURE in 5m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9808eeb771cb4fb49d79b85683b3eb44 : FAILURE in 3m 30s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/70dc033351904c1794c3e9fdceec4ae3 : SUCCESS in 9m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2fb1bedc154d44b988656623eddb90a8 : SUCCESS in 24m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4d9c0890f7a14dfd923fb94036e35c59 : SUCCESS in 10m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/838bb231c9b54bbabbbbfc19c2ddbbc9 : SUCCESS in 14m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e8e9de3ffaca492aa5e5a6c05d9d582c : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/772bd7b6797e48f2af455558a93253d1 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/922bdbf88cab42aa9a29b09bf795b2fd : SUCCESS in 24m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/77f06bd216064b86aad709fb2c48758c : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/93ebc8b528fb4a7480bb514213022ccc : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/773a3524d8b8445abbab1ea7e4026204 : SUCCESS in 7m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ba169c10b94b44fd8896ecf5f61ebf35 : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4518fc80667b4fa1a91f88b2ed1fa2e0 : FAILURE in 1h 02m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb95921e96cd420ab75b3e104b494d04 : SUCCESS in 1h 31m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7ab7c21ec1f789cd83383ce529c9498ea0c82ed9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 22:35:47.000000000","message":"Patch Set 1:\n\nrecheck\n\njust in case...","accounts_in_message":[],"_revision_number":1},{"id":"b3098047f5eeafd7785865855e8c177520c93d38","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 22:39:45.000000000","message":"Abandoned\n\nclosing in favor of 14.1.0","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0390f772adfe82bb6e117f12be79a01392d91f5","revisions":{"f0390f772adfe82bb6e117f12be79a01392d91f5":{"kind":"REWORK","_number":1,"created":"2023-08-29 12:36:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/893043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/893043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/893043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/893043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/893043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/893043/1"}}},"commit":{"parents":[{"commit":"130d068736a6e4ac6039c3d3520d1d0e712ba0ee","subject":"Merge \"update constraint for taskflow to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130d068736a6e4ac6039c3d3520d1d0e712ba0ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 12:36:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 12:36:55.000000000","tz":0},"subject":"update constraint for oslo.db to new release 14.0.0","message":"update constraint for oslo.db to new release 14.0.0\n\nmeta: version: 14.0.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Id4f39b7e118ca66f464b5ac8b991245cc50b1cff\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3ad2c40f2e53ac33afb4400312cc606cf58d24d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0390f772adfe82bb6e117f12be79a01392d91f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0390f772adfe82bb6e117f12be79a01392d91f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
