)]}'
{"id":"openstack%2Frequirements~893060","triplet_id":"openstack%2Frequirements~master~I7080abace3fe17850c62e0baa961be8f7092a60b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7080abace3fe17850c62e0baa961be8f7092a60b","subject":"update constraint for keystoneauth1 to new release 5.3.0","status":"MERGED","created":"2023-08-29 13:21:27.000000000","updated":"2023-09-01 00:06:08.000000000","submitted":"2023-09-01 00:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893060-new-release","meta_rev_id":"d2405f1ed66dae99003cf903e8dce98121111e54","_number":893060,"virtual_id_number":893060,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 00:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 13:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-29 19:32:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 13:23:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-29 15:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:32:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:30:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"1dc1c092065e351346971c56ff69aad2d17f35d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-29 13:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8431bb4f364afbb822039628cd6524f58ffb41ca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 13:23:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33960fd073317e9c6cf10456494c01bfd9c374bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 15:12:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4625085581094daf9d8cf2c2826767ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40bca4f7f53d487587351a74c4f84dfb : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6f824cbcd44f3d999a4223f8c78e59 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e30424d9427f471192dcea0db740a8d4 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb217dfb036042a398e5d62a412c055d : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1ea92eea99a495c90fa8619384831b4 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78cc8bb645034ae18a0eab73b14780ac : RETRY_LIMIT in 2m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/301a298f6c8e4663995b6a0561959e5c : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6455ed7c61d465e999744dcffe4703c : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da5aabfef6c645d9aca6d23518f8cf12 : SUCCESS in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a40787c24dff49609cb33e4916721d91 : SUCCESS in 13m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54d999d125384f12af06991fcf857b08 : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a308e89a23be432697c04fd7ac701bad : SUCCESS in 12m 20s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/91bae976cd1042b9991ffd6946e6de83 : SUCCESS in 9m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/27055ad573c74b5ba6f91d8d7e1e0690 : SUCCESS in 6m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/03e309d88e94453ab0c63702993da62c : SUCCESS in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7fb32a81dbe844429d5e23a3f6744e91 : SUCCESS in 12m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/589c54c4a49e4c51895c02b9fb844a8d : SUCCESS in 4m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2f45880e64ae4f95b5bff27b6b8fd41e : SUCCESS in 6m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d49152d85dd548389e5df0e6592c1e56 : SUCCESS in 7m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f31bce64a6b34d83bd4dcb3c118e7b3c : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb81f836882c45ed9f7d61fc832c2634 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d1c45762aa884a5c981fbf3782afa3fe : SUCCESS in 9m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/996b60498b6a421a87cc217c4d534195 : SUCCESS in 12m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4b25c354f65e471080c7e64856200b2b : SUCCESS in 3m 55s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/91a7a0ecd89745b195ddd93385b06c52 : SUCCESS in 10m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ec633050fd0e42bfb9e4231c4e0ea283 : SUCCESS in 4m 36s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/323eb4f486e8411db4305a2992feaa46 : SUCCESS in 11m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8ce5d398e6b44973b87ba2b1f594fe4d : SUCCESS in 27m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/13c59352f5bf4c918e234784fd46a458 : SUCCESS in 11m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/53b4a242959741c0b4151d7877eb0e00 : SUCCESS in 14m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/62383a219b584b91b709719e73f33678 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c54621570da9403bacf01580511eebca : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87c9feff2a514f1eab687ae96ecee0ee : SUCCESS in 27m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/755bc801dece4dcfa058523b8aa25141 : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e315b69d143b406f9f004676c387d89c : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b59c53aa3e2a4cd9a16dcc517e90c004 : SUCCESS in 7m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/917430affa824e45861d3641662e6794 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8769da5288c1411d94073cf7d8bd8bcd : SUCCESS in 1h 06m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d00a391887b9460a98f7ac4d12e4e177 : SUCCESS in 1h 43m 39s","accounts_in_message":[],"_revision_number":1},{"id":"78bc5f28574c9285e6a035cfd379b9460fa27ad4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:32:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bcc347e381db91e37de2de12cb4b89679397f4d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:33:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce074ab18e869c852c123b5736f76211d2bb1dff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:08:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d247bc98171407cbc41df71af2cf9b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3e50603af34ce9bb63f056984404f4 : FAILURE in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6863d32783143df97a3001b1c4bc7a0 : FAILURE in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a421747878054f3b9feaf31c3bdd7a5e : FAILURE in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6137b10b35c4e93a6f3960b3b1f09fd : FAILURE in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e48a4a48d524a17bb9eb36c7b393eb3 : FAILURE in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70ae56f83e474c679aa7bc412bcee02d : FAILURE in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/344b4be8bab8477d8bb024beb15fa8c4 : FAILURE in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/000aadd8a0af4ca5b9e6aaee10081901 : FAILURE in 7m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4d9247b90131461199fbf885f4cc6f7b : FAILURE in 11m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e8bc977663454bb79f994bd33020e1e7 : FAILURE in 7m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2648ba8441924a1ab323d4137018e7ba : SUCCESS in 14m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d92c30aef78847508735549bcb50886f : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/427d9cdfe2eb4b1494717340eaea159e : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b08c9b88732d44eb9e73bdf0b80a431d : SUCCESS in 9m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d219907104244514adb805e9a9a51374 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6624bcd2b20b4f07b5c47e45e0bd20a1 : SUCCESS in 8m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b92ed2ba15db478798d79f9c1d2b08ca : SUCCESS in 6m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5d7bd00bb9844e3f90802eeb79f067e6 : SUCCESS in 13m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d7494f97b770481685ddc5674bfcff13 : SUCCESS in 4m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/557eee37390d4498bec4713c6c89d57c : SUCCESS in 8m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7331210069284b6cb7e1af0559fc88d6 : SUCCESS in 6m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4559cbadbc334adcb40326fd8e8a3572 : SUCCESS in 27m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7b11fd71ae3d4a9190b7884a3425cc51 : SUCCESS in 12m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/293103a869804aaaacb8b7ca1f3fd19a : SUCCESS in 15m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2b394d3f120940b2a2972759ec2d5f75 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc074b6c413a4ee48e5697bfa4e0014d : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/247704ef6b4c4feaad4331bc63a50861 : SUCCESS in 24m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/95edc86dcb8f41c3b6d45b6321842331 : SUCCESS in 5m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2a25523d9e5b41c19b504b60235030d7 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a2c0ce6b2bd41389d331099b515aebf : SUCCESS in 6m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/288268d062b6442196d22da99c9ecdd2 : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f94fe5bad78d40e2b3b6d30bae41d766 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af72bb25f2604caf8d7a480afcada867 : SUCCESS in 1h 18m 40s","accounts_in_message":[],"_revision_number":1},{"id":"54700b439fe2a26e19bedfadf680f80dce9b579e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:30:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29f47a146b60f82f0a825e22372cbf52473c771f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:21:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20b7d6f06594450b4628816a6b6ef29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791cef5268d94f748db9ec4b58bf6c3e : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96930fe3dd834764b84569f39b0876da : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8874aaf502fe42678aa3f09a3a376faf : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ba6881f43d94d2e802b5dabde01a624 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd68fff801934fcb9681e8ade9a59bcd : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca4f2de09efa45a8b29abbe17eb904ba : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d15babfa42494c5aacbd5b02c5153caf : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07d5ff7ffc3c49d1ac2c0b7ae53750da : SUCCESS in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a8fab7985d9a469087b34b60f25edde4 : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/245df1711772484a851e6775a49772f5 : SUCCESS in 12m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3babd48d4e7e4cce8cd1b409c6ab1f29 : SUCCESS in 8m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ef4f148592424de18ad035e404ee6417 : SUCCESS in 13m 11s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/769d0e7e61134163881f74d8910a3c4e : SUCCESS in 6m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c9fbfa2f782e4164b79c75cb49a0e1ef : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/40c895df9a1c43b1b428a3c328ddd999 : SUCCESS in 5m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a9a88c5301474f5ea8d5ef6c7eae5678 : SUCCESS in 11m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c049fcbb8a03404db66200ff2f78a55a : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/498e00fbd8bb41f3b30dfa7be547870e : SUCCESS in 5m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c3d0f4460ec44c5da17be99faf759db7 : SUCCESS in 7m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d0742891bf7843dfb2b6969f938558af : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ba1cecb38174c5b97aa7870559a5d59 : SUCCESS in 8m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/81176b9c4f834a75aaefaf7447d473f6 : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/961d3451f13c4181ae88c4aa54dc81d4 : SUCCESS in 12m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bcf5626803c342e0aee1c5429ab42429 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3c738697f80d4b9c8b06d4f546492f11 : SUCCESS in 10m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6d72337d39a6437b9781dccf93f5a3d0 : SUCCESS in 5m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1709fc571e61442fb71760dffde925c3 : SUCCESS in 9m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8bd061a4cf3d48cc998eab10ee120fb0 : SUCCESS in 26m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d7ea18ab586545c6bb41674850293d04 : SUCCESS in 11m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9cf097df15b44edeaa3e35be8e037b29 : SUCCESS in 12m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/263eaa7e55fe4d1f8ba5222a1adc6a25 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/309548289a1948919a757c3704ec0329 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be20a8a7c4e949a48a30a4edd907492a : SUCCESS in 21m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/616dfaba1bc541b1a20960837cf3b41f : SUCCESS in 6m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/df1a43b724054086b232019e72711ea2 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c38fb9626f124527a1fdd7807edf2084 : SUCCESS in 5m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0b559eb5ea6d468daaf98c6a8fd8b1f9 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cbdd337577a4403a58842bf9ef0a50b : SUCCESS in 1h 01m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82f8f2e409f3498da3f86b11043bfda9 : SUCCESS in 1h 40m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3c4a55269a91254ee59cd06404c39cd775f9b386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:22:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52ae5565cb73531cad5ec0f81d556bace866c168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 00:06:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae0471eca9c9414d84f7b6209037985a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f3706b464348908970a63961ced470 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c139812b624c45995757fa6e97b655 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/254e05ae8fa544c0ac91b05476740335 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95bfa2b5a6dd456fb8e2078e02f04429 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f95e166c9b54e68967c8ff464e076a2 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0a841ce24fe445d9a3b43473affa032 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f17302c9b4c344caa7ca2346fef886bf : SUCCESS in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3a247f0412014f1c945a9026a29f5172 : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/30cf927cbb974871af1f5a5b23d482fd : SUCCESS in 14m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6177813d3ed840d38174ca7a0e2e7f0f : SUCCESS in 8m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1ddf73f9109e4ce7b92413753d112ea4 : SUCCESS in 13m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0adeea27ac8a4ff997679bf4ab51fc8f : SUCCESS in 5m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a5fcbee231aa480ea38a204358151177 : SUCCESS in 6m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bf2252cfdcab4ba195c6e0a727a8fde0 : SUCCESS in 7m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1edfdb47c1274b2d9d54be2c5fe7c7dd : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83023c305ccb4985a1d0672d7e0c454e : SUCCESS in 8m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/aae8b4ae994c44f785c7a372da50ec5e : SUCCESS in 8m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cd0552c8f0734f4180ff9367be6a6967 : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1aeabe3a5f5a4367a436c7427e2f716e : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8219821fad1b437abf8f841fb489576f : SUCCESS in 10m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bffcefc383e54683ae1e7dbcd74e65b7 : SUCCESS in 5m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/35ed1a18080c471a95ab8fbff4c14a3f : SUCCESS in 23m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/87845d16791c45afa7dffa27aa1e2cb4 : SUCCESS in 10m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/861bb791d9ae4103b82a7344118d66e7 : SUCCESS in 12m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9e0568f5a9ea41cd8a5a788d7aad6edc : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9617759dc7364fbea4c7414fee3aa858 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0dbc41ee93c438e88334c7972b47ed0 : SUCCESS in 26m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b2b7e73e789e450589139cb96b4bbb44 : SUCCESS in 6m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bc74f06b1db74133b1b8f650ddab0d72 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6542fd750304b8eb7b47515620e59ec : SUCCESS in 7m 29s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a1cadd8553974301bac033bf299475bd : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d6094e1768740458ab87b547b0f21c8 : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7de958213c0e438f8ac30ceb9c0ba04e : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d2405f1ed66dae99003cf903e8dce98121111e54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 00:06:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c5c7277028bab848010d1c211cf96fe89504b3a","revisions":{"4c5c7277028bab848010d1c211cf96fe89504b3a":{"kind":"REWORK","_number":1,"created":"2023-08-29 13:21:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/893060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/893060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/893060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/893060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/893060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/893060/1"}}},"commit":{"parents":[{"commit":"130d068736a6e4ac6039c3d3520d1d0e712ba0ee","subject":"Merge \"update constraint for taskflow to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130d068736a6e4ac6039c3d3520d1d0e712ba0ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 13:21:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-29 13:21:27.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 5.3.0","message":"update constraint for keystoneauth1 to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie51b3dce149245a8450f254d24ef1b99ebac9e7e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7080abace3fe17850c62e0baa961be8f7092a60b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c5c7277028bab848010d1c211cf96fe89504b3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c5c7277028bab848010d1c211cf96fe89504b3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
