)]}'
{"id":"openstack%2Frequirements~893150","triplet_id":"openstack%2Frequirements~master~I2b7c995b39dfc5af28cc8cc1c0d5ca54758abaeb","project":"openstack/requirements","branch":"master","topic":"sphinx","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-19 14:04:10.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-07 03:44:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b7c995b39dfc5af28cc8cc1c0d5ca54758abaeb","subject":"Bump Sphinx + deps to latest version","status":"MERGED","created":"2023-08-30 10:48:41.000000000","updated":"2023-11-07 03:45:10.000000000","submitted":"2023-11-07 03:44:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893150-sphinx","meta_rev_id":"5677f1dcd3a7f2cb35f1dfd3c252c9194026355a","_number":893150,"virtual_id_number":893150,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 03:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-11-07 01:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-10-19 14:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-11-07 01:54:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 12:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 20:46:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-06 12:23:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-29 11:54:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-07 01:54:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"70b1ac7179bdc008c7ca800aa1e9a2f040f3270f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-30 10:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8de894abe01e650a86423199c715c7eca9c2b4aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:46:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1e69fb4db23455c96ce36d6e827b6c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a58da522ca14643b97de156b2d82643 : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6963d6c9b784427aa7d4a5fffe30beb2 : FAILURE in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/926a74aa4c214e7c96c80be39e7fd1b3 : FAILURE in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a8890098e784e969d08eae335e4fdb3 : FAILURE in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/753108d21fee4eb8aa967ca4e50a8e20 : FAILURE in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eef07325493e4e7fbfc44b7f95cb2623 : RETRY_LIMIT in 1m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90d642bdfc614dcda7c35caabcd97aa8 : FAILURE in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7823c9d3db034d37b37ee06294b051a7 : FAILURE in 4m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b34174ccecde46b6b9c1fa691e31d2a0 : FAILURE in 5m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7efa0672cb964d0aa726e5025bcbb6cf : FAILURE in 5m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1272c327cba3470a9952ee45b94024e0 : FAILURE in 7m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c9ff94721da145769c3c1763a9e0995d : FAILURE in 5m 50s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6f0d43898c9c4bbe9679bd9b6d3ec771 : SUCCESS in 8m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1411af24f0e04eaeb553749230c76ab9 : SUCCESS in 5m 23s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b1c2571711b7423a95bdd7f162cf690c : SUCCESS in 4m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9200934c08294fdba39142ac9b3646f1 : SUCCESS in 11m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8252faef75aa4551b47c013c6a20636c : SUCCESS in 5m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0eb1a679608d40ee87695d944eca3c73 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/468915bb8dc4497f94e4974e5a0ab740 : SUCCESS in 7m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ffeea4a45a99426e8a49b48482f8e50f : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1a1a2917da94a13a7b968eef0a5a63e : SUCCESS in 9m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ec5a3d73c20a40d7ab0153ff3bb9f72f : SUCCESS in 10m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0e34fe20bc344c8c9046301940b353f5 : SUCCESS in 13m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f1925d17b288446cb287a0eb33242b0a : SUCCESS in 4m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/14dc7623b49f46f3a2cad144c460780e : FAILURE in 3m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/faf9173fa9fa4b97888f1c48cbedb7a4 : SUCCESS in 5m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a60bfc7f86d24382930e6cc0dfe82650 : SUCCESS in 10m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/749e2a3c25874a4a9c74691cd51db904 : SUCCESS in 27m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9ff9e31f4bed4044b56ec60eccb56db4 : SUCCESS in 11m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d8b4d654f4a64cdba1bb6635eb14d039 : SUCCESS in 13m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8b51fd9ab3d24c64aa6ea91ff689edd9 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e60a1c8095bf4059b27d7ccb70bf5478 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c50d8e939e14583b92152b03d5816b1 : SUCCESS in 23m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fecd9e5066c54b8a8a421e36970c33b9 : SUCCESS in 5m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/95fc167699d44d308d9a3765d49b674a : FAILURE in 3m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e6efbead8a14ca586a0455bd92f12dd : FAILURE in 3m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1ed8cc6034974fb1b6649aeab5e3e461 : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d944d1845da4906917e22f05458ff34 : SUCCESS in 1h 10m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81bf77842cec45e3879f8ac4c3ed0623 : SUCCESS in 1h 33m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e570bd35404aac65c60bcad3db65a70baad6308e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:46:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29e6f3a0b269d2a106e72bea61a00d7324d21fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-04 10:17:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3737b25da51aef24be4eea3d963ac1ed78e58625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 12:08:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ba7601d7db14b2ab98ee46708319251\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83229fb4d9524bce91450760fe45cf6a : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad5e6cf322d402ba17dde305497d4df : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e91dd4cc65a340d1b7db5f932dff87cb : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4514350c56c54b87b161c520fc907b00 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7aa6bd8b1cd84064806edff85e1c8162 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97513918bd4a42c095f58f480cdb5a53 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29392ab5e87a40df84d9caefa79cd2da : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2a72c964f12423bb2c71a1501e459cf : SUCCESS in 3m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ba5f884c84464f8aa328671d3caf2399 : FAILURE in 4m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dda142d5b09141e9abf1a7fd2499a039 : FAILURE in 6m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6bad2148b6414fcb87aea8b22df4656e : FAILURE in 6m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bee1efc2352d415f8a6d0896eb1d8f6c : FAILURE in 5m 08s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a40e8998f939421cbed28424417d84a5 : SUCCESS in 7m 28s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d3bf57804a2e4cd3b6190032d3b17028 : SUCCESS in 4m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/adf311b2d38d486a9033457ca1da53d2 : SUCCESS in 5m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/771f51dc98244775a8238b562631db1a : SUCCESS in 9m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6a78c178ce834e9591e80738d8659b17 : SUCCESS in 4m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8481b86df0ca427fb5bc6a3977af88bf : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d191ef5225dd450eb52f6850e2b58a43 : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b324b67fc26c4252a08359dafe737024 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb967b637cde4e15a5bf4a9c11e06d0e : SUCCESS in 8m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c0ae98c8179f4347b68808ede71477b2 : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/65e80a36e8c84f98870fe3d5d2f4c611 : SUCCESS in 16m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e3c2f8daba774109a910284bd4934ce6 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5551015e402b4fa7b5f060797d975793 : SUCCESS in 8m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ea37b252e54e43d39b945fd874a2dd4c : SUCCESS in 5m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1a6828d0bc834256bc10df05e0ef52f5 : SUCCESS in 10m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/08ced189b3bf4467a5ed6d60e9f6dbb8 : SUCCESS in 25m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5709a421e38a4ba39718b49f50517e10 : SUCCESS in 10m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e76865cf65424544864a862be7350510 : SUCCESS in 14m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f2c4e5eda28c4cc799282394de35c81d : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18e38b7aaba44d2391ab70103d6e6aa0 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/622f7fe880c64d2b95ca98c51e219ba7 : SUCCESS in 24m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7d378940bfe048ab85ae84d73a573cab : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a894c3fc30534141b989e9e265c3b99a : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/535012c5af8d4724bf485df1b2b135d9 : SUCCESS in 6m 54s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/00e3889e49964a318f4053dd61a84978 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6998f8f958b44300adf7092e9329b743 : SUCCESS in 47m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fd8fa24cfb541da8c8c6ac717cb50d7 : SUCCESS in 1h 45m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e26971343b6ee30fa150e63efecb54d656bed3ea","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-06 12:23:39.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d683469ea21a66eea0071c06fe7bf0155245a78b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:19:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a809bb5bf0a745899f3cb30fe7e4df49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660627d3a8554a57b0a75c5d131b1470 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adcdaf060a6341d799a852a278ff807b : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5137a9d53564520a8f68f805ad0fafe : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc3f884851cf4f4eb0eecad794630883 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ead62f6d7d8844ca930d5669a6d3b59a : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50cb1615b88343cdb3763ff060752168 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/207b4fd3d11345a1b132cffb3cf58745 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/360d53f2b6464629939fdf062cd5ef2d : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a7b3cb3feef14938b5eea9d333741e5b : FAILURE in 3m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b4476d2117ab405388f308bfe5c1356a : SUCCESS in 12m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/44ac588023724b7cb7059746c744d6df : SUCCESS in 7m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/37e2dfa249824f9389c206d9dc60f26a : SUCCESS in 11m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0e77f6591da84465be9af732dcdbb0bf : SUCCESS in 7m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ffdaca04b8a84414a2a9c7c2260ff455 : SUCCESS in 5m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fbac09ed1f374a3ebbda33f40e9de806 : SUCCESS in 4m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f98360fedf2b4624bc454f857504b69f : SUCCESS in 12m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9e76302b3b9e4d42aafa20c82e3bf41c : SUCCESS in 5m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b9aaccbb03f841e1a314025fbaa4c4ca : SUCCESS in 5m 33s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6a249ec416a54400a3de2c542bd0b947 : SUCCESS in 8m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/16601ba620244721889ebb6567cdb6ce : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ba2084040da4f059594f5cde1545748 : SUCCESS in 7m 43s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8fbba87d0c45462686b3454dd467a9f9 : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d655f1ab427b49608c299d841829a8e1 : SUCCESS in 12m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e8e7eb34f77b484db08c1351781c9dbf : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0c08ecda19ed4254ae3c069041db5614 : SUCCESS in 11m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c5feda98e5104f9ea0a5e7071417224a : SUCCESS in 4m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/46f0c16e9bb243ada93fb6041199cd15 : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e4820762304449ef9478ded5ab9ef2ca : SUCCESS in 25m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2adfa8ebc55d49369338b6679a0f6df5 : SUCCESS in 9m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9c9d558aa52a43898499a36135905f8a : SUCCESS in 15m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7f40ee9b0e4549aca481ab827c6c8287 : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/424b26f82a2b4d358eeca5f7136d3dab : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ae34282283947d0bb6a44e686dc877e : SUCCESS in 24m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/149ac62e4ffb43f4a07a1553d19ffa42 : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f79943b4711949d1b147241da636740b : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e3e3af9ea334ed4859e76bce400fa1e : SUCCESS in 7m 58s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b74c50c56f434d849c9f2c9fe5f3c73d : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb7e1adcd1ff4f9ead9ba0aed4df45af : SUCCESS in 1h 16m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ead77fe2f4e3481f89d535f16e04b1a5 : SUCCESS in 1h 48m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8b9ae3591efc722db72694803fc58b209d154b1e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-29 11:53:36.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2dee3f02e8b7d2baded48531c15ad80027c6941d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:53:51.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893150,2","accounts_in_message":[],"_revision_number":2},{"id":"99ed69b420fb3482f3f1d6a3128053117c14a4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:53:55.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893150,2","accounts_in_message":[],"_revision_number":2},{"id":"2554b1983e117341ad7bb672cd5f4d52669eb387","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-29 11:54:02.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 893150,2","accounts_in_message":[],"_revision_number":2},{"id":"d3fd383344d0b4b55e0eb34ec64ba9347b6fcd53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-29 11:55:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"df512a89f13c64e9562f786f3778117762a46e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 13:25:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce4905ec8b0942cb9b2914bd2de284d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c791b93567374d0f8b6170541f1940f9 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d988bab3ee1d4fcfb166a9b237fcab2c : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e305b5442e84492ab36fb3c8459d0e39 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fafa30da9c4478e98d3bf48f1fd9318 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e4a4854bbf54b12a102ed450a13fbc3 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fed613be2aef4444832daa80050fde12 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/110cedef19654b0dbd75d903a0528aa8 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3ee9d335cd746699732db5701e1a310 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1295130b5aa14054b6a3a647c1de3742 : FAILURE in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ed93cf08c3f0493ebf4a63136b9533d1 : SUCCESS in 12m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a90d3cbdbc114b4fa770581c17e61d03 : SUCCESS in 6m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5d8325d8e2154e50a1009594002ff5af : SUCCESS in 12m 00s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a72e43adb703440598d0b4080e86384e : SUCCESS in 7m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f4cbd874a8014e4b84dab899dbea17a5 : SUCCESS in 4m 59s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5b36dc73916d4d05a6d1332f406baa21 : SUCCESS in 4m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f52b28e0952b425c9f4cd29ac91dcbee : SUCCESS in 11m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dabf53663fe3404793c74d69c041c59e : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fb8f7fa215154e3fbc30664193d964b7 : SUCCESS in 4m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/100dcefff6c9491c9ecc91cda155a08e : SUCCESS in 8m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1a20ee6c1185416c97c624ff28f21f9b : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/542b4f8a837b497da82dbe7d40d72c16 : SUCCESS in 6m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/84687ea256644cf6997e662644d412e9 : SUCCESS in 8m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d7949f9cb9924bc8a4de357ad2c9b165 : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2dd0a1b071b64005be89cc8eedece7e5 : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1eb1cc137a414b2d8ee828f9b3baff57 : SUCCESS in 9m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ca18ea584e374e6083ca49b740a455d9 : FAILURE in 4m 08s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4cc31674ed70444a9f0d51e8a9e692bd : SUCCESS in 10m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1c234ec919c843ebba7bbd0c7245c554 : SUCCESS in 23m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/99f69c9cf6234a50bd6bc78782b6ec83 : SUCCESS in 10m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4064e39590e749458026017346a4776e : SUCCESS in 12m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d43744b0a9ec4a8cba9d5c40911785cf : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c24f19ba19b545c9a2b6bea468cc397d : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d31c6cf4cad4f02a5d466db77047714 : SUCCESS in 20m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/06c7ace2f9dd4353adade13d620cfd2e : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fb4ff72ff6004de081e884ed6867a0a3 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97bbee69805845cabba2160a5b272ebc : SUCCESS in 5m 38s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/cee5e207e4b84f66b8a8b04daef72ead : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f0d68a0824d4c9b83aead67df25a172 : SUCCESS in 48m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a5fecedeeef4519b361d6f7244628e4 : SUCCESS in 1h 25m 22s","accounts_in_message":[],"_revision_number":3},{"id":"db0d61622da640204900c6051a6dab60162ee05a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-05 16:56:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8291e534dc23a5c9bd06780cb7820519c83b8f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-05 16:58:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fd221799f719fd42488f5d078c91d7c0ff28dcff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 18:14:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b21676dd5d94711b689d8f5f7ef1c86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0993b6570f14f2996bcee3c8440afea : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a9e6988eaa4b3aaed689119624e25b : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60e875bdc164827b98ad40c267e0e98 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a004074b6d4340d28982b0c12324c9a4 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18443dba2d264027851f7ed114da5e4d : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ced7ab1b0c694bda9a324cc718bc3939 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc0d7a471a7d4e44a6a41ecdc8d04c79 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c0b8857840040b99ca07ebc8598d78a : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6547172dcbdb4fefb31c67e80e5e184d : FAILURE in 4m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b0bc237d54c84a308be74b23658b229d : SUCCESS in 13m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/afa9e897244e4d6b97d45593115cd9ce : SUCCESS in 11m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b0fe422a9d724bc4b3692b0318c6fda9 : SUCCESS in 12m 55s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0ea843c6e1de470a8a696ba7f0078680 : SUCCESS in 8m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3cb844f8cc434c8b8cb11388b4e21853 : SUCCESS in 4m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c65f5b674a3d47b598b0128615af8821 : SUCCESS in 6m 09s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d808dc750b884c8999c32a8eefbef8ef : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4084427441f44a2d88a8baa0ea77f425 : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8928b18f88ca4b5dabd5d75d4ec06633 : SUCCESS in 6m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eb654fc2b5514081844d4f37c027de6e : SUCCESS in 7m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bf09ee8dd9d04c93a3be4e4731032c7b : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08fe161b2aad4f179c862637544cd749 : SUCCESS in 8m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/69fb69553a72451cb88aa0f8aa20d8ed : SUCCESS in 8m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d20adb003a0f4b138dc7b251b2aa17bb : SUCCESS in 12m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0608dc258dac468ea337594e87e04b61 : SUCCESS in 4m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fb0e5796a87345bb938620e4f7052412 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3591db161cae4c2c9dd31e0bb0194b08 : FAILURE in 4m 41s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/aa7b370027e944e9bdb03cb4b62c0929 : SUCCESS in 9m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/21cefa8abb134094b660a53d15cf872c : SUCCESS in 22m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a045929c57b84efc9fc197c591dec01a : SUCCESS in 9m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4ad464d712d84b34992192a78ff69cb5 : SUCCESS in 12m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b3790910923747a99122ccac277394b9 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/683be70270cd4c2abb844b6ed5a6fffb : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0996b211dbd448af9b9db28b6046d599 : SUCCESS in 19m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d2dcc0aeff3b4f9f95869ec07d296bb8 : SUCCESS in 7m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/61446ff19b46453fba6bba700061230b : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d65cdbd1cc554cf4a6958639d7030efc : SUCCESS in 7m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bf4579b0d56b4c148cea3bdbd923a34b : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd74b0cb7a274b39bd23b91c0c7cda82 : SUCCESS in 1h 03m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1d515fe65f149269e97124ebf67836f : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":4},{"id":"4f79fddaf0166ad98ff821dad84819ff90735afe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-10-19 11:13:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"05ac4684e63807103d380334bb1a3adf3f2d3b00","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-10-19 12:09:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a67291dcf5383dbb374a2094a353a11c21e4ca65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 13:01:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e34d0602f40475daebdf1d227edbec4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a822b9a3cc24a4bbc745a430aa771c0 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d88447ee1f4a4996a7323d9555f7fa : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1660d30a66242a8aab1bd147f5bc9be : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea20cd86d72246b18e8ddfec0ede8f45 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8c458f4c1824c99b187f9b619783f37 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fda1f92adf3c4c54a56d38058e910dbe : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9afb7c5e0ac5466f908752bf448538a4 : SUCCESS in 8m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fae1b83d28df441e877902b966921c34 : SUCCESS in 11m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2857781d4a9a44d9bb2bf2c176a8ffd5 : SUCCESS in 10m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/480a9b8b6faf474da4d840f7d94308d5 : SUCCESS in 16m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f175e1b93bd84ddfbf3f5480c25b66b7 : SUCCESS in 7m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6908439b96d14d459f7ebc80acb7a206 : SUCCESS in 4m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/54c8478b78dc424e89f46efa946c4124 : SUCCESS in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a5a6261a18cc453f91da5066d1846910 : SUCCESS in 12m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6d3046f1d55b4fef992ba25c1f3e3fd8 : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/73fc0bc3342d429d8ae4557e840bac2f : SUCCESS in 4m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7fe7df5024ff4f51b516e5bfffe62d65 : SUCCESS in 7m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5bafa2495281453fbea9973abba3df47 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ebc23840fa047068e7779adf0efffbf : SUCCESS in 7m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f25886a46baf477ba9d29e762bd31d99 : SUCCESS in 7m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5c8185f2c13545eaba933c31c1608a9c : SUCCESS in 11m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/16670ddd579a45aab435c88d8a32e62f : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2df8ca7c21b94fe18131717f8dc2d0b0 : SUCCESS in 9m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fba839e5d68947f29d37ce6f37a5d20a : FAILURE in 3m 51s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8d57fa57062241a4980f3ec1d06817de : SUCCESS in 10m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/43e7543cd3b249b5a738b2041a49ba3f : SUCCESS in 23m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e23870d0f6e4663bad49e2b29c0e325 : SUCCESS in 11m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a840a0b23bae4fd2ae0a3a304f8502e0 : SUCCESS in 13m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ecc5f790a28d43cc9576560711737bd4 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/018b219d2eb6443293a7f6bdf04d6985 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea4be6bc1ba641bb94cd558d39623aa0 : SUCCESS in 24m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6c769c1341a9406b8ad90f9f108d3417 : SUCCESS in 4m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fa4e940c7fa840ecbf7acd6675ce6dd4 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e6dc9f9ca6845f4a563fbc3fdbf7533 : SUCCESS in 6m 19s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/114a7745cbbe47e290df2f48f1d48002 : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46c4ac8321e24cafa82a36dd1cf397d2 : SUCCESS in 1h 01m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2daeed97bcb24a1f8f0547435745b419 : SUCCESS in 1h 40m 00s","accounts_in_message":[],"_revision_number":5},{"id":"217d19785598058691952432a280664d4093ded0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-19 14:04:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ad58003ca409cfbf311e88a3b4a59baffb9f88be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-07 01:54:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"723f3d6f5151647477821d0c21096ab202236c46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 01:55:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b53c37634edc7c509b90c5d0022e89b11efa39ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 03:44:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2c36c0602904074a0093de02a48f391\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a60e68caf2f4d68aa36a97b211f1bd1 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43944222ba0841fe829f3b92af83eb52 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a5b15fb8a9c4eecb12e583035185c31 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a54a3f318b0141858e7bbb92282e56bf : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/511493561a214f159882698c12112f02 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9899ce83252c416caa529015ad32d5b0 : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/baca7356159b4c6ba93d37dcef0a8eb5 : SUCCESS in 8m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/64f1ba05a5c84b00968bc1e6e8b9783c : SUCCESS in 16m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/18e802d3c5bd4755858a34df0cc542b4 : SUCCESS in 7m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3c28e894fc4f462db58e1853be7099ad : SUCCESS in 9m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/df2772e447c64c6bad453c3b95c860ab : SUCCESS in 5m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d969d3fe94ab4713bb768a562739a4de : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/880a5a07bf4945c9aae29a2b04899672 : SUCCESS in 9m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0a901309df324d6a98c1158308a1731a : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6bebb58a88e2406dad3bc00dfb1c02d7 : SUCCESS in 10m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cdb985ecb94e42ce9786f188d194c884 : SUCCESS in 11m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/62197c88f2674112a5927a6fead62304 : SUCCESS in 12m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eefb236acf7240a389b842fe0e2ba235 : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/98a9e6efa8964bcfb928fecd7bc22c0f : SUCCESS in 11m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9c3e1d17527f4c1281bdc91961c329ed : SUCCESS in 26m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8c21adc4d704b8d8cead7e6fb6860b6 : SUCCESS in 12m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9c916743af0049a488122ccceb887a9b : SUCCESS in 14m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/13b9404bab314a2a9cbe190e37505c31 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a87c70231b342b3bb79d3dbdaeb52c8 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ce84ed376b740cc95add3649e34b464 : SUCCESS in 24m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e22c4dacb5554a3cbbbc98fc00c1b274 : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c3d14a1744dc475b8a79a54120d9db41 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d9f3de97226e4b2d9d7ef8c376d7be48 : SUCCESS in 6m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2b170d7813754fc2a2430e7b478a77e3 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5968b5b1afee49f6a2870df8fb61976e : SUCCESS in 1h 11m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59cb131073dd476d9205e79b99693703 : SUCCESS in 1h 43m 49s","accounts_in_message":[],"_revision_number":5},{"id":"6bd5b546c7f6d92796092780726e3cda90a3293e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 03:44:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5677f1dcd3a7f2cb35f1dfd3c252c9194026355a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 03:45:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9a7d6f1e8c44cc2bd508ab1a9de382b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccfd434ab54449ae90f8e99d91deda3a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d1c5e98c0e151f089ef31b9a74a4a87770f27589","revisions":{"cfa9d5c1e0f20859343a1acf5a37c759b4e66817":{"kind":"REWORK","_number":1,"created":"2023-08-30 10:48:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/893150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/893150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/893150/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-30 10:48:21.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-30 10:48:21.000000000","tz":60},"subject":"Bump Sphinx to latest version","message":"Bump Sphinx to latest version\n\nopenstackdocstheme is updated now.\n\nChange-Id: I2b7c995b39dfc5af28cc8cc1c0d5ca54758abaeb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfa9d5c1e0f20859343a1acf5a37c759b4e66817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfa9d5c1e0f20859343a1acf5a37c759b4e66817"}]},"branch":"refs/heads/master"},"bd57b99a578afe5df0ac138f47eb1da0c1069226":{"kind":"REWORK","_number":2,"created":"2023-09-04 10:17:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/893150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/893150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/893150/2"}}},"commit":{"parents":[{"commit":"9463129a82a49fec61ad2ba536ac904cf50fa48d","subject":"Merge \"update constraint for oslo.middleware to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9463129a82a49fec61ad2ba536ac904cf50fa48d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-30 10:48:21.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-04 10:17:08.000000000","tz":60},"subject":"Bump Sphinx + deps to latest version","message":"Bump Sphinx + deps to latest version\n\nopenstackdocstheme is updated now.\n\nChange-Id: I2b7c995b39dfc5af28cc8cc1c0d5ca54758abaeb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd57b99a578afe5df0ac138f47eb1da0c1069226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd57b99a578afe5df0ac138f47eb1da0c1069226"}]},"branch":"refs/heads/master"},"b7b5e96cf1625f97d3ada127283140f3b7a72f58":{"kind":"REWORK","_number":3,"created":"2023-09-29 11:55:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/50/893150/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/893150/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/893150/3"}}},"commit":{"parents":[{"commit":"a3b055de9631c3149f721eebdd74740be06f5b75","subject":"Merge \"update constraint for oslo.db to new release 14.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b055de9631c3149f721eebdd74740be06f5b75"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-30 10:48:21.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-29 11:54:53.000000000","tz":120},"subject":"Bump Sphinx + deps to latest version","message":"Bump Sphinx + deps to latest version\n\nopenstackdocstheme is updated now.\n\nChange-Id: I2b7c995b39dfc5af28cc8cc1c0d5ca54758abaeb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7b5e96cf1625f97d3ada127283140f3b7a72f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7b5e96cf1625f97d3ada127283140f3b7a72f58"}]},"branch":"refs/heads/master"},"3d0ed484a9cfaca0934f317b84c7e73b819a3927":{"kind":"REWORK","_number":4,"created":"2023-10-05 16:58:22.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/50/893150/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/893150/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/893150/4"}}},"commit":{"parents":[{"commit":"a3b055de9631c3149f721eebdd74740be06f5b75","subject":"Merge \"update constraint for oslo.db to new release 14.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b055de9631c3149f721eebdd74740be06f5b75"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-30 10:48:21.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-10-05 16:57:58.000000000","tz":120},"subject":"Bump Sphinx + deps to latest version","message":"Bump Sphinx + deps to latest version\n\nopenstackdocstheme is updated now.\n\nChange-Id: I2b7c995b39dfc5af28cc8cc1c0d5ca54758abaeb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d0ed484a9cfaca0934f317b84c7e73b819a3927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d0ed484a9cfaca0934f317b84c7e73b819a3927"}]},"branch":"refs/heads/master"},"d1c5e98c0e151f089ef31b9a74a4a87770f27589":{"kind":"REWORK","_number":5,"created":"2023-10-19 11:13:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/893150/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/893150/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893150/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/893150/5"}}},"commit":{"parents":[{"commit":"79faab4a72a5876090e9dff295aec99e3c43176c","subject":"Revert Pillow version pin to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79faab4a72a5876090e9dff295aec99e3c43176c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-10-19 11:12:59.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-10-19 11:12:59.000000000","tz":120},"subject":"Bump Sphinx + deps to latest version","message":"Bump Sphinx + deps to latest version\n\nopenstackdocstheme is updated now.\n\nChange-Id: I2b7c995b39dfc5af28cc8cc1c0d5ca54758abaeb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1c5e98c0e151f089ef31b9a74a4a87770f27589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1c5e98c0e151f089ef31b9a74a4a87770f27589"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
