)]}'
{"id":"openstack%2Frequirements~893167","triplet_id":"openstack%2Frequirements~master~I9cc3ed60d08f1db7377256d40e8b69f2f3cce381","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9cc3ed60d08f1db7377256d40e8b69f2f3cce381","subject":"update constraint for python-vitrageclient to new release 4.8.0","status":"MERGED","created":"2023-08-30 12:41:32.000000000","updated":"2023-09-02 05:33:05.000000000","submitted":"2023-09-02 05:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893167-new-release","meta_rev_id":"5d38b97745b4f90da632167813f87e23761b77a8","_number":893167,"virtual_id_number":893167,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-02 05:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-02 01:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-02 01:55:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 14:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-02 01:55:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"171771c9c402ce4f709d3bf58cfa99b533fdb0c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-30 12:41:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b91893391a5734fbe04bd72de9a2240fdfe972aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 14:46:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11b0f338fc55469b85d94252aa302e98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a522ab6a51cc4c1184dba3ceccdceb46 : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda360b2003447e9a8e07af3225795fe : FAILURE in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15d64cf0dc2448feb8d9af16c0c1b48f : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d96dc90628af4953a13577593851959e : FAILURE in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31e0e19fb10c41fca0ea9f13c14dfedd : FAILURE in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/769f0d0f31124d6884e2f611b62a2837 : RETRY_LIMIT in 3m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0131ae36e37e4f60b879addffc7785bc : FAILURE in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c8dd10ca22742579c750bcf7a4e53d2 : FAILURE in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b38278d963d245b5a038465c20db5f3b : FAILURE in 10m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9ad21224f0cf48948f6bb746c2405f3a : FAILURE in 16m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/07e9ce25f6f84ec1ba4836cd94bd6b3a : FAILURE in 7m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4020f3ee05d4428ca6d3a04f4da159fd : FAILURE in 11m 35s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/44858e5ef28f4698aecc767aa555bb14 : SUCCESS in 6m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e9c80c9daf064092860bbe1e70c22dc3 : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/deda7dc1f4524f0aa8a719f397e73a53 : SUCCESS in 5m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/db8bbd041b9d4a01bee64a9b66fb2fba : SUCCESS in 12m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/69cf2f74c9bf48dba9e83a9715eebcd4 : SUCCESS in 5m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/67bc49b91d7a40eab7fd31cc859ec1be : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/70365597d17343bb9438f7169805183a : FAILURE in 3m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3699d11328954d78aad409454f7fdb26 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94a12cd0ea9a41a393575abb9576f764 : SUCCESS in 7m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ddc4687e48cd4d879749a056260e13b5 : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bf2ea0373e7242a0b44f08f9e2f9aa09 : SUCCESS in 12m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3503afa330e54b27aefa93f9f543fccf : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d047a0928b4948f7bf124cae0f231f68 : SUCCESS in 11m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/85aab7f57ea54ad58ef381bf15fcf2b2 : SUCCESS in 5m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ad58d64c5d4a4602a368bd8a2dc7de2d : SUCCESS in 9m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/db29cbd850fc44c58ca229637d36c14c : SUCCESS in 26m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9f92e9cc6ad54ff2aea09c28a169bd1a : SUCCESS in 11m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e16db751892349d8bd5e1f4de75ea198 : SUCCESS in 13m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a5ca5051854d4085b9192d73e6d83153 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29e837df8f794b1cbe5e0058da2884c9 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f3bca1beeb545d08aa1da598c5ddfc7 : SUCCESS in 20m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e99494955c3a4dbba8cf85dafb184a1d : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/72d93857be1b4d23b6cec3b306c38960 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/649b3cf1058548768746d322c9c32b33 : SUCCESS in 6m 38s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ed396a7567e648b99df86ba194929add : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2ebc589b5da428ea3050d4860e69f7b : SUCCESS in 48m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a820832b92948178345c31b0837c731 : SUCCESS in 1h 55m 59s","accounts_in_message":[],"_revision_number":1},{"id":"31613bd579e45c4e371524c7dca38b1abce59c34","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"48a739b8823c942d7f4f44dc96f69028a165fb08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:56:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f3e8e2a3e14543963605c8917632dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892d5fd88e8a4431b5eef5fa476ea43a : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5fd8b082d244363adf01d2ca7322977 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4c1479aac914087bde6e239027e0b31 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e224326ce9bc4c9ca00c3911970ae650 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a14bc02fb7994cf49037431a33ca2bf0 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ebad2d114e34859a8d594c74cc8c0f3 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9600874fe2904758b583a59c2b8732a2 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30945a3c64a84da197694484bd9e4f93 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da2aab5fd8a94ebc8f68df4fe743e9cb : SUCCESS in 9m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7ef53f1b60654495826319ade30a2490 : SUCCESS in 13m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f92baeaca0f040f282b549af8c29bc48 : SUCCESS in 10m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cf2766352bdf40119558153aca8bf532 : SUCCESS in 11m 32s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/83c165b953cd4a60bf7c3b0c4bb379ed : SUCCESS in 9m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1c7c0e2915c64d97926a5c78f7e64b0f : SUCCESS in 4m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a5d7983b61dc4b2991e9e0f7395e4d74 : SUCCESS in 5m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/08bfdb74d82949e2a35ac0a00e67628e : SUCCESS in 13m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8a8c30db6c4d49d5b4724475a934be92 : SUCCESS in 6m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/345e22b6108148da8f83992e94918c92 : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/667a2e43c9c244e7ad9c449424080fa0 : SUCCESS in 8m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/25075ae0a9e645cb80ba57f2f6248687 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48016fd4f7c041e483bdbe119d5a520a : SUCCESS in 10m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9cb70fa126c64f0e9b4a213333a6a6ad : SUCCESS in 6m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e001a84058d24131a5d39dc2b9da2570 : SUCCESS in 14m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fd6b62b1b8d14c3cacc711fc96c4cd76 : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e32057d022a64541baa7740325a4618b : SUCCESS in 9m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2b60dadbd723415e81ecd17265217667 : SUCCESS in 5m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e981f31ab5524320894cb900749d1a6e : SUCCESS in 10m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a7206624f1bc4b4fa4ae5e801448e9cd : SUCCESS in 22m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/35271431d5744c4992a0a65c158228c7 : SUCCESS in 10m 35s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/953fb3585a2a46a7ac30d2eda93b6ea1 : SUCCESS in 13m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6c210d272e7b452cb4b448b959298a4c : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f73485beb6047be9f0d70cd1e091199 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6ec7e4c7ee44f4eac83642c03e233c2 : SUCCESS in 21m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/90d449da67624cb8b08b9c05442ae136 : SUCCESS in 6m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7071be2adcd4431d9c06b76bfd8c7e20 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fde5ac90c2fe47c5ace713ec34c25634 : SUCCESS in 6m 42s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6e102352ee1e40e1a5db85743afd474b : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b6b6d080d644479a083ceed1db49e47 : SUCCESS in 1h 01m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0f84dbd19a74b49986d5b86758576fb : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c4fc838e72f200b80c143301e19b78284889eca3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:57:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01c5b43d41303fb99da4144b25c44fec334be91d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 05:32:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b3f5113e3e4aad85f63c7abcec874b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53489230dc914dd687d473f26a94f818 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738294d109db4edc81d7d0e0db31f7d2 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb8ecba078c54cd8940dbc8e14e105cc : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/598ba6a09cec4fe8869361507390019b : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75021800744c4ae7ad6f6602ebba32e0 : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0397f4b32c564b178cafbf6d52c8ba7e : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/821a7cef8f0c409fb142347a8d7b239d : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7dfefa08fa614e19992c1ce05d5bedb9 : SUCCESS in 8m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4ef0b0b1782f4794ade7542d1cdf08b9 : SUCCESS in 10m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4f50ce3959064db0a0dfcdd704d36bd6 : SUCCESS in 13m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/326e2793b8eb45ffafc4a0bdbbe72d1b : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/81f1ae51926f473e9cc115591cc2a690 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/40e7986135c44c7392d32d698a47592c : SUCCESS in 4m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e2f988f60dde4efeb0cb69f4defc6b57 : SUCCESS in 9m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c60b9a41ab6c44d2a8b6d85314d058c5 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4cdde513936b4885b93fbf42c990573a : SUCCESS in 7m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a5d2bebb42d54cf5b0f42f418d8dba8e : SUCCESS in 6m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b04f66124a224dfa8fd6391778dc9cb5 : SUCCESS in 12m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/89906c63a6b64a9f950a1cca49558564 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/65146f159ffa4c329546f67b1e056a55 : SUCCESS in 10m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/edd9147b53fd4be5b2965fd6207f02d6 : SUCCESS in 4m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/83d51c9c9c324f8abff1e4e0aaa8a9fc : SUCCESS in 27m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/afb6b4173f774fb9912800d39bc3b47b : SUCCESS in 9m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/27d96515025f483eb512694c8875c1b6 : SUCCESS in 10m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a694227eb5df471ba5434abe0557ea71 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/442fd86a51e24afa8b63769386663c05 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4238274133147c78d0b82849b3046c6 : SUCCESS in 26m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f72134465b134e5a88301b66f6242154 : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8c4ba121b5d747edb990225eae959bf3 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/998c181dd86049c299395a2f8ff584da : SUCCESS in 5m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/29b063ba9a5e4d5489cda5bf154537dc : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a6167d6a8f44a6a876fcbd6a8292955 : SUCCESS in 42m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/153057e3fcc54e099b32ee8c0d6f2b09 : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"61da2ce8c6ef8ebc92eb3fb3bb00d635f43b7d00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 05:32:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d38b97745b4f90da632167813f87e23761b77a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 05:33:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82765bd922d9413cae77ba0b231da662\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c79afcf74fb4e288d397e64ea2f801b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"868c7312bf9d750afb99700b2daed42c79709bab","revisions":{"868c7312bf9d750afb99700b2daed42c79709bab":{"kind":"REWORK","_number":1,"created":"2023-08-30 12:41:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/893167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/893167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/893167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/893167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/893167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/893167/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-30 12:41:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-30 12:41:31.000000000","tz":0},"subject":"update constraint for python-vitrageclient to new release 4.8.0","message":"update constraint for python-vitrageclient to new release 4.8.0\n\nmeta: version: 4.8.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I55c38f2113b1dbaddca308740eadc0221d579e7e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9cc3ed60d08f1db7377256d40e8b69f2f3cce381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/868c7312bf9d750afb99700b2daed42c79709bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/868c7312bf9d750afb99700b2daed42c79709bab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
