)]}'
{"id":"openstack%2Frequirements~893347","triplet_id":"openstack%2Frequirements~master~I7e7c06839785f929f6d1296a413549ee543d2598","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7e7c06839785f929f6d1296a413549ee543d2598","subject":"update constraint for automaton to new release 3.2.0","status":"MERGED","created":"2023-08-31 10:10:50.000000000","updated":"2023-09-01 03:53:46.000000000","submitted":"2023-09-01 03:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893347-new-release","meta_rev_id":"cccc884a201bcbcd2934f6938a3349628d58179c","_number":893347,"virtual_id_number":893347,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 03:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-01 01:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-01 01:52:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 11:44:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 20:48:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-01 01:52:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0388d39dd402724b1c6c7142a771ca4acb8903f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aafb37b5ee848f9e4cee7020c90b35a3d03c0692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:44:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75a7820c48eb4255a25fcc7044f83147\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf6079ccffe44cbbd1d0602fcfedb26 : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f69e4f8c9cec4b8bad009e1c1c39cc13 : FAILURE in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6648bb95653e44e4afdd9c28bd292a86 : FAILURE in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59377eb3fa8c4093994628a77ab5b518 : FAILURE in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/111a6a41456044c3b53e3f49064a7b37 : FAILURE in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be2e1b05325141a681e4a5a899605336 : RETRY_LIMIT in 1m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9be63985f2b247f1bc98aa9d7bedde62 : FAILURE in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c977491b735845b2b0bc95086c31eb07 : FAILURE in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2cdcf0db4f9b4f6c88a12aee05474118 : FAILURE in 8m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3263df2a334345a4be820d9e795aaf25 : FAILURE in 13m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f7f0455147f84330ab68513b10d63cec : FAILURE in 7m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6743951aa5db49cf9b6bc0bd614a2569 : FAILURE in 12m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bb14dd05c1834b9ab0f71ab93baffc35 : SUCCESS in 7m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4092fcb5650641129aa4699b06314934 : SUCCESS in 4m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/36fbf41da9414671aef0c9ca5512ceab : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6dad75c0e17f47fb80a4b757a728b51f : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/02fa0b2a92c84bba95d6dde101267814 : SUCCESS in 4m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5b896ed379ce4603ae825472e63e1290 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f8b115364c42479d843185062de9b725 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3b11d13f186846f2ac114bb0bf6080de : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95963f3c38084a70b8654ce3a162013f : SUCCESS in 12m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dd1e363e59c24fa89ab48028a79b8b74 : SUCCESS in 11m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/747bd083ee3f4e60830f816a981887e1 : SUCCESS in 11m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0941d35ab374412198766dbaef88c66c : SUCCESS in 4m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/491ae60888874786851d01d3d6d580e4 : SUCCESS in 10m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a16f9e6cd7b243ea8ab54eeee5292c64 : SUCCESS in 5m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ebe59b9d0c12420a9a9942d1a11da359 : SUCCESS in 11m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c6f04def61f1467e9402d6ffed8c04b6 : SUCCESS in 24m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0341caaa90f04658abd4c1b1479b70ef : SUCCESS in 11m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/46ef4830a2b34c209b9ed33e6e601fbc : SUCCESS in 15m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7cf11cf70d0246ceab1ccc1e9e31ba82 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91fa6c426070477f8a629d41b248489f : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f49329f6bb284ae3bb983fab018e7d37 : SUCCESS in 24m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/64f85ea5ee74408b88a26cdf9b9f626d : SUCCESS in 5m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/deabcaa29e174d60986ee8c09c30fe9b : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d55183264ea84f05ac129b0828f2cab1 : SUCCESS in 6m 00s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ed6f3629c2ff4451a4e00df5e3fcd2d8 : SUCCESS in 13m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f4b8f27ff1142eb8c0c0047570da1b2 : SUCCESS in 1h 03m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cee026d8cf5c4973b3cc465ee5f13149 : SUCCESS in 1h 28m 27s","accounts_in_message":[],"_revision_number":1},{"id":"468d622545da204d23b1ea98e54bcc1cbd86f7cc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:48:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"458e08671449f4165ca27b72caa18bbce49089ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:20:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d24cdff2fd314b7d815d4971978fe351\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97d65de3a184e9e958b683915430136 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5b03939b2b4c3d8041da86004446fe : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/726e71a11989497dbe0ffe77dd8f9a55 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85bda6a79a564a2ca3227a97752032e6 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1afac715c215499ea7565f2e038c41d7 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/830950e1d26d4c1c9d5718d6633dbf0f : RETRY_LIMIT in 2m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/795253db60eb432da66c0ec4262aad35 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea770ba783b844609efc87c80d445c4a : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b1c146767f984df7a8de8317d94c19c5 : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/045ee952cb254453a54fc89f8f5636b0 : SUCCESS in 12m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8e1baf2cb2774b05a32f53cd9554bf66 : SUCCESS in 8m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8be76486d67e48a2a07e973a2e0e9535 : SUCCESS in 9m 03s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c7462e35b1c14a90beb84fea375da7dd : SUCCESS in 7m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/150ca89128b64c7cbbb6f512f481f834 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e5325b4190fa4f0bb5d1f7ec8ca4f58d : SUCCESS in 9m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/444ab0b0a5d54711be20aedb954dcd16 : SUCCESS in 14m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8183804650834d36863c023c7f8b3917 : SUCCESS in 5m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d3299c3494314992985ac6f7b183787f : SUCCESS in 9m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eb1c1f07ca384c848f573e81a5b8a007 : SUCCESS in 11m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/95f91398c77245e9ae2cff3515e5d81c : SUCCESS in 13m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e309696bba594e5689238e4390fad090 : SUCCESS in 8m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a1cc826415244a2c95bc9b8fb3344168 : SUCCESS in 8m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d6c71a7e881042f28666970dd51aef61 : SUCCESS in 11m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9268ec42f80c478da44d837613bcc208 : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/282f82c5a75c459380066dd7215d2612 : SUCCESS in 14m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b787387a91bd4f29b79443aec855fccc : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/846b882424d24accb49d8a27a1bd6bf0 : SUCCESS in 9m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a8ec45fbb7d64c1390ab9463041b7f5e : SUCCESS in 28m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/56726df2f7cd40b18af7f8806bd8709c : SUCCESS in 12m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/20d6589ce36348da9a61958ec841fb40 : SUCCESS in 15m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/61c99b93460647b0a6a368b57083178d : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/66ce6b4f37cc46f986a5e96a72025eae : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7fc24b87272649ee800238df4f214e1d : SUCCESS in 21m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/eb38e9cf39154a80aa2095df74b46137 : SUCCESS in 6m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/98a65feb851d4d778b47b3296c743632 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ce6d402cfd24f3aaac9e59ee81ce2be : SUCCESS in 5m 29s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c9da2df17b4e45b5a319393a8db3521e : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5df9deeedbee472b858f6fcf06b10695 : SUCCESS in 1h 01m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/722cce61e154465783329baed76f4115 : SUCCESS in 1h 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0f139b4439edb515465f69d3f4cd68648baded16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 01:52:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97c59153420a11c0ef949e3eebd47fa144bed3d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:53:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad74780727da00cdfe7129273581cf2775738f30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 03:52:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7d4b52791cb41f7adfe2f4f6a6995ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611414d1672b4c09b40fea08a74fa5d7 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f918f4d933db4b4fbb9c717bc5782966 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32eb8f3f5f534d4895220298cd3dfd39 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5ad1ee3b84b4e4494de35a05078e028 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f27def96fc7743abb1f7aa470c26df3c : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7362d0aebc404408905282aa2d1b666f : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77ceb90fb3e54ab8993dddbf9451f89f : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6bcfc49d441144d0ad79c618bb893f8b : SUCCESS in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4aab10f381944f9f98184068f4f7f266 : SUCCESS in 11m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c777da1bbbbc4d53a6ee766aa6f1a558 : SUCCESS in 11m 09s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/beb363b173ed4befac30d228edbce17e : SUCCESS in 11m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b37791cc448540fba46db8ae19a3f1d6 : SUCCESS in 5m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e6749c3e1819462cbcf368036413de19 : SUCCESS in 5m 42s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f605de8751f2418b81b32dc728bc323d : SUCCESS in 7m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/982b7357871e470a8ed580d3342f4945 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a76b4f46c1c0452992d7891512d79f22 : SUCCESS in 7m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/48883a109ab4476db04a3e60f903b667 : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/aea59e08104c4e9ab804555d314555c0 : SUCCESS in 12m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1bce582f9df643d1b8405df09fe872b5 : SUCCESS in 5m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b496c981cfbd45498b45e34b9ec873e7 : SUCCESS in 10m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/426854619bb04f52b7ae4d9ecb5e017e : SUCCESS in 6m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dd98c23b73bf47eab41d2b3ac28d07e7 : SUCCESS in 30m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b6a940378a294ef1be856be18ea025fc : SUCCESS in 12m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4bbdd958227e4302bd1f842f63f4efc9 : SUCCESS in 14m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1b7bdd2147444c4585138d4cc869966a : SUCCESS in 7m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31a8643208224d91abcf5367f0b92c1a : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26ef474bc899482d82d3d31fd84a2daa : SUCCESS in 25m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5d95490afaea4f7184293547adc400dc : SUCCESS in 6m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/88828e24093c4b639016846c95583e1d : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0b09fbdd6504192938edd4a06a67d3d : SUCCESS in 6m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2f4b166d2419477390d4de66d4b152de : SUCCESS in 15m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcbbb3cdf7474ea597b2c1c0ee2d1d17 : SUCCESS in 49m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b45f4e470e1c4a6f8215d0ec0b50c6f2 : SUCCESS in 1h 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8df04dac3463fa2455e1724f934017849e245a96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 03:52:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cccc884a201bcbcd2934f6938a3349628d58179c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 03:53:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc9971dc13c414483ca5a3d7baefa4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510bcd4157784b848617043e381d98ef : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2e97b491190d0d0f0724e409a6bf9a9c1dc9f6b","revisions":{"b2e97b491190d0d0f0724e409a6bf9a9c1dc9f6b":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:10:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/893347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/893347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/893347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/893347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/893347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/893347/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:10:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:10:49.000000000","tz":0},"subject":"update constraint for automaton to new release 3.2.0","message":"update constraint for automaton to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9c90a1a42a18725151288ba13ee4192b48f911a0\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7e7c06839785f929f6d1296a413549ee543d2598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2e97b491190d0d0f0724e409a6bf9a9c1dc9f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2e97b491190d0d0f0724e409a6bf9a9c1dc9f6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
