)]}'
{"id":"openstack%2Frequirements~893348","triplet_id":"openstack%2Frequirements~master~Ic1e2a1e2de59146cb5876744a800789a8d188432","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic1e2a1e2de59146cb5876744a800789a8d188432","subject":"update constraint for python-designateclient to new release 5.3.0","status":"MERGED","created":"2023-08-31 10:16:57.000000000","updated":"2023-09-02 01:10:37.000000000","submitted":"2023-09-02 01:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893348-new-release","meta_rev_id":"36280459938743c4ba3d9c265da967c9c9ca7c7f","_number":893348,"virtual_id_number":893348,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-02 01:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-01 23:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-01 23:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 11:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 07:45:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-09-01 23:56:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"14cf281e10ed71293a789a7528542d89dbedeeba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:16:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e616099f79aa587c257d3d442ee44b23ee06f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:51:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/240deeb2e5a849738cdb9fe9887a1c2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905cb52e75c340f8abdfd86db093e609 : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82df55afc034b71b6cba1f56262aaa8 : FAILURE in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26eb17c646144653a8107c411cdc1de9 : FAILURE in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c785413d2de849c1aad560582a0033d0 : FAILURE in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ba98d9081a74d5a8379821a635f2d09 : FAILURE in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10d0c2e7c8c44cf2a6f1e0053ad140eb : RETRY_LIMIT in 2m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63e4b0ba7bd840eea440a68567da7d22 : FAILURE in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6da9ad072824f539ee34402cda292da : FAILURE in 2m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ac8595d25b246e18d253b6905aa2ff1 : FAILURE in 9m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/71cdf923ed4d4245ab4c3e3f8ba9df75 : FAILURE in 11m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6fbd20e2132b4516a3b83105e11da3ab : FAILURE in 10m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a80bb60168b7481dae3b56ccd4f21f6a : FAILURE in 12m 27s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/55d46fde6af04563815995f207de1d49 : SUCCESS in 9m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/387d47eaa4ef4dfda28a604c1f4ec37c : SUCCESS in 4m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d0d09068e7ee4009a59c09ce956e9e33 : SUCCESS in 4m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/40993b55668c4b07a3565b0084a40e82 : SUCCESS in 11m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7b57037a75ac4dd9a4adab4aa00006b2 : SUCCESS in 5m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1d71dd3445c741ccb34fe72afe2bb97c : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4f715e72315d4f35b3f6a0ca384bdfc4 : FAILURE in 4m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7cfc993c6d28476db315891d46d99bd3 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d60b39c258324943a20973d9fca87c9c : SUCCESS in 8m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/74b6c40037d8415d97f4d05138e9bbf7 : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bedaa0e37e564b38869227f676de1a7a : SUCCESS in 13m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8cffc11dc67045b7b2a539bf87262529 : SUCCESS in 4m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4e129bdb40e143d4af198085f4fa1b4a : SUCCESS in 11m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/71c309e7160b4e38b49593ff452b38c2 : SUCCESS in 4m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2ffc664bb9c546839625afc399d90551 : SUCCESS in 11m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d732bf9c6d86442fb7e6f5bc9bb71daf : SUCCESS in 24m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5b659ba3af0e42b28843061d7da79fd0 : SUCCESS in 11m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2f9f0c7873754e198c9884efc9d120cf : SUCCESS in 14m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dd4142a3d5814538aa200d17fae8b595 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20ceb9d3da114ed19a1d1a7a379fd7b2 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c09af71109e94dca8fffc27dfcce410b : SUCCESS in 23m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e6aaca66bbb44f3ebbf07aeb3d380aff : SUCCESS in 4m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b5fc897f8c2b4206a42b130a6177d726 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05e1e54b3af84d0e8441e2e599b7b796 : SUCCESS in 6m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/fbc9ad08c59e4801b31477ed60f3cc51 : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e280cc2f8aa4254b2f1d88cdac90ea7 : SUCCESS in 54m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3486cff46ac746419de0dfbff3704f29 : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fb1e91583e5397e6d776791a5f48d1b8806456f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-01 07:45:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc7c418c05f6d32e9656aec349948a33326df08a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 09:29:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/328eb421d7b04f4db485378a061f8c04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871e30dba5ab45008b89adeff8f7eee5 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13abf59e36984ffa87a98315f2eff047 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/709e086e898d4b54916210cfd8d64b8f : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98a722231cf344ec8fdd73da030d99f4 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/367d43d2e92c48909d3cc1e6dc2526fe : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b803ba212ad6448c9da3c428b88d6e7c : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/674d35fb67434514839652975bef6f80 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/357dc2ff9a954adfa80df7d05429b11d : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8922a8501e08412e90048837004e37fb : SUCCESS in 11m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a844d7d594c04dc88596a9ecd1487b55 : SUCCESS in 13m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9ffa4db9de3448548b35ee4aaaef95d9 : SUCCESS in 10m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c22efb514aa541fc947402dac92c2f39 : SUCCESS in 14m 00s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/71fabe572b364e05865c6676b40eb351 : SUCCESS in 7m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c5db1ff84084444aae6889a063a29048 : SUCCESS in 5m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/cee168bc19bb46c299802fcff211663c : SUCCESS in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/82e33a4e2dd1418ca5875ea2b74210f8 : SUCCESS in 12m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/98dbe8acc915464a81db6dadccb5e1cd : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/53ff51b09a224950bf756675d541bb20 : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2fb9b928a89c4d6ca6d8a97d8d54eefe : SUCCESS in 8m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/58759dde2b214fb99762772d55c66340 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/edd4e8e48ec649909e9fa26002d075ab : SUCCESS in 6m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0c66d203914c4b30a1f0b091dbf84ce3 : SUCCESS in 7m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ee062b5911594a4ea4c2650aed4682d6 : SUCCESS in 12m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8ee16451bdc64a6cbd821256d8955c99 : SUCCESS in 3m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7906c43599ba4f0992b0c2fa27a978c6 : SUCCESS in 10m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2f5e366b540440849e22ea6cd43d0d79 : SUCCESS in 4m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/05e1ed6a00d548219ff280f8aa2c126b : SUCCESS in 8m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4651755ecfcc4c6289480bba7e1a4894 : SUCCESS in 28m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/328edd01e724410ebdb6822a9dd5d9ea : SUCCESS in 11m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6b8b8867b7584907befd3db6f80c8445 : SUCCESS in 12m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ab287ad549ad4736a9e83f9b977e9970 : SUCCESS in 4m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/888e885fa8514ef083939dc7e7362115 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec9969e3efab45998d539da2caf5f9c7 : SUCCESS in 22m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c6f2a552de57486aa9c78359d7ec5e6f : SUCCESS in 5m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1858023f87b14e288108422d525a9caa : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5eb2c2a808d4795a037258a8d50dd44 : SUCCESS in 6m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/665f40ad70cf452e880873636c239747 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b352aa6456fa423a9cf6aaf74c05f09e : SUCCESS in 1h 04m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46b6a1378ccd4feb906f5021607dc902 : SUCCESS in 1h 36m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9a7ad9915e823ac987a4533c260db4562d9786b8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 23:56:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d03c37f7a420e3d7954916c46226c6dd5462e39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 23:57:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54bcc3ef07116759ba9c8aac50b05af69aaf9193","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 01:09:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44efca8872334b1a9a631572fe7caa78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3c1f38ac7b4f42b913c61f95db65dd : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/172da76c2ccb4ca8825b31abb6d38878 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f76697f4b9f444849b88110afff7143c : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7af8773682c40e08887ff971757c874 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2478a714443a43eaae1aee077bd36108 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e795f964293b4283a049024b8b2e0d49 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d213a0c1da2149d49d4efcf1dfc13afe : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc79e005167e42bf82e22ca5ac484b4e : SUCCESS in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/699e35b6fa684754b89f5805ac1d46ee : SUCCESS in 10m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/67a8e289d873467a871d81d8b2cc6fe3 : SUCCESS in 9m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2afeedeafb0b4fa5a249936f30fa12e8 : SUCCESS in 11m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f35fc8c204ae4112824e236975ac8c49 : SUCCESS in 8m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dc47f08a96ff4260bfcab048fc62cbe2 : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4f53030fdd2b488f98177a528beae3a3 : SUCCESS in 6m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bd2cc5d579254f7595563685459802bb : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6d553e5ec734821b48e157ca6048f6d : SUCCESS in 6m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e9020cc43fad423a9605c3a1df6d93dd : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5fb2bc464c0246db863414e9b7b7ea3c : SUCCESS in 14m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/83be0508d35f44afb61ff87d1d7779f6 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f453c2a1b6044d9bb6590fadd455ba57 : SUCCESS in 9m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/dd182c056fe94faa98a3df06bbaef950 : SUCCESS in 4m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1dc82ee61acb4584b10f50159fd17a9b : SUCCESS in 25m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/34c9b22edd3448c5ae74c314031a5645 : SUCCESS in 12m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e4b214eeea7545f49746e585db856ce9 : SUCCESS in 12m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/406eef971755435580e8f74b70396e4e : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4509b9f50d2f4f88b51342f0cd88f9cf : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70c6ede1356b4bee8d34f4ba82b9abf0 : SUCCESS in 28m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/eab1bb8388504381acfc41fdb7a750ed : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3f1987d4ec304e97be172ca2bda7191d : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/afde54446d5f4cbf8a9edd7454025e6c : SUCCESS in 6m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9320781cb9ba45789486bcba8e49b999 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5b7fbf6fcde4f04b1ea6e9e4115879d : SUCCESS in 43m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf3f3477393448599becc5b11170abf5 : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":1},{"id":"80adbf4d04c864f1ba428c8f755b4161d553b49f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 01:09:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"36280459938743c4ba3d9c265da967c9c9ca7c7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 01:10:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8627486fdbe441178f0ff6429d6eddd3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c92e43abd04db88f34db471b8586e7 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f265e2cbb1e51cec297cab9ed0e346c2da24fce7","revisions":{"f265e2cbb1e51cec297cab9ed0e346c2da24fce7":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:16:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/893348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/893348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/893348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/893348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/893348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/893348/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:16:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:16:55.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 5.3.0","message":"update constraint for python-designateclient to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I271f3ecf1c60a0405289b72d3dff97c1813c0eaf\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic1e2a1e2de59146cb5876744a800789a8d188432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f265e2cbb1e51cec297cab9ed0e346c2da24fce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f265e2cbb1e51cec297cab9ed0e346c2da24fce7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
