)]}'
{"id":"openstack%2Frequirements~893350","triplet_id":"openstack%2Frequirements~master~I8475e7cb101f58dc3e5b9e51427a7649c27e269c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8475e7cb101f58dc3e5b9e51427a7649c27e269c","subject":"update constraint for python-zunclient to new release 4.7.0","status":"MERGED","created":"2023-08-31 10:17:56.000000000","updated":"2023-09-01 04:32:46.000000000","submitted":"2023-09-01 04:31:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893350-new-release","meta_rev_id":"2eece5566538f32e7f94d9ea26b5dba3b00a202c","_number":893350,"virtual_id_number":893350,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 04:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-01 01:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-01 01:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 11:52:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 20:27:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-01 01:52:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e10ad86459389edf3946ab3df6fb2a3b070af430","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:17:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31e016a5f88a9821beced85ed01d14f6d858287c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:52:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c218e31ae63946428fa7ddc4c1fb00f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88ee65b8f254f49bded00c871ed21c2 : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c700e05b06429e9221f80e693ff5c7 : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1184b62a3c7241edb193dad138b9f0c1 : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad397cbbdd804eae99e53403bdb76649 : FAILURE in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cd05aa7e2d74f3eabe45439db9f1cb6 : FAILURE in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d761ee2aa72b4b28b71588633d558ac1 : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ceed8c5167b24a0ea36a82cef10681b0 : FAILURE in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94b95d874a7e45e380b2429baf66e74a : FAILURE in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd74a05118f34c6f8675bc6d91af73b8 : FAILURE in 7m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2bcaaaab98a24a7cb1670bfd527553a8 : FAILURE in 13m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e17b7bef5bb4a0d94a3fb2f958b4cab : FAILURE in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f0f178ea837448b8bb611b382a0efb84 : FAILURE in 3m 53s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5435b147b2764f01a806cb4e6918659a : SUCCESS in 8m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e6e5c6298edd4e728665fb3125cf315b : SUCCESS in 5m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/75d3a7dc3d1447ba9468526d1ab3171c : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1c62f4a94ae54b37a3c8d3cf05266749 : FAILURE in 11m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b0863754177f45d99a42b2828e6c3d54 : SUCCESS in 6m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/80aaf15a9172489b9c047a3ce3bf3aae : SUCCESS in 6m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3058198e37544a61a7d7b83007cf64c1 : SUCCESS in 10m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c65f822759244973931e51716e533f4a : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/057e554ea2b8422e9c948bccbd336a3b : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2053d499c44d4626836ed3a7ae70710c : SUCCESS in 11m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/49445f6d0e8647b0ac98060ef1a1ad60 : SUCCESS in 11m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/63ae7c07841e4c7c8e6ca06bac58e026 : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d430c7e0b5cc4c9b8ef5217214641c3e : SUCCESS in 11m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cabb561ebf714473aac7a622ab620484 : SUCCESS in 4m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/450b154a259c4f8293a59dcd7ed6dd6a : SUCCESS in 10m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b056232cf711411484c459dffdbcd590 : SUCCESS in 25m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/925daffb20f746f7b18a27fdaacacbba : SUCCESS in 11m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c8916f53fa3348d6b0864775600f238f : SUCCESS in 12m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7129dc7458af47698740e835ea079de4 : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/004cedc6b07a47aaa6fe5b1cb4ec125e : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed5daeab58b44530af985c640af9bfe2 : SUCCESS in 21m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1f7c3d1904cb4a28b71f0218652b8c98 : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6c6e92945cec4f0aaa8df4053b356deb : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/218de44cdd04457789a87f69ae49560b : SUCCESS in 7m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f16a18960731427b92bff45db66a5664 : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ab4875ea30143c1925692f4f0a6d1b3 : SUCCESS in 1h 05m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/445bafbb21454ddc8ba4db84b74008e7 : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":1},{"id":"623c5b09ec0db20740c3aba8251dcb6bf373fa41","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:27:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46261317ba7ea353cc137c0b534a565e3b7d97c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 21:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f4f8d5a1d74e8db3c170a0d9e93a51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65d4b631e8341be86e1b54aa1755737 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4290405158ef46ee95020f4b987c56ff : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c8af40c7b0940dbb3a2b104c29c431b : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3116233f82b642d3b01eca687d8333b2 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa0de33dbee3484eae7f94b83344713e : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33e5232294af4d93988594f899967a66 : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7c38a23c3764adf98f1435b30697bc2 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/033d958447284f7c87602a22d69276f0 : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be1462bf490a4ff3962d1954b559f460 : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d2cc4c528a734434aa99bb1911a86540 : SUCCESS in 14m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/91af0f1f559b4b098bf62a67954a0666 : SUCCESS in 8m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a6e6b4a345604d2c9792aa48df1d4bb4 : SUCCESS in 12m 03s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/190d1577d4744c4b9fa88ab000d2a521 : SUCCESS in 8m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ee242da339b844138d61a527c142c362 : SUCCESS in 5m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/70a16d776c03480ca53bf7c58ceec559 : SUCCESS in 4m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/402a5a68f71547a2846244f8be6a63d6 : SUCCESS in 14m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b1ec01bd8f424548a95d7923229099c3 : SUCCESS in 5m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c05c9ef7fe8e48c8a008d07d6cc17019 : SUCCESS in 6m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/05040913a08e4a0fbbb628b5ea975d61 : SUCCESS in 8m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b1672de7043649b8b69a4d21ea09f75f : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ac7a6254a154a95b70ae950b31af05e : SUCCESS in 8m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/68e03ae89bd7426b8bac7927ce999014 : SUCCESS in 9m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c7458969490d4f058d10b754ea6cfd35 : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e899896749bd49b09f9f0cd44b09a658 : SUCCESS in 5m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/94f373341578490ebfee5d43b375a0da : SUCCESS in 9m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c0ffba7d82ca45f291654a91071c3ca8 : SUCCESS in 5m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d16a00516bcb456cbe55bbdef3ef1f95 : SUCCESS in 10m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/64d4d52127be463299aa41467f02248e : SUCCESS in 27m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/686101d88fc141cf8d48b3e34cc7cb76 : SUCCESS in 13m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c287ac543d724bd2b88210f1a5036edf : SUCCESS in 14m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cc4762dd11b54e66903f6d3cf413527b : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac91ee433f8d491182fe76fc9b6301cb : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae4cc5b464484bf3bdef7199be08601f : SUCCESS in 23m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b6372adf18644f30be596ef3f673110e : SUCCESS in 6m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/568b95a981eb4956b86f643f53d2b763 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a93e83e1aade40fba6084b18cf4ae759 : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b28695f40f004074b5b3cd59334b1a4c : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d13ffe6d11b84a67aa6b33cf5d736d33 : SUCCESS in 58m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31037b0e3da04dde9f1505107b3ea363 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d03c82073ede2b190e72d3f03bfb9bc1131836d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 01:52:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b08efe5a9bd30ce5f4f7964b4ccdeda1600b29bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:53:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a23ddc4cc3d0c4fc64580b3b0d158ca15e58539b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:31:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d34ee9712b4641189e78d793350303dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa386a5c1b6483ead7f6fef12770ab4 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/146406934a71424499d9ebd972e3b1cf : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54209341af1b4c60b9571c7264abf63c : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5df8b809538843e380bef60f44d853f9 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f39587273c14f64b29a37bdec6508f0 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3859fbf2486f4212b85c7902cc68074b : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/def1ba6876c447eaab11cf4e2fa05600 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3a07fc114c2b4ecf9fe44dd7aef1342f : SUCCESS in 10m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2bdc6d97c02464e85f1a43777e35935 : SUCCESS in 14m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/960724f7187f49fcaa7d0e7a5cac5aa6 : SUCCESS in 9m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d475033993424989b05c0c00cb8da4f9 : SUCCESS in 12m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8815b88cff9a473cb2b092b3bc5b2350 : SUCCESS in 5m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5e56e1ba74274435ae3a57ff079892b0 : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/48246e650219453e807c56343e20bd96 : SUCCESS in 11m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/953bcb8580f34b768eaf9814c3cf72f0 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee4857fde19a4ee99a12af69a2ec38a9 : SUCCESS in 8m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/159535ad19c145fb808bafa98fcff36c : SUCCESS in 8m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/938d1938e4e54a789fe9cd1224d42261 : SUCCESS in 14m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f9afb8c34f41444da6717575394879c4 : SUCCESS in 4m 55s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c03ef849be914579a38d3167ad59f1c6 : SUCCESS in 10m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/13b4f21c6b1e45d695ab581529477a25 : SUCCESS in 6m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/39133ae228474a418b2af23d5f128745 : SUCCESS in 26m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/88713965830b4717bc8834c10f9e367c : SUCCESS in 12m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7761a008dc53446c8713aa46bd6d0b39 : SUCCESS in 13m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0cad820102534e05861814e1ff78b92a : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b417a4fb6f084d31b9cf813d73965665 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/854ac8ec325f4c6cbecee48dbb96d780 : SUCCESS in 27m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/292a59169ccb4b248a24ee79b8109557 : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f2da3f8deeba4a8083a1f4a82bc31082 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1dd972a10ae74283974928364807cf04 : SUCCESS in 8m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a97a49a329f646179fc45f150b8f3049 : SUCCESS in 14m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0f9219f7c69476abf1d7a4967fae756 : SUCCESS in 1h 13m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2e4ced0a52d4df4a4b1df698143ae25 : SUCCESS in 1h 48m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f69b8df74982e0afde063ec906b0a504dd1bf9f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:31:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2eece5566538f32e7f94d9ea26b5dba3b00a202c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:32:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c91203e083649c689d66aae318005cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09f2c6c894f423d87a1e45bed2d99bb : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee72eb3526d47d05fda419fd296ac4b3460ba6c2","revisions":{"ee72eb3526d47d05fda419fd296ac4b3460ba6c2":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:17:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/893350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/893350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/893350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/893350/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:17:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:17:55.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 4.7.0","message":"update constraint for python-zunclient to new release 4.7.0\n\nmeta: version: 4.7.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ifae084610be836ffee2915aea121c3992526007d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: hongbin \u003chongbin034@gmail.com\u003e\nChange-Id: I8475e7cb101f58dc3e5b9e51427a7649c27e269c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee72eb3526d47d05fda419fd296ac4b3460ba6c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee72eb3526d47d05fda419fd296ac4b3460ba6c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
