)]}'
{"id":"openstack%2Frequirements~893351","triplet_id":"openstack%2Frequirements~master~I2d06fd16894a60e37e2e7a0d0283aad744e9d1fc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2d06fd16894a60e37e2e7a0d0283aad744e9d1fc","subject":"update constraint for openstacksdk to new release 1.5.0","status":"MERGED","created":"2023-08-31 10:18:54.000000000","updated":"2023-09-01 04:31:51.000000000","submitted":"2023-09-01 04:31:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893351-new-release","meta_rev_id":"bcacd96a06f484e0f632d0abd1074505749a0bde","_number":893351,"virtual_id_number":893351,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 04:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-01 01:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-01 01:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 11:59:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 20:26:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-01 01:52:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"de514eef9b8d3fddb11f6bed865059e460087d74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:18:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4100a31b12968bd0c20658784df4558030a90037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:59:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/455466b6150a41e193f854b87534a955\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dce6ee345e840ccaf43d8a93807197a : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c067a7b10fc437383b0414f40c2c050 : FAILURE in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fa40099af404438be60544395f07c3a : FAILURE in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/166ada0058c8426d8b204a5a2857f469 : FAILURE in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a343b1160ce848af9b4c2565ff24dac2 : FAILURE in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59f4f43b2f054e9ebb74fc2e952413e1 : RETRY_LIMIT in 1m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57ffe10f847e4184b3cf89a3cd1e020b : FAILURE in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34b37e72ac854e97897b6638f5657a44 : FAILURE in 4m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1e6a33418b1c4c34b845ea000528eeba : FAILURE in 10m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8208298e7298457a8e1121b897116669 : FAILURE in 12m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/45052ddca04046fea789ccbb4cb2481b : FAILURE in 8m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1a0643f9764b4c0984cc38e6bd5a86d4 : FAILURE in 11m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6508008e72f84382a4920df48a5a656e : SUCCESS in 7m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3cf700db0a704a279744fb45069d1ce4 : SUCCESS in 6m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/75e13eadfbb34e929f317f27fe7e815b : SUCCESS in 5m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1a5af2c9d6fd44be9e31fe18efa7b1e0 : SUCCESS in 11m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8aab4e4bd70142c2aff839cf73ed0f8c : SUCCESS in 6m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bd31a5800c6b4b71b8d8725e865a4d07 : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/85472c79030e46b28cad23a8a45faf94 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/56a47f0f473f4eb09c4341c351370ed6 : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/862368904a924f01be9e6098383b1817 : SUCCESS in 8m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/64ae615b4c6541a899ca82918379647d : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fa678cd99ce54c57b0b2e576ab182937 : SUCCESS in 11m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cd3b752b37b04339b7741c96a1040f49 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2d3a9fa76fe94dea9ca444d040b9e0a9 : SUCCESS in 10m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/280c98572f5a4820a48d3a827e19a70e : SUCCESS in 4m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e4d44326754448b5bf7be0a68093002e : SUCCESS in 10m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bdb24e5762c24934a1b316d9bfe9a728 : SUCCESS in 23m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8b9394d82242466c99f2bed8a791c5cf : SUCCESS in 11m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/de5f3897339343b6abb214fdb4cf3ef5 : SUCCESS in 15m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fc2e87e0ae22456682b93b452cc803cd : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ccda48c1bff447bb83eeafd11f03e0ef : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/806e084a4b084d10b2668a44ac71f0af : SUCCESS in 23m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2385f406e19242bfa13e87c029801d32 : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/822dff19d61343848f84fc1c4eb4bf38 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61b62c8fff0148b3a76078cef7c59862 : SUCCESS in 8m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/271e0d34b6b74c6abe42b0033bcb5f5d : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/798a292e384646eb93c40306b09397c2 : SUCCESS in 1h 06m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e6e6ee360d24e01b7f3bd2829c3f52b : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2bef54c8963adcd4b5159f1575c3df227b14bf06","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:26:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51bfdec0eff557c30b4912d21307cbe8f484c3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:17:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523bd978de214d03981fe83120ec4d1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c736c501d7ff48dba805ae33a5c257e5 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88357786728047e5816199fd12a07602 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1025b288f4614260ba466cbb59ac1156 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2938e359c9a1499e977b46d3352325c9 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3c24b1875304a819f12afd0cf62490b : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94cf19a47843416188bf906f0139bb91 : RETRY_LIMIT in 1m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e31db8198b34d9288ddededacc60786 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e99fa4d40f64e0abee3a42f1dbb2bb8 : SUCCESS in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/daa8d633866f445aa02722795246f1b6 : SUCCESS in 9m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/812ea6d81c58467798086df40e5c7396 : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c6bc17b7101448a2b96f00b68d65a466 : SUCCESS in 9m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d4432d80087444e6a7a03e1acd16425d : SUCCESS in 14m 01s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/72bc03d9e38a4e59aff499b408eb8a8b : SUCCESS in 8m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f80f97ff61104872a05b88bc21655a2d : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/70ac7906a3d4441ba60390244c961869 : SUCCESS in 7m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/404990153894485fac2d72467df2b9cd : SUCCESS in 12m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aee8366784494564a45a3b1d62d5f319 : SUCCESS in 6m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4b4bb5f9849649e089f3d661822fda73 : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2782480fdada45c08cec1002812bf14b : SUCCESS in 9m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ec90e81870bd4ddabff88b22b3d7ed45 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46b90101f84c484d821a96d9fa159038 : SUCCESS in 9m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6326b3ebe1984854ba598a4dd6dd9e43 : SUCCESS in 10m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fa6eee47b40f4c91b4db96c7be912519 : SUCCESS in 15m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c4d411ef8a50444fb1e5c7090d41b8f2 : SUCCESS in 4m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1ab407cc1bdd48238e98aa116fde0e66 : SUCCESS in 11m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d541be87a9a7498c8ba6b92c7158939a : SUCCESS in 4m 11s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d1bb2ecf82f944ceacf889f79a032978 : SUCCESS in 11m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2b3e402e805e484d9ec361881492f2c4 : SUCCESS in 28m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e47148c3fb364804ab17bf961f6435a7 : SUCCESS in 13m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cb53fb8fd9064251896a656b907abf73 : SUCCESS in 13m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f8717667e88d492e8c4cac8a96678d18 : SUCCESS in 6m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bacf93c286a54db890e2df2ef4c332d3 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0134e3d6ea2c4e32950d8082f2145be3 : SUCCESS in 27m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d4b16af0cec04702aa4b557792fdb656 : SUCCESS in 6m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/541b4c448dba4287af3bab7950c95059 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a31ece99c55a4f989f4b2c4fa1234bd4 : SUCCESS in 8m 29s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e35d51e9c2d8457f8e6b2c3f482da3c6 : SUCCESS in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38a663715d6742b4bef1d78fa3151c2b : SUCCESS in 56m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d786328bc0e845389c1a4c41fcc7462c : SUCCESS in 1h 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"69df40945953953e349eb8da6f7a9138efbdf0f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 01:52:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca0a94723a20441ee91c2ff9843f5264b8983c6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:53:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"103d12fe8c5a7e417a03565e43733f0453e30303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:31:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbed5a4b273e43e3b68b84810135b391\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56d39076c424edc91b36d19348174be : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97efafcaf95b4dba85681e55bc14af7f : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ec83c709e64c08aa7cecd3db6a8419 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e9524a89c324264b57dab483edb82ad : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c35b79abc31a4b7392a54787fa19f7c3 : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70013d45876549c39df833fa2ea9504c : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98de8dc8aa79462d80250dc2168d9a9c : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01ea5aa8cd5f413fa23d81409c6fa172 : SUCCESS in 10m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9dc8f4fe117643c885334a661075eb1f : SUCCESS in 15m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/97bc6bfc59944290b68e6a2b786a2aa6 : SUCCESS in 11m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/243a8fd23dfa4d27afe7637c83514c52 : SUCCESS in 12m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2153a2f1d5374114a66216f2e5dcbb8a : SUCCESS in 6m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/52b502aa5d9843bdb85387384dd2acdb : SUCCESS in 5m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/64b3672f8dde427683aef6265716a4c1 : SUCCESS in 8m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/73ea816b553847f18bfafbce13b263d7 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8283adb88ea42fa8370d1e5463dc8c2 : SUCCESS in 9m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5f80d129456643779b5d2e509a57b883 : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c61547ff178b4dad98c53987927c9fbd : SUCCESS in 14m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/29b178d90bda4f08a5e7d4ddc289c04b : SUCCESS in 5m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8f6a837e247f458696c876024cf1c8d3 : SUCCESS in 18m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/45bf96903ef243a2b35d029ce3a1ddd2 : SUCCESS in 5m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f0d32b3061e1451e8f38fe4846bdc8b1 : SUCCESS in 30m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/069911b48f1540d5814aa1680d28806d : SUCCESS in 16m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/02c487cc50dc421dbd6d1ea4bc943e40 : SUCCESS in 15m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b4715b356de34746a6c9a2d36e4da957 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32d8079bb270412fa4d815aec5b26968 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5375747f8fef4ac4a2efd3909238fb97 : SUCCESS in 25m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5bb541546630411280fbeeee7cfcd43c : SUCCESS in 6m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/409e1a16678d4d9e956c9c8aa4f7131f : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0fb8b3fb11d24349811d212e44981c8c : SUCCESS in 6m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/54787d25277540b888615f0d2b8b1c36 : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21018c551441493f868216d456707043 : SUCCESS in 1h 00m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4a97276bf92427c8487f42ed3e296f8 : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bcacd96a06f484e0f632d0abd1074505749a0bde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:31:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68d42f7512bc6a7b030a024a6f3b0d8f10f1f223","revisions":{"68d42f7512bc6a7b030a024a6f3b0d8f10f1f223":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:18:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/893351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/893351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/893351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/893351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/893351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/893351/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:18:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:18:54.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 1.5.0","message":"update constraint for openstacksdk to new release 1.5.0\n\nmeta: version: 1.5.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: Ib894e366e7916aa240b70c7462cf8783b2ab0cab\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nChange-Id: I2d06fd16894a60e37e2e7a0d0283aad744e9d1fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68d42f7512bc6a7b030a024a6f3b0d8f10f1f223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68d42f7512bc6a7b030a024a6f3b0d8f10f1f223"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
