)]}'
{"id":"openstack%2Frequirements~893352","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I7e1de832d4d1eec1d443bcdc6ec63511948970ca","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I7e1de832d4d1eec1d443bcdc6ec63511948970ca","subject":"update constraint for oslo.messaging to new release 14.2.2","status":"ABANDONED","created":"2023-08-31 10:20:47.000000000","updated":"2024-01-25 01:44:25.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"187b9d98a2423f3ac3b2f5abdbabf1d7823e69fc","_number":893352,"virtual_id_number":893352,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-31 11:51:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 10:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-31 11:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 14:36:48.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-11-09 17:19:49.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"fc87bc937f0e608733e4d990f2c9cc9a0ec7b40c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64a2381655b290863a5c4193aacb68cacfb880b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:37:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/439c607f79cd461e9dbd453643549abb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4e84104aba8b4fdb8ca5fe0ae49606ef : FAILURE in 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c2ee957e0018e7d068b34c750ae6bc20ec2e7baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:51:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ac435d7052a4f96955a5b7ec4ff7be7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141a22d1f8ae4cd1b54f5aacb1a67d9c : FAILURE in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/166a8e5026d540e6b1550b3507a52c00 : FAILURE in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd1576e2e9194f2e8ec0844b2e0f91fb : FAILURE in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/808cc7df16dd4c6386015a1a08e26dcc : FAILURE in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d1da1b39a9941eebefb285d33828e9c : FAILURE in 4m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e242a9774ec4f03994e6fec4236da2e : FAILURE in 7m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c086a6ce170542038f4d782c50bffa39 : FAILURE in 11m 25s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a9ec2955d2c14739b69f60a58d633544 : SUCCESS in 8m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/634b367df9814e2296bd0412ce0ada29 : SUCCESS in 5m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/55109f54a5fb47bc9248fe5df5419ad9 : SUCCESS in 6m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/023901e7a965478397a02c62a269ba8c : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/61eb7a4b84ea4116893e7b6c8e959ce1 : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8df423e7001a4d9e8e3705a301307fe4 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/413f21aa34ae4417a5aeb672d2707661 : FAILURE in 4m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6441368490dc4a63aea90a4913a4b80d : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cf4aa8ab9604e2cbe0c3558dfde11af : SUCCESS in 8m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d03eb13fc6184fa1b80f446bcc3c4bef : SUCCESS in 9m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2b67f2c1e20e4baf93bbd64dc77dce9a : SUCCESS in 14m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/99c28c06f9e74daa9de04163105b7d3c : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6d4277ba68c745f2b4d79409b116c06a : SUCCESS in 9m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1ebd6546dc5e4f7886961000f99688d1 : FAILURE in 4m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/43e443a6f6514c64b58691dde3579ca1 : SUCCESS in 9m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2ab8a12e2a7e41159bbb27bc15a9a785 : SUCCESS in 26m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b68a0b9d20324ecf81bbe75c7e6f071d : SUCCESS in 15m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ccd86feb130a4db4bcec7881ebb4d691 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/148dea896e704e329a47eeb35b59d4e3 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1ce06e5e4cb49839c1a6b5bdba22577 : SUCCESS in 24m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/317067a221d342e188d7c797f5240e20 : SUCCESS in 5m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1c55913cedcf4834b953d65fb7df639d : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fbdbdf4acbf4709b87822084542a17b : SUCCESS in 7m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/29105d6024714960ae3158f18126b1b4 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e18fd72bbba34344ba96ecb311c1e113 : SUCCESS in 54m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b15d9839fd3437f86f7364272651a9b : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7698f5c27c7f3e443a208e6e9db27ef2b63b1d94","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-01 14:36:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c972cb298e58f731f19b38e3fe4d1afcd30dad14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:43:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20f0a43bd3d43ad9ff59aa034f03fad\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1fad4d7225514ec984a788e508c7ab94 : FAILURE in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6a07ba42d8d94406960a65179ec3d6a4a4c19480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:06:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/967704dbf0624cacb9ebcdb8fcbc5b02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0c14afc7084f6e942783566c111ea1 : FAILURE in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3be2776c7cdd4b33a6d2f1f900d18b98 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cddf5ca33ec64957bddefb2a0424ad6e : FAILURE in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c11439a795e14270aa0197a4a7a5c353 : FAILURE in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7be590156f124c1f871eb304e70bd08e : FAILURE in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03227bb1c050461bb6b866e864564c9c : FAILURE in 8m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce59ddae15bb4cb6b3d9243280556f96 : FAILURE in 11m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1870acad26fe49d786f38cb2e900c299 : SUCCESS in 8m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a083a51248d24f14b5cb1174950ecf4c : SUCCESS in 6m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/911fea97f11f4b2fb13aa2a8ba027963 : SUCCESS in 7m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c16352dd49554660b2bf906a19217f8a : SUCCESS in 10m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f880cec85865488b8302e44d76201eb0 : SUCCESS in 6m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/83c38519e641400c9ff95b0d13f0f913 : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8ec78bc54743433197819110b4679469 : TIMED_OUT in 40m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3f735a5c99b44e5688ebd3cbf0ce00c3 : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/493f852de64b44b5a4c56e12c442720d : SUCCESS in 8m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4743d602505c405ebec55f7c6ea4493b : SUCCESS in 10m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bc378098991443528f0568ab2afc3572 : SUCCESS in 10m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/673ad8a7d0c44d92ae28ca2c0ec873b0 : SUCCESS in 4m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/780fd18100d1458998b397546828453e : SUCCESS in 9m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9311252ec9b2456e981e43452688fd71 : FAILURE in 5m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/afa9071594144c539ba24e6400612ac8 : SUCCESS in 11m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/85728e9497bb485ab2bdefd7e20e2fab : SUCCESS in 20m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/60a01b8617914ce7916752be19c253b4 : SUCCESS in 12m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/20b1d0212e144ef5b045fa01f41b4456 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7bb2088aae484d17ac22e9451970595c : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b8a13e0312b41a19dd94ff3f3768cab : SUCCESS in 24m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0ef84b0f21524d5dac113b0bb10cbdef : POST_FAILURE in 6m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0bb7f3fa3b1f402985eeb5b95e289585 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c808784e7e194af488574c8b89873cf4 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3cc0d9d4ff8f4b53965a20b3fd6caa0b : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b269157a82c540c8951b97a479fbe993 : SUCCESS in 1h 03m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b866777a72f429b87854b29433f34f6 : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c5df1eecc11bdd917838680ddd06b668416e256f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 17:19:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b23474696b4e8ffde3fb1090f9cde4441543bcee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:25:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6750a0c70364466d83bb4e53339d1460\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7959f1086cec48c4b1843275d3ed9d04 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5f19d74b6a59af1e37599b2602e2bfe1aa4d9d3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:32:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf4861f4094e436995b197cc4794e028\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363a2659c19c40738d77c52684de72ae : FAILURE in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18f24282fa5c461684084d788e00f92e : FAILURE in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e22d4c65e3fd4c2fac04ef84c29776d5 : FAILURE in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c5fb4e286f44036ad07e65201b52db5 : FAILURE in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85815ba5c99b41d69a4d4ce9a7824452 : FAILURE in 8m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dab876cff2554b0a8558c2ff2699d099 : FAILURE in 11m 48s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/be59942695d2410bb1dafc1b045f4fe7 : SUCCESS in 8m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d01d094dee2f43bf953f4c1d0af6d457 : SUCCESS in 6m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8f97d038b1214c75a2598876ed750c9e : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/56e53dd5bac548919349ec5e4d861cb8 : SUCCESS in 12m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a3247b83ecba4a6ebbce9d1ee76931fe : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/44c7f51dcd6048d995fabea5263fc21f : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/230a4655ff034baa9af1bf2df6bb2451 : FAILURE in 5m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e8ae58e9291b40158c443274c5c7b707 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f5cb27a44884bce95dd03f17589120d : SUCCESS in 8m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1a8473db1fe6409aa8ed48c242ad4a4a : SUCCESS in 11m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0a1d11e5255741f180aa65c53ac0264e : SUCCESS in 14m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5e833daa90d449c18fb474cb34a19edb : SUCCESS in 7m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e711a0b3569f46e699c5b6bc86a0be63 : SUCCESS in 11m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/48f2cf2649784a9fbc9ea9e40d540d2e : SUCCESS in 7m 02s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/deba16ebc5ff4024b814b4454edaea0b : SUCCESS in 12m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e794a7671f1343a39127be4040e10399 : SUCCESS in 25m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1a0eebc527a44b4b938a36f8310f9248 : SUCCESS in 12m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3e6141796afc44e8a55b8a331cac4953 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16d1fcf6ee444850ba8ec9d34ddf59a5 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fea02d71dc274ffc8210c0aafed5cf4f : SUCCESS in 26m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6279462a1eae4d448684e625ae471f72 : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c2aaf1c01a2441ed8cfb949c65a155d8 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf9d1b506c7b4a6f8b49b866247383a6 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/59c2d7404dd4454c9e161556d15a009b : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dafae13325f14176b14244eca14c00ff : SUCCESS in 46m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11ab160a07424019a3ffc342b228c77a : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6f3929dd35580e9d31d91cba8255c08b147b815a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 21:34:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"618bd38bd08a04e7e56c8d5272970ef9115440a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:40:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/932d7118453c42458a4a5950209ae1a5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ce0f2668ade493aab9a509913fc30ce : FAILURE in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"062f870e07f1b433a552bd7ce540d539c741c9aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:05:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58c4197ae461472da84f0762a18dee43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf0b9298141490b96668586bf8dbdb2 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c0fc51d244e4f1c83e5a7dd5716a629 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/107b300386f540ae8b6caa3310e2a6ac : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5464f28aada14bc9b3af71e99b235fbd : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9cf84c61e90445b78277d81eee8062fc : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e2b3b49954b4148a305627332ed6856 : SUCCESS in 13m 21s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b8a48121de38421dbd59e361d1fc2fe5 : SUCCESS in 7m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/04b0bb0049c743bcbbd0818b6865a096 : SUCCESS in 5m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0f9e748237a147d5800a796d28b425b0 : SUCCESS in 5m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/86b263c6fab74cdd991a2f9815f91924 : SUCCESS in 12m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5dc1050d7bdf424895ad3c37571a5a74 : SUCCESS in 6m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a15a7a40c1b04194b61bde7d045ec67f : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/299efaad9d6c4194a4fb8e9fae8c71ce : FAILURE in 5m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/290a010c126246e1ae906b6862d5f81a : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5dc0072c9c044a9a8f74cf595fde9f0e : SUCCESS in 8m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/23b8c84ce42b485b91aa9b403748936a : SUCCESS in 11m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e0c7b857f52243ae9678e9371458c834 : SUCCESS in 11m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cfd14fcb728f43d4af37913d97e7403b : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/06a256ee3e184fb3bc1676ad0d90189e : SUCCESS in 10m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/49c506cfaa954bd6858fb2b6ef48aafb : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ac3d2c6bc63f4669979a8c0fb4506449 : SUCCESS in 10m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8b014fee58074cb8aa9af40323177524 : SUCCESS in 22m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9d9d7678b8764ad3a740a202e81559df : SUCCESS in 11m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ca62f5f6a9c348c28912ccc1610080f0 : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0726d7ceae6844cd9e3cf0112152f3be : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/511d544de2ee4beeba56576b896a98ee : SUCCESS in 26m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/04acd7aaf34a4e3da5b3675e0403fa41 : SUCCESS in 6m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1a7acb4de59349ec9f6f3ee1487e353f : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/315f2f7e178c41b383322bfd21badb2c : SUCCESS in 6m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/354d13fad520415c87bc2761937bfce5 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a338ded837c44e32b74295f4c9a37c43 : SUCCESS in 51m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe64d680e6104b22b30657634143a4a0 : SUCCESS in 1h 23m 28s","accounts_in_message":[],"_revision_number":1},{"id":"187b9d98a2423f3ac3b2f5abdbabf1d7823e69fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-25 01:44:25.000000000","message":"Abandoned\n\nNo longer needed: https://review.opendev.org/c/openstack/requirements/+/901128","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fea1a5db31a291bb0b20a4ff0761060c418eb5a4","revisions":{"fea1a5db31a291bb0b20a4ff0761060c418eb5a4":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:20:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/893352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/893352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/893352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/893352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/893352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/893352/1"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:20:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:20:47.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.2.2","message":"update constraint for oslo.messaging to new release 14.2.2\n\nmeta: version: 14.2.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib93ad4296ea1b2daf671fa6e88fedd66ad990e07\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7e1de832d4d1eec1d443bcdc6ec63511948970ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fea1a5db31a291bb0b20a4ff0761060c418eb5a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fea1a5db31a291bb0b20a4ff0761060c418eb5a4"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
