)]}'
{"id":"openstack%2Frequirements~893353","triplet_id":"openstack%2Frequirements~master~I75fbed9895fa8b71ec33645a6e520413a96d5ac7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I75fbed9895fa8b71ec33645a6e520413a96d5ac7","subject":"update constraint for metalsmith to new release 2.0.1","status":"MERGED","created":"2023-08-31 10:21:26.000000000","updated":"2023-09-01 04:33:28.000000000","submitted":"2023-09-01 04:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893353-new-release","meta_rev_id":"0dfa622f48a5b46f08af3b4be5e3830f4d2c1982","_number":893353,"virtual_id_number":893353,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 04:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-01 01:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-01 01:52:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 12:02:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 20:26:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-01 01:52:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ee2e4a4cd449ba339c863207feb20ca0cf80acd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4e50da6f05c3f59c2a80a020f3673b1ebf378b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:02:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14c942d4a28446fa8778586a39e8b989\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20f6bd6de604d8c9e2ace7d732faeaf : FAILURE in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38715e76fe2949d2aa1179440be080b1 : FAILURE in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e12bdf78a1d4f8b980c6d71eff336eb : FAILURE in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86a1931c2a0f42c3bc3e81a9ba8930e5 : FAILURE in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/364d103f421e4e36a8754c6f1970a61f : FAILURE in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a77290a498a4f528aef0e755920cfd8 : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3749ac09a87643c19fe5f81e28dc3e95 : FAILURE in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/741de0e5dfcc44ed8e559060aabc0935 : FAILURE in 4m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c2a21483084f4c64a6194b1a4cc02492 : FAILURE in 7m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b78764a44cfc46caadb31c4614c332aa : FAILURE in 11m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1922adcdc5974cdb847d4357ff873627 : FAILURE in 9m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cb6593d78bf940d7961010b9de835ef8 : FAILURE in 13m 21s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ef8f0da1f2b24ad58e32fdf4352369f4 : SUCCESS in 6m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e82d6b8e96984c53ac9021bfe9612b58 : SUCCESS in 4m 59s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ae3b6fe28e3c4c7197ca283a0a073c53 : SUCCESS in 5m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0ec8665ecde94b71829f6c4f31ca04c4 : SUCCESS in 10m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2d7fc97cac8b43bd9ac4a08e5ad5dc11 : SUCCESS in 5m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f61ab3741f64493dbf38ca04be282330 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/52279fd24e5745c281b174ad4457c527 : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e13ab5a8426a4192b2c6a10aa3a156aa : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8df6202c8584e4e946e35884248e524 : SUCCESS in 6m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7b0f621a72d94033859ea9ac3461af52 : SUCCESS in 9m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1788741264a145ac9d5f9dae9472826b : SUCCESS in 14m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/36b363e4c33b44e5920f4f4d5925e962 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f259b771d24844d7836910605d491628 : SUCCESS in 9m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2b44e3e1f10a4682acda7f9dcc8d2a39 : SUCCESS in 5m 08s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bec2107e4c284bd7a3e9ae0f2805393f : SUCCESS in 10m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bfb67f5d252849dcbefea5fa0ee9d771 : SUCCESS in 26m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/263aaea2d7234bdeb88506181c896a8f : SUCCESS in 12m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0cc1f022b64f46ae9adc6ec366de4b84 : SUCCESS in 12m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/585ca27818f9403bb9db065e8907741c : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea7f38437429472db0d4331cf8ab3207 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2dc43c80a634247ac26883403246b01 : SUCCESS in 23m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cd30e25405ec43869c3c7a979c8fd894 : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f5939d9512a64af58f62a7ae53fef44a : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd6d5c61e7ac43009e3c1105fd103aed : SUCCESS in 7m 10s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/428ae270ca0e481eae74f6efd47d0045 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd705a8187464b0cac9bbdefe398726b : SUCCESS in 59m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/409f50b2ca984d1bb035f54df1da53dc : SUCCESS in 1h 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f7d53576c3f6258546b03d2d6d9e8c6784a88273","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:26:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cb6985f9705aeb029a7e27415278fe9c0fb5f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:10:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2e07e8b1ad45d3b5f5bbadb9b632d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f04247b3b04e58b0451d43a9c0bb06 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce7b6a910101498c8f6a58286bd0e802 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1da2c74f4085450fbff350163b61328c : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b61ae455dec4e97bde3bca6dba69c91 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/055f3ceac30b455d9a0beaee2fcbb18c : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c55370a46b244bb19a4c10e39c917aa7 : RETRY_LIMIT in 1m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/71721d518c0e41f8a14cf13cc275751c : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/643fdb0bf66d4021903fc9ee64b2e3b4 : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0bee86bb592f431ba26a2230550ec768 : SUCCESS in 10m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0913c55306004c89834789853db48d0a : SUCCESS in 16m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7af593b079e541eea40de75bb8ca1572 : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/edac585635d045618c1211856fa7d8b9 : SUCCESS in 13m 20s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1c473a977bd44072a0574a168736cf77 : SUCCESS in 8m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/328dd786dc1f4ae68fcedb8f96cd8f4a : SUCCESS in 4m 59s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f66130d5a0f3414b8a3b9e9202a7915b : SUCCESS in 5m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0ff45a443a394797b7d6ec8777d90caa : SUCCESS in 12m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fce64ec037c84e6182885ff7d8279c24 : SUCCESS in 5m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b59a361de57542aa97878324951bc8a6 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bc829134c2ab4b68a697462c7b0b9da4 : SUCCESS in 8m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/19ecbabd7b4d405bb0e720aae52d74ca : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcd7df9f10b54738b4208e607ea08846 : SUCCESS in 7m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/949c43aab02d4c7eae5631c719e83e03 : SUCCESS in 8m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f1ecb4f2b5b14d31a60323e3b1c0d0b0 : SUCCESS in 14m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b9a2173de6b34909a43e2fbc687b2dc5 : SUCCESS in 5m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fd8c697e76284a0ab560dacc1e47e7fb : SUCCESS in 10m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/39801d6990e64848942b6fb573eac446 : SUCCESS in 4m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8ce1fb0446df415bbe595960359d56a8 : SUCCESS in 12m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1b177937a80a47568cd0098d0cf41cbf : SUCCESS in 25m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0d94eca72acd4f899bd8e9c06d220a9a : SUCCESS in 11m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8cca275941ef437ba1b33c86157fd115 : SUCCESS in 14m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b8aa2eab632742b7831d3baf8f50301e : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f1aa641a4174348a056b28114a5f474 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/874d3f40cc524623afc9e8639717e7c0 : SUCCESS in 27m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/264daf506df442409398f5cbc8ed13ed : SUCCESS in 6m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/24c6909c8703404f91d7867c9f272f17 : SUCCESS in 6m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01a4d66147be43f9bebca3a3303bb910 : SUCCESS in 7m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/559a1bc7e86b44b5b46675a495ab262d : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08030c1e69424aa1ae162f637e33f848 : SUCCESS in 1h 16m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93b79bcb431d4db4b8b96415456bed54 : SUCCESS in 1h 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c9c54688c8d573a5f48a7187329f4da7aa0f821f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 01:52:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e01eb1479a25949908bd46bb71d0c07d57f738ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:53:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb5ff97e00b706c15d4bb871262716aad8cd9962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989d26e5c5414d79b469d209bb5a1fa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd38a662d474d81bea3628778a03b79 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf245ede9dc4d87925f1c0215dab3e3 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46a25c3abb2043cfa00b4130ed3241c2 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad1a0ef96dc3443e89cfd24039912a8f : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ee67d09b1034797b39552a85599ee81 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22233c4f8f4b4ac6b361739aa55d82a2 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/582beb1355134713b700d60e2e40e618 : SUCCESS in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b03cfd1839364321a60866766ed32fd9 : SUCCESS in 10m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c3e61e7fb0624aa3b33edce11aa0bb78 : SUCCESS in 14m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4a1f842a79314a87bac78775237280d4 : SUCCESS in 9m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/dde9bd53564d49ccbc848a3fa8f38334 : SUCCESS in 12m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/97a4d6b3206e4db083ea7b5c04669cac : SUCCESS in 5m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a7d9a9538f2047ff89cb8310959b92b4 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/476acf4aea0940e69b770a75b2393359 : SUCCESS in 8m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f7a8c39e120f4ae7a7879555621123dd : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65a85defabb44868857ea0a3785ab408 : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b01d72034e044385bc4729987b09d4a9 : SUCCESS in 10m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1094996459304481b28ab059f937818e : SUCCESS in 17m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5083a272be4f4cdf9b1e2755a6f2dfc9 : SUCCESS in 5m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6ef857aed3f24ff3a4931f23da76f141 : SUCCESS in 19m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4bf1a91b9c02436392f5098cb8e6c780 : SUCCESS in 5m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fbf9bc6bc935494491a69033579ca0c3 : SUCCESS in 26m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/33372053e1be4a62a3453939b738abe2 : SUCCESS in 11m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e1f08b3da3904179bc0971fffc864259 : SUCCESS in 15m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/76a95b220a7e430286021ac7a7754334 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afac1cf070094698a4f87de40dc13b3f : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3f42ad81ff042b081c677e8f01e4bcc : SUCCESS in 27m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5e5fb2517b5347d6897f80c5647621c1 : SUCCESS in 7m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6501f516927341bcad09d99dc3ce3c7e : SUCCESS in 8m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5fc4a30c83754badaa12ba36eedacadd : SUCCESS in 7m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d92467ff377c49b4b103ec8e954b9e73 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4cb8a3964dda4c098a418dfdd9a56f21 : SUCCESS in 1h 12m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ade3eb51238a483bac85139f96ba6ce3 : SUCCESS in 1h 51m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9ef596973d57c5797eb51479067ede731560e215","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:31:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0dfa622f48a5b46f08af3b4be5e3830f4d2c1982","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:33:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9aa760863b548fd97703eb96039bc69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b94ee01d8c4ad89cb6ca6bdd0029c8 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50ed40d772899e65e1d62c894565a8333699b02e","revisions":{"50ed40d772899e65e1d62c894565a8333699b02e":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:21:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/893353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/893353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/893353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/893353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/893353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/893353/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:21:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:21:22.000000000","tz":0},"subject":"update constraint for metalsmith to new release 2.0.1","message":"update constraint for metalsmith to new release 2.0.1\n\nmeta: version: 2.0.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I2f5264b77374a76047eab4657741ab2d7cd947f9\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I75fbed9895fa8b71ec33645a6e520413a96d5ac7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50ed40d772899e65e1d62c894565a8333699b02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50ed40d772899e65e1d62c894565a8333699b02e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
