)]}'
{"id":"openstack%2Frequirements~893355","triplet_id":"openstack%2Frequirements~stable%2Fzed~Ia83abd6e7c79bf252058b38d14f87fe3a1f49508","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-12 00:49:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia83abd6e7c79bf252058b38d14f87fe3a1f49508","subject":"update constraint for oslo.messaging to new release 14.0.2","status":"ABANDONED","created":"2023-08-31 10:22:33.000000000","updated":"2024-02-12 00:49:36.000000000","total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"61da483f25dfde7065030ff97060e5cc524661b8","_number":893355,"virtual_id_number":893355,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-31 11:45:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-01-25 09:42:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":-1,"date":"2024-01-25 10:06:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 10:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-31 11:45:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 14:36:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-11-09 16:54:47.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-01-25 09:42:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-25 10:06:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-25 10:06:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"d5e0e3c332e0ad8f7fe2967be31d32a6016e9013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87a1ab514118b6cb357d5a38b846b9f2de5d88a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:27:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28fe29ae9a2b41ba8d4bf48e6c319c98\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c32ec500df70454591dd8f1920a1b5d2 : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d919d01c86d71f86f531c97f7128bfff99ea799f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:45:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b97ed69732a54465bd4ba8975d6e0173\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b77ed0b591f4b098e5c5c37d5895351 : FAILURE in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd1f0fbb487404eb9d23f0a7ab4b0c5 : FAILURE in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/363ddfc5633b49ff9658f63442c1003d : FAILURE in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e51a07da0d6e4c7498550ffcc3e8cc53 : FAILURE in 3m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/652fa9db2e534db59cbd6b9755c99bdf : FAILURE in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8244fc07fea245bea9e9ed6f04eea837 : FAILURE in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44c06b3222ed4807b2480453320360aa : FAILURE in 11m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e707ebe5bc1a4d53a92eca3761c11797 : FAILURE in 11m 22s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/bffdffdf37e047ab91448a80789c932f : SUCCESS in 7m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2f2633d45ee44956964e3e3636a6a716 : SUCCESS in 5m 08s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/677851cd212848638bd3f4fdc1fcf4f0 : SUCCESS in 5m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d72e8fa247854f58b8c50624d43e931f : SUCCESS in 12m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/331a018ba75842228f956b9cb754cedb : SUCCESS in 5m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bf5354afb79341939697da03bcca2655 : SUCCESS in 4m 52s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/9f7b2a0867754287bb75a1b42fac749a : FAILURE in 5m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2fb937c3e3f34a6e8927991ebcd36a36 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2baa2d311c6e463882733dcfd809eb71 : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e5087249840a49b9b4a7c5cadea6bc42 : SUCCESS in 7m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/55045de2193f4b42bd029d8b211464d0 : SUCCESS in 11m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3f195a013fb343edb98694cda43222ab : SUCCESS in 4m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8d8d6dd8cb8d4653ba1cb93658af7aa6 : SUCCESS in 8m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d6f550e173d849e8a53c77184c5c5e89 : FAILURE in 4m 16s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/89f6c43b62bf4b2397dc40a5298b0314 : SUCCESS in 9m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/181a73f4aa84428fb1a8d244adf7b3fd : SUCCESS in 24m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f72f7f9d28b84370899fd330ee82eb74 : SUCCESS in 11m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1c6ed96a3ed449b4b059ecfa2dadcda7 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/db2f7ccf8faa44b08304cf648cff7db6 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd769f8181ed4f728ef4e65c624dbf0d : SUCCESS in 23m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/29943096451d4d28a758dca564e150fc : SUCCESS in 5m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/39e05327bd054fe393dcdcc126600920 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2bba79fbebfb43b59d962ca748850b90 : SUCCESS in 5m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b6a9d39a19224345b36cd0020c080902 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dd108d3945741e9a7a1aa6d54250a5a : SUCCESS in 1h 02m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb52fce5b7224b938c9ded94277f00dd : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1bb102f119cdf4ae166dc74a64d701790f9d7c39","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-01 14:36:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e854ee6b6d9a7367a2deb8a894eb8426a0674d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:42:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b880b43330242e8b841e20b745e8bf0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8e54179f92d046f8aa6fbff7550429e5 : FAILURE in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1cf8ccdabcc802087c57c1c54124805511c66136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 15:57:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/662d1bcd71ae4c13a22ab672196435d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a727aff0ec84453befe273212688437 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b67ced779464460f9079cfc36f837ba6 : FAILURE in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc0625eb70ba4797aa5d37c9907ffcb8 : FAILURE in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aee9422e12824d93b8f05222726a8d1f : FAILURE in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8a92c7415374034a684d182457a9d5b : FAILURE in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/623cd9754bcc433c9c3894b3f43e1d13 : FAILURE in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c89dc09a74874398ac1541dda1dddd1b : FAILURE in 10m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c3662eeedf9e4a79ace79a52a4b63c36 : FAILURE in 11m 51s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e9cd7e63b6284b3ab49d7e13edfbd634 : SUCCESS in 6m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6be20332165c409ea9fe650436945487 : SUCCESS in 6m 25s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/121c3f825f184a1088f8bf029dc2f027 : SUCCESS in 5m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eb53f911088d4533ae809a915f99059a : SUCCESS in 10m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8def70bdb952474db0d2af3280e77e18 : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8036173f3b8f495ea461f26d6161a0c5 : SUCCESS in 5m 43s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/00c193cbd20241ff9b0fac553c2b37cc : FAILURE in 5m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d46eb9f598634c7ab7a988d3060707bb : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b50c616fb3e435eb609a82fcd50478a : SUCCESS in 7m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d3eb5cded05544d1b12f6c247f25fcf6 : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/71c1308f58314ed5b4218c1203f79bdc : SUCCESS in 14m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/727af99c15754a8ba0da13f307932d74 : SUCCESS in 4m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/93e525d2e6374386bbdf4cb7341c9de3 : SUCCESS in 9m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0adc14fdde2945348568208f3c6c38b6 : FAILURE in 4m 23s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/f988843d65074b1b8421bbcddec18443 : SUCCESS in 9m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9f9bebba12464ddaacf16a6c4d7d98ba : SUCCESS in 22m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b89152e4b9f2421a919639567827ecda : SUCCESS in 11m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/84174adc66ea4a04a7015946aad691ba : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d87516a8d80d410193645da1f4f6fafb : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c3390e6f34c4ab5abb21393e8997ace : SUCCESS in 21m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d858c1ef61824d32a75847eaf5acd397 : SUCCESS in 6m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/56efa6c9707b47d6b38747705c0ebe5a : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cbe5dda9845445da9b8dbdca45a46293 : SUCCESS in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3c5f2bb71a2441d9b0bddaed6be83e60 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/606bc93475bb4468950c1e59e3942f9c : SUCCESS in 1h 02m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3dd9f44420604845af0e41e6ea854143 : SUCCESS in 1h 15m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6aa660bc5711377f0d5f9b370c54d8b8738488d4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 16:54:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ae93e2984a67bb0b3152368ce687eba526c054f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:00:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab610c0f559343149201ef5125879418\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/24e71bcda7b94674a069e0c607d8d6a8 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9822fc44537daba6f0c5f354d7acf4ef05f6b56f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:28:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4d3a4539e974d81b8f878d3be211baf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ddf203242ee467ba212557b25dc8eb6 : FAILURE in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ec1f7ee461b40c28d1f586200935f64 : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28840ac4a41a445a9b30eb6f1bfb3805 : FAILURE in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/457dea61ea184dfcb38bdb94899a396f : FAILURE in 2m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1602df71740405282305f878651ce88 : FAILURE in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d62a8f6901b4c8db95761c5ab230ba9 : FAILURE in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/41ea22730b974c5892200e8c011c0fd7 : FAILURE in 6m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9f87ac33d0224ce0bf7e85bece108683 : FAILURE in 13m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f902bbd97f4844c2b6fb1702e377fefd : SUCCESS in 7m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bdf1c706c27940479cca24858ac5773b : SUCCESS in 5m 50s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c213528aae3d49ad8af1d93e38c40ba8 : SUCCESS in 5m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/59ff5f874c8949a6821cae9e8dfd1e2e : SUCCESS in 10m 27s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b045dd2bbee843c08b82538ae6b12a4d : SUCCESS in 4m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d52c90ac2adb42fa92ec133f3965fd0f : SUCCESS in 5m 32s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e24f82f042c44edd8802e54ab0293c62 : FAILURE in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/08671273da7c4e16aaec0309702fc445 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d401ab3b34b74b1a81c0f3fc6acc7c25 : SUCCESS in 6m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/82037948ea094d9ca2cdc6a9f3203853 : SUCCESS in 7m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/44561e64adce4ff99153d57c43154185 : SUCCESS in 11m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7c63462261b044ce9bec7691fb7f148b : SUCCESS in 3m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e6f9c1164aa74c6ead3e4f8b736a4322 : SUCCESS in 9m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f1329c659bd440ebef08558affbd6bc : SUCCESS in 5m 09s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/13b9238a823f4fc0bc98ec9a8e7d86f4 : SUCCESS in 8m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a6b17b04f88f4c0e8f5d8af1b7c107d8 : SUCCESS in 24m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/310ead31eb0f4264b1bb0daebe95cb1d : SUCCESS in 10m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/733a0172243f42ad89b2d0be3dbcb387 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e01d858e15944f32b012f9a2e97e7c62 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e4e0043ee39d4befa7d62f5f3d24a15d : SUCCESS in 20m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5173a3a1e07a4207a1b827e6138e62da : SUCCESS in 6m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/463854d89ce848cca0ee887652cfc619 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c96eefafc38b4ffb84b8a5ddcd44b6d6 : SUCCESS in 5m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8b66425dec8340d99cade385ca22f7fe : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/840ffa3098e34632848ff91d66c083f7 : SUCCESS in 51m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06ecf95db2f94a5bbaf61c58552d7bc4 : SUCCESS in 1h 25m 25s","accounts_in_message":[],"_revision_number":1},{"id":"14be05b3628a7728fd4346ad3c63d47f77d545ee","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 10:03:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f0d0e135f187ac204463fd74254285cc6a6ec19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 10:09:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90dd4a5b637142828d9fc2a2520d6a46\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bc9db40fd55a463ebb9c518ee1c0c81c : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"baa6a28dba2347a12c97030002ff6043936e51a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 11:26:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83cffdddb23f499cba31df25a6042811\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c7edb612904f6aa6bb1631b261dbb9 : FAILURE in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07beef4fbff740378e31080f15051c83 : FAILURE in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/533d85ee792a4aee9b46be883b34d049 : FAILURE in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0593afd979e34c1886259b264399a668 : FAILURE in 3m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56238d3cf10d4f10a1a09346b78e4a3a : FAILURE in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa6c510c6bde4ebdb78161767b8e1ca0 : FAILURE in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8e9f0c2fa4024034ba2f0b7878f8bdd4 : FAILURE in 9m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e5269493baa148cabf298f3091427b25 : FAILURE in 12m 33s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7bc47df6742845ac9e9c94d2b8bdd583 : SUCCESS in 7m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e656c3b3a1684b9f8350cf399e058580 : SUCCESS in 4m 57s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/21a53f8c747a4e10aca4f2d8b1e7d9db : SUCCESS in 5m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ccce90e8390f416897293f417adf756f : SUCCESS in 10m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/36b09f12338e42cab586672145ba2bb4 : SUCCESS in 4m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b03d3334a137452780bc4ec599302f08 : SUCCESS in 5m 04s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8dd727c2365846a1a02d0935a68abcee : FAILURE in 5m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ef23b93978c34be098a2b85153386382 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/854f967c400642ecb9cf575db02e5ca4 : SUCCESS in 7m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7fc672d5686b4478b35a3b3843c42b2d : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c33251cf067c46abb025f01f3e99f529 : SUCCESS in 11m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/78b9c77e61ee416da4d9597c62501ee6 : SUCCESS in 3m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/782e4e7805ce426bb58eb8e687ea1c7a : SUCCESS in 9m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6634e56dce76416dbed81a6ec162a9ef : SUCCESS in 4m 15s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/d27b89b99c4c47deb1982c234695b4c1 : SUCCESS in 10m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ed915b2bd8964a668343495fbf76def9 : SUCCESS in 22m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20d99859ca6c451390d9347d8e2901f8 : SUCCESS in 13m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e761d007315746b1a76cbd98dba17662 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/00229768870f4528b8809897b5c0d8d7 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86dfdccecdf94cf3ab9fe1a1cac15185 : SUCCESS in 23m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a5663ecbe38142a3b50a03ea88a3ede3 : SUCCESS in 5m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d0fc9c73067140cb98c20b726bb54382 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be1a6ff2f59b4aa29554dcf564a868cd : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e0c244fc01444537863b4fe7a972312e : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b670731c08fb47649b7999fbfcc9dfd5 : SUCCESS in 45m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1190d8e3e60240cf9405f129d7e4dc7d : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":1},{"id":"60812302965170ba1ccdbd01563a36a2f41f062c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 21:31:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10e299bec6bcfcfed2e45863ca4a71c265a527f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:37:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c3355a81b6a46bbb937b9d0d2ed3805\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f03ac9e5b13d42ae850b1f99d64640e7 : FAILURE in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"54e696228f9f97b056e9c661d007526e43d4ff2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:41:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cff1c1b24b34558b15946d2a5a52aba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de253fa290c4db79706346a0becf04c : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f54afcb39bed455eaf021f2c25b41000 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39485919ead84efd9b3581aa6343eba2 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1292c143c39c419e919cc816e33042bf : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77dc43e13d6b4689a70d73dc947b9978 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1acbb5430f85456bbb3ecf40815a68f7 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d24e39fa37174f088b927789cc38175d : SUCCESS in 15m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/596589eef0cc4eaab5f9956694c8f123 : SUCCESS in 13m 46s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/b176d3e17a424660a83b17bf18cd018a : SUCCESS in 8m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a3b68297e90e434f9c962bdd2035e558 : SUCCESS in 5m 46s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a78d601db4ca4c2d9e9d19da589abbd5 : SUCCESS in 7m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fc251849cc7545ef9f019922f65ded5d : SUCCESS in 12m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/604d7a1ceaec4d188e2ec49395286699 : SUCCESS in 5m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0d268745564542b28ad5577167923cdd : SUCCESS in 5m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/31d5a3ad3bc24ba1aeb202fc65708591 : FAILURE in 5m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/97b9fe0cc37047afbc0734348264bf8d : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65d63a387ec940e398637cce3182cdb9 : SUCCESS in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/afbcd08a9f2c434db8d418f3ff1ee49d : SUCCESS in 8m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/596c156a6b9942699a40e506a66c560d : SUCCESS in 12m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/15bffa27271241bba49df8afaddaf012 : SUCCESS in 4m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c480043b3f9e48f48aaf59b5fffd3908 : SUCCESS in 10m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/99f800ba8a374b1cb047895f8427851b : SUCCESS in 4m 41s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/532ce81815e447109eb7f4346ffa8a41 : SUCCESS in 10m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f068cf132d2c47f2ad12d25bfd8c411d : SUCCESS in 22m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e1570748266e487d9ab37ca2d9a631e7 : SUCCESS in 13m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2bd634d323d94258bfe8c2904fdd53d5 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5de945073874acf8b238ed74f0c4973 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5c43725090d46dc8fbe299ae3450087 : SUCCESS in 21m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/62dedb36bca54c21b80161893f342f0a : SUCCESS in 5m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5475ceb54c444c4ab99ecc992371c131 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cf5f7ff9d1442fdb2f890f5ccc2baaa : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/61dffd44bdb1429ea3d15270b606b42b : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2edaa0c7c9d4362a8f367ae4e1b7f75 : SUCCESS in 51m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e3c29f56e154866b2b7c3fa045ef882 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e657a7e2088db6eb0c0d9aac7fddea8d58a82f88","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 23:53:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d138dccdb7f6b8201ecfdb8a579cabb2391fbddf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:59:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01679dd8c57a40699d842626663c5724\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/79a72f761db340e680b9e55f466f8363 : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ea92598c5e0e42566d719b7b34dc461532d34ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:44:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7930dd50d58450397e63a8f70c93ada\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c06cdd6dee4905a1f3a9accdae7b1a : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f77004c53d5849fab145e4fb69e3ac98 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/358f44e381054bf0ac854266bbb869fb : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b566122a4dc4a4da544beebbc150154 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9648782a0f142ebbec8761771611c69 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e3d13a783444e29b5ea81ee2b95e38c : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d2953324ea964ba6b14a6638dae18387 : SUCCESS in 10m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/01c81de879ec4289a19bb864b5fccd48 : FAILURE in 6m 24s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/50e6cc8a64394a509646b319923b0b7c : SUCCESS in 6m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/beda0a8a18384b9980c762a3edc125f3 : SUCCESS in 5m 50s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/191efffa02f2443e8eea26cf59ecc1ae : SUCCESS in 4m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/48ce39bac2894cd28fd1e2d9f04dc518 : SUCCESS in 10m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/29246cb8c79045b48b6983bca370eb62 : SUCCESS in 4m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6cfebc15ab73411280745e9f48e85933 : SUCCESS in 6m 04s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/527986e069e545cdb58d685e3ae31ed1 : TIMED_OUT in 40m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8bf95e7f837742bab29f8ad84e8a3815 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d3c2a45c5ae7466694d7b11cc3cd14cc : SUCCESS in 7m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/42fa37fbc389420891f2610166f088b7 : SUCCESS in 9m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/50ca442c518c47faa90aa9daf2e91cb3 : SUCCESS in 11m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3c0da1ca929c4ec8945408d182c25702 : SUCCESS in 3m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a83ca34dcfec40c3bc6d5ced7bc18a6a : SUCCESS in 7m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f190f31e7ca42de91883004f75cf599 : SUCCESS in 4m 29s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/cb003601f43b49b48916f1ca776e151f : SUCCESS in 9m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/66df5eef49cb4e81a2f24cdfc5b5d924 : SUCCESS in 24m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/839a5fcba9784c958c3ad3fa314f649f : SUCCESS in 12m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d27db9bd14994e0c9c19b020c761d7d5 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9132d1eecc3410ab91a587451aa23e2 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/098edc3774a54a8bad66e546fb798bad : SUCCESS in 25m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bd036919e32f479980ec6959a1e83605 : SUCCESS in 5m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7fc75b6c397c44f1a1f3cec2b3de5c66 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c4a1e14940744d4c937bab524d39a44c : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a8f0a75564af438a97cabcd12916af2d : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a8b45b2478744e69bc5e6ac462af208 : SUCCESS in 42m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8acb7eff61b6438f93967666f1347233 : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1e264b86675c55d697ca603f1520f6176c6fcfe6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-25 09:42:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d989eb199c3b91992e6e265fa8fd7d20dbd80a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 09:57:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0baf902f08844efa92793c4c44e99568\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c2164d0214b24814acae722c435bf228 : FAILURE in 9m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3d13e04c5380797d02e3868f14d092b15b7bdf3f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-25 10:06:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"268f9567b0656e3c94ca91a2e72390b582b944bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 11:14:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cd0b375d1b94a72b9738917b20e73f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa14710cb5c498daaa85e1c5cf5ac60 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f41d504602264e8d853dcaec42905484 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f673ce703e13483498671d1882b1bf91 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a24985db3344b96b04cc14a37483a57 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/348e718a8a6a447190587b64a0a0b227 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33b892a39d4b49aca3d93f66072a040a : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f86a460fa6a747c3a1976189c153074c : SUCCESS in 10m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/111a25f8d8af4ba2ba9564f7a2ec4f76 : SUCCESS in 14m 42s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/fe60ed6fa3cb46829bb5b8a34528276e : SUCCESS in 6m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/22a13aad30664ea790da9e8d41461366 : SUCCESS in 4m 54s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/0987f5ae2c774ac6944286a570238234 : SUCCESS in 4m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5679c7710869465b8916c48415063d1e : SUCCESS in 9m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/60dd31a187444bb485f030e7c1baaf13 : SUCCESS in 4m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bdb47eca07c6489ab525ad6a4f38598f : SUCCESS in 5m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/cd0a3fcd8bf54b98b4c1425a1d80b22e : FAILURE in 5m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d32cacabafef491abd74364654d4048b : SUCCESS in 13m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3caa3069b094492ea4200a238e4edfa3 : SUCCESS in 6m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e802fc03b2e74178a11cdce2f1945b45 : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/29cd87cfb6ff4b1b82bbc9483adc1b88 : SUCCESS in 11m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dc6b8e809e92442ba4a992473590c088 : SUCCESS in 3m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/93b2f06a9e964696b5df08d8ee070142 : SUCCESS in 11m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1687b6b5d6644e06843abbf3c2bb43d7 : SUCCESS in 4m 18s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/0c0bca63849942feb553ae280b7d653a : SUCCESS in 9m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/21076365ffa64615b32fec0b9a4c85c3 : SUCCESS in 22m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f74806ef0cd24b9dbee5896b3e72906f : SUCCESS in 12m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1ee94de573bb4cd4a4e7805f0dcaa82c : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5a6f4d575c841ed9a49a1f5733a87ca : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ea10332838640acb21c638cf816a15b : SUCCESS in 26m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/97b2b6db56224e76bcce3f62f9a2f1c8 : SUCCESS in 5m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/412ac64d3e7040de96db562ca5cc952e : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9820d5e4756842cea750078aa2f6fe05 : SUCCESS in 9m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/579161a8474743a892f166f67113ada3 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d94504221aa54805808c16726b819d43 : SUCCESS in 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fd6269fe07f471d90d2c32e2180ea28 : SUCCESS in 1h 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"61939560b0ca0a9d748862cd57c8fc8bd065ba5e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-30 06:32:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61da483f25dfde7065030ff97060e5cc524661b8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 00:49:36.000000000","message":"Abandoned\n\nNo longer needed: https://review.opendev.org/c/openstack/requirements/+/908439","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aacabfa25694a85fb7e9a6736365f34eac2e7717","revisions":{"aacabfa25694a85fb7e9a6736365f34eac2e7717":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:22:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/893355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/893355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/893355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/893355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/893355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/893355/1"}}},"commit":{"parents":[{"commit":"ce23ac59f84228b1747c22311992169cf827cff9","subject":"update constraint for oslo.messaging to new release 14.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce23ac59f84228b1747c22311992169cf827cff9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:22:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:22:31.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.0.2","message":"update constraint for oslo.messaging to new release 14.0.2\n\nmeta: version: 14.0.2\nmeta: diff-start: -\nmeta: series: zed\nmeta: branch: stable/zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I5942f57805269764db4f3fcc796bedcd8ef42376\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia83abd6e7c79bf252058b38d14f87fe3a1f49508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aacabfa25694a85fb7e9a6736365f34eac2e7717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aacabfa25694a85fb7e9a6736365f34eac2e7717"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
