)]}'
{"id":"openstack%2Frequirements~893356","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ib04c299c36ce6f5d7463e5b13d8e2433e4aa082d","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"Ib04c299c36ce6f5d7463e5b13d8e2433e4aa082d","subject":"update constraint for metalsmith to new release 1.6.4","status":"MERGED","created":"2023-08-31 10:23:01.000000000","updated":"2023-09-01 03:26:54.000000000","submitted":"2023-09-01 03:26:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893356-new-release","meta_rev_id":"18373c4828a3a9daf63ff937795ba0d5f06567ac","_number":893356,"virtual_id_number":893356,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 03:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-01 01:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-31 15:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-01 01:52:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 10:30:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-31 12:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 15:46:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-01 01:52:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d3b48e4d295fe046a830d32c22aa33d0452c82d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 10:23:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f33a74c40eb0becbf9d3110dd798ddc8992f9e6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:30:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70725183d61543dab0f27bc2d0ba393a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bd8a1b14ab224d11ad7d084a122b1476 : FAILURE in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c4e696e756b1e43230f5308d88523091f9014623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:00:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5073eda6173471cb1a1207fb2f46ba1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e55780377d3c4a6190fa916239728826 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/176d7ddcb36d441b919fb94d569d2d3e : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf0386f4db4f4eb3ba7c6661e65d05f4 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03c2fa7d19fe41ba812b4578a3bc892d : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e76bd2772534d99a6d0b354f16f17e8 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c47c7d62db34d359d7eed60084411fa : SUCCESS in 12m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/66c63589b73d405286d8bf3cfceef866 : SUCCESS in 11m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4febe41c1c794cb495eab39ab46f6697 : SUCCESS in 5m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/38f57c42d3534078bc18c5b34b8faa4a : SUCCESS in 11m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/040786a3cf154f80b343f7b794ef99cb : SUCCESS in 5m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1852a14514524d8395873af98822c856 : SUCCESS in 5m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b27b653ea7db4baebdc54f1de93b4355 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc21ca36a088402cab5574ba44a5da44 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/01a707565843436da095b7a3f3d0e973 : SUCCESS in 7m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/10252c3b5a49400ea3cdb233a86670fc : SUCCESS in 13m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c713c3464fae4e5da2ef81c8ac50bd1b : SUCCESS in 3m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2dd2584db3414d21ae554e5cd8c7c558 : SUCCESS in 9m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5844c01e9bbd43fe93a52b56e53dd590 : SUCCESS in 5m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/402a3b3e896b46dfbb977e332a1f244f : SUCCESS in 27m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/df1069720fe44265b56044135f41e1fc : SUCCESS in 11m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4405160598af4332a9f1ef527828b535 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e25a6b329e84736b0b47862a64b0d81 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a57346755a834bb6b9d0731274f003d2 : SUCCESS in 20m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e91e5ed38e25494d8750d2a240122c42 : SUCCESS in 5m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/67d28013dcdb4bb0a89aeee7acf870ae : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acd998183a0a48a2a84d3956f92e6037 : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aad2bbc9dc3b4873a109ff92941dbe6a : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcf36f1ab4864267bc94dd5b89fe8219 : SUCCESS in 1h 04m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d6956afb43e41b2b95f20b79aef49d5 : SUCCESS in 1h 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9e22da484151215512bc36ef525fca07d8fc56a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-31 15:46:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"006ed76e540dcfed80748f719579fe3ed6d1c20e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 01:52:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79a00bf37e14c2a4cc4d0116e563028dbc8f633e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:53:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ea431282a6c0da77c590d9e73fb5a5e471e9b5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 03:26:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4b812f942643579be524ce857c07a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce343d47aa44ac3a7c17b936dd0e1bc : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe105a36c624f81910e46869cf79dc0 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ed72df5df384f1caff71bd378dd1ab6 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1829409dee7347e3875d04ea85049ef6 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aeeebe9168e54d20a6069f336b5935ad : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5cb9f62851c84dae823a279a8b014e27 : SUCCESS in 11m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/086583bd4ece4004af5fa1d7b90815bb : SUCCESS in 8m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f999c927e2054f9db7a85cacd9120304 : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a6554d919cd84aff81ed2735c0851562 : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/21e1cfff21a54a89bd4d4b99a5dc79f5 : SUCCESS in 5m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/06bf2b2356bd4ea4aa60292dd3ef7be3 : SUCCESS in 8m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9cd8dc800a7489a93bf6ff6c4b1227e : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e02efb1ce19d48298f1f2702d4a47012 : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/14b7bec27d5d470bb28869c4d36a56be : SUCCESS in 16m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8c53d53e5a0b44d6b957b5f3d1971530 : SUCCESS in 3m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/41b871f0099147b2894f0a94ed12f732 : SUCCESS in 9m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d3163fd571864d6ab908010687ebecac : SUCCESS in 5m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/513ad96c285d4cd697dd2358259ae57d : SUCCESS in 21m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/89215c251691425a8b2636912882db35 : SUCCESS in 12m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/20e191a72cdb408d8792fa23a1fb7acd : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/716ed8d921834ab490ca0962624a3e3c : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbe0d0ac8dbc4153a11a956b07d615e5 : SUCCESS in 22m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cd62af22933e42909d78bab63d694394 : SUCCESS in 5m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/99515fcf8fac4c48b668d87088cef3c1 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15cefbdb28024e3383b9d617a41fef22 : SUCCESS in 6m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a9306cea97724326a2548bb37edc12a1 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db19c1a754b549539e61cc8b4710ca4d : SUCCESS in 1h 17m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50ef4bc42c1f4fe0bce7ce3dfef53571 : SUCCESS in 1h 28m 19s","accounts_in_message":[],"_revision_number":1},{"id":"18373c4828a3a9daf63ff937795ba0d5f06567ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 03:26:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b31a68d71f4a04af9434749be881eff737d3769c","revisions":{"b31a68d71f4a04af9434749be881eff737d3769c":{"kind":"REWORK","_number":1,"created":"2023-08-31 10:23:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/893356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/893356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/893356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/893356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/893356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/893356/1"}}},"commit":{"parents":[{"commit":"9076d9f06d2b58ef3bd259d9b5214e38047b0c42","subject":"update constraint for python-solumclient to new release 3.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9076d9f06d2b58ef3bd259d9b5214e38047b0c42"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:23:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 10:23:00.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.6.4","message":"update constraint for metalsmith to new release 1.6.4\n\nmeta: version: 1.6.4\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I3582d14073008537d22125e8549eba73c41a968b\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nChange-Id: Ib04c299c36ce6f5d7463e5b13d8e2433e4aa082d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b31a68d71f4a04af9434749be881eff737d3769c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b31a68d71f4a04af9434749be881eff737d3769c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
