)]}'
{"id":"openstack%2Frequirements~893372","triplet_id":"openstack%2Frequirements~master~I2badf0740fe8ca2575835912202b523649ec91fe","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-07 18:43:15.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-09-07 18:43:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2badf0740fe8ca2575835912202b523649ec91fe","subject":"update constraint for osc-placement to new release 4.2.0","status":"MERGED","created":"2023-08-31 13:50:12.000000000","updated":"2023-09-07 18:43:15.000000000","submitted":"2023-09-07 18:43:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893372-new-release","meta_rev_id":"8a5cedcbd90816f07932c70b867119bbfeb03333","_number":893372,"virtual_id_number":893372,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 18:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-07 15:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-07 15:39:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 15:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 15:46:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-08-31 20:25:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-01 01:52:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-01 07:44:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-01 07:45:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ccc200683807d77abbd8bbbc29a4468ed22c8488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-31 13:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaa934f2c9e9b771f9c6eedd5c9e010c859bfdc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 15:07:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cdb0b6d4aa8478cb5c9443123b0352a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb8ad26cb57d4ffd8f9ad9b666bab39a : FAILURE in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d280a8903cbd43f2bc466dedce6ab4a6 : FAILURE in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bcd3736f1ff480aaaac542950df388a : FAILURE in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ab70f4713f34da3b89077b82703af0f : FAILURE in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/714e565c42fb48c788a27b0ac818588d : FAILURE in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c232b91c347c48998ca33fb18207f55f : RETRY_LIMIT in 1m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66930612ee1c4ebca77c163bcb5af2a9 : FAILURE in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a300855256bb4f4489863aa218c78e04 : FAILURE in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/92ded066b1a8472a86b6be56982a7259 : FAILURE in 7m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ebdd844e87b74009aebd22c59fffa35a : FAILURE in 12m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/729b17874e75425dadabecd88b00d74c : FAILURE in 7m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/072dafedd10c46aaae8e46ec53dd1dbd : FAILURE in 11m 32s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/45d6297852a4493abf1b23e6a3d8d797 : SUCCESS in 6m 17s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a96d8aab759242b88979a65a9047d29e : SUCCESS in 4m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1a9574ac368543aa909c3485bdc27c6b : SUCCESS in 5m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/34185111dc1341728d2c666ba7ce0726 : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2087fb202a254ffe9d0ffcb3e2c052de : SUCCESS in 5m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8a76fc9f9126481aa86377d411b71299 : SUCCESS in 4m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/58be76137e864647a943389b6f9fff55 : SUCCESS in 8m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e5acb7c0999a402c99871e331ccaa985 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d72547c9c7b74d2483d89e827451f391 : SUCCESS in 6m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/582e378063bd424ab24b7ebf69d3a1af : SUCCESS in 8m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/91e0356692f247039e3be05fed98d445 : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/799798aa2d0f4ab0b34e7ebb9ecb86b5 : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/58f82bad3a8a4202a30312bda7789b19 : SUCCESS in 9m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b3784620b11943eab442286d603c0da2 : SUCCESS in 5m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bc7e23d8af2746919affccb2c53142ba : SUCCESS in 10m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/de5198823cc1413bbfd7a901673f7caa : SUCCESS in 29m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/78b6272ba46e478db75c044315510589 : SUCCESS in 11m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2a761cd731c1472cab4f8e390a4f13cf : SUCCESS in 13m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/48244bf2d6034db58721e4ddd22ae015 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f0629eebeb3645a38c0f827f1dfb35f1 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de77d372ef5c4cb08ed70df726a8a019 : SUCCESS in 25m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aa97e050edb84913ba62dafc66971f8f : SUCCESS in 5m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/84716cd7e97a4579b1458c005fc470b6 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9dcbf7f415904f1d82eb9ca52f7133b6 : SUCCESS in 7m 00s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/121f00e2731147afbc7d7e4251a14377 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/535508e37e2f411d8a9c0b59d8c00dd9 : SUCCESS in 58m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e1683a022cf49f4b9f0d2a949c99e6e : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4bcd759bb27b5d489bc8c8d61d1856c7cc3c14e8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-31 15:46:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd7d75227ec6122b2615499df4c5c914abdb6068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 17:38:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91638f3e2c1144a485a3300dbb14c917\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065c5734a91042d4a94f3969b64b80fd : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8693648f5c6947f6b11a14ad65b9645b : FAILURE in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8542d59f9b924cbdbc267c6e517d0af2 : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28143cb8fd8b442dabfa5272397b1df4 : FAILURE in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/370773dc496a4a50b3f23e1dcf051504 : FAILURE in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66da4d192d2b4bc19a452acf2387ef2f : RETRY_LIMIT in 1m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f93e521bcc4e4ab08569253941de996a : FAILURE in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8762c2586ef14b61be119af47fcf0d3b : FAILURE in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/836f271397924296ac479b71272fd0fe : FAILURE in 7m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c00c5b52b7624501b5c4526194a89b4b : FAILURE in 11m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6cf95a24df5a4f3da36f2195aa02f9d1 : FAILURE in 9m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3b30505d62ed4259a6e0d6c120464920 : FAILURE in 11m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/42cd84f5e32c48f5aa200518b4047b62 : SUCCESS in 7m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b74959b2dd9c431db2772028908faaa1 : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1c9114b1e45c4e5390cf42362d55b690 : SUCCESS in 5m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9a2c806dd1c3423f887a4a50e228c85c : SUCCESS in 11m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d1fa199f18814befb5638423bdac4206 : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e6af6f67c8d848e98e7baf49cb3f958a : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d7b3561de2d7444981047e34e4c2504f : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/064fd0d79e51450493a7383b25f8032d : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4325ab0f7a84c76a0455461e111ad82 : SUCCESS in 7m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/45429f1343cd4db4b99b9b5cd76b14a7 : SUCCESS in 8m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4c67a7b508a24bcdb577212f1c69addb : SUCCESS in 14m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9b1c079724b248cfbdff3afdf0b348c7 : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d2f0e3d641434a04bfca8c53854d39f8 : SUCCESS in 9m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d5a2c784119d47f58b1dda3b84dc30a8 : SUCCESS in 4m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9841ba90813e49db9f5522fca8f2d13a : SUCCESS in 10m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0589b71b462e4ef19ccabe6f90673b06 : SUCCESS in 23m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/04839535ef4c46f2bc666f8dbc525814 : SUCCESS in 12m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c7bb10dae7f1454eb81315901bf72d6f : SUCCESS in 11m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/757677a58054412dbd4d5b9312829e11 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b26895f0ad0478795c36fd8eefd6d66 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c13a3647cc14612af3de78345df6309 : SUCCESS in 26m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3c2b0f825f38422d9d685a167b21dc29 : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/174a3fce27484fccbf87cd0d968efcc0 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb720a78de5c4ac0b9b3f2b1aebc10f3 : SUCCESS in 5m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f428dcece58e4e3495ce9a3fad3c363b : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2589a767e5dd488bb4233abb00b1c859 : SUCCESS in 1h 04m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ab337582b0b4389b4fc96c9c37f360f : SUCCESS in 1h 44m 01s","accounts_in_message":[],"_revision_number":1},{"id":"cf06f98b65a3ac11c64ed53e6fbd4d29aa6a1dca","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:25:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25b7438b9c80bd88d457638465b14019a0bb2119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:02:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db40264be2a947a38c05cd3875861b44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/672fc24d162c4653b0ca5dd3adabd660 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23b850fb9d2340d4b27bcabd3dfd70ed : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f5456f82e41439ca9e7c3686ccb68db : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79038c662e094ce8890d1252eba8541f : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e855a4cfa8cb44f4ac08c541e51c407c : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e8a128ef7444eddbcea710342a06c05 : RETRY_LIMIT in 2m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3cde91a4fca410c986be71c3ef68f23 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c3fb9899e96497d9782fb3558461a0a : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29bc04b7e02a4d0698a2b376a3da5da8 : SUCCESS in 10m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d2185c4cae40415e9450a680ce0968f5 : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/76672d4111f342d79d2f99a41408699b : SUCCESS in 9m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f604c5e61d4e4e0fbed7703ff24bfd25 : SUCCESS in 11m 54s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/56a371ddb0e645409febf4df830b72d1 : SUCCESS in 7m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/19a5767d6fe1440cb75cb9c05a289834 : SUCCESS in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/006ef1bef444432e8075d074cfc1ee4a : SUCCESS in 5m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/822521823d0b4985828835ad8bf1b99d : SUCCESS in 11m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/985c0c786b8b48d1b06d3fcb272430e2 : SUCCESS in 4m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6d3de23efad24b5ca6162c528b5057d6 : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ca2810fd8d544f7d8a191ef6254a2bf3 : SUCCESS in 9m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/18fa3ab3d3ff4ed6960c782fead588e6 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9bb10a18ce81478ebf58dc7bb75c3bc2 : SUCCESS in 10m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/401f857211214addaabddae642bb8113 : SUCCESS in 10m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/120ca179649a499b9a58a8cb352f8f6f : SUCCESS in 14m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/18e610246c544a5dbc2dab3244a5ad80 : SUCCESS in 5m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6372e3b0780a49e1b2978a380aca809a : SUCCESS in 10m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/052ab31bb491402c9029edeca583d635 : SUCCESS in 6m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f688946cfbc04b95b2b45cb6e8e8a5e6 : SUCCESS in 11m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5c3f41b032c94b3a9645bd8355d6055c : SUCCESS in 22m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6fa2333378604e49854ce86ccae73e62 : SUCCESS in 13m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6a38d58cdf774c0b9ccf63e8059183c2 : SUCCESS in 15m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/340fcd0f692a4a1eb7ffbea5307b58f6 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/06a1d9fe3a2f4cc1abc2dd0a7efe05ff : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b083d174c1524a05aa08228fb08d461e : SUCCESS in 23m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3266bf243b6545b0ac0dfd236e99fc31 : SUCCESS in 7m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4d2d16bda41c4604ab3bc78223bfa713 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/561f5a706fff4c9fbd2837f81701a226 : SUCCESS in 7m 42s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/88693538b2264d0db0fed87863e05c87 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f017bcc08714b948c9e32375bc8272d : SUCCESS in 1h 04m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/789b8945aef549a19e73ec5dc1cd7bba : SUCCESS in 1h 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a347d0c0d3e0ec64f767960fdfcbe1ee7948933d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 01:52:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"970b6c10de925de80bc6633bc1ed35534b284bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 03:39:48.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"82db35194bbb30d31298703c5a52de776e2e2605","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-01 07:44:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81ef8944ab6618a31aafa2ccf6b563ea0189f9c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 07:44:54.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"3f0433912d418779a9bf4e40f5d4dc8a1b10da16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 07:44:57.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"6f7dcee7b8a819ff6a363d6a87c95cd73656b2e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-01 07:45:01.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"04f86ee76d015ef376e7d66b541fb6f71f451ca3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36b9d0f831853c2679f4b7f394f3c4b181206b48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 01:55:38.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"2550d6a547ec30a3f864c54a4c5b8c588a280f7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 01:55:53.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"ac23cfd61070d73f845e29c31b6508d182de6775","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-02 01:56:12.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"c9c3196c860675e004b91a5627c5c7c26e296805","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:12.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f915e0c455159b52fbbaa02c6774e741b11c22c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:37.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"69d13afa0b100006c78db48bb08b2fd0b5f64ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:52.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"699460abcf7b44920109885832cbb0889914708e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-02 17:57:58.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 893372,1","accounts_in_message":[],"_revision_number":1},{"id":"1a4e35a07b233feeb9de7418fbcb22120c0b4bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-07 12:17:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c9dde0b2b9528b1faa4e9617f3c73f06af40f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:03:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/169a085466d241968bb59641c6e7b4f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99764e8f56e34b5a92250fadb10593ba : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb6fa1fbaae4712b11b188978bac670 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fed8db0d0ea4bfcb5fccdfe57c1b7eb : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e5f0c4760d9418990ceb320de30581e : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/137e5e5909304095b06b8a27b0598d83 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbe8bf42b795402f9e551065f48007ec : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ecd8549b1484aecbfb31247d1fd549a : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4c835355fc5416e809dca8f4d142f27 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/761c37693f114c71bd3e7ee80b84c3e6 : SUCCESS in 8m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/db58a05703f647a9acf4a7724f3c277a : SUCCESS in 14m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/96cf04e6cc8443ad90d77dceb7775b21 : SUCCESS in 10m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9027179d03ce4103a6d9210bff1bdc85 : SUCCESS in 13m 42s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1c0998d6c80b425891eba2749cd2a671 : SUCCESS in 8m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e213142114e74b4186ccc2a6b0b6c746 : SUCCESS in 4m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f0e39a5387e34893a3a59a18fc1730c7 : SUCCESS in 5m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0549de6ff1d9467abef9dd84c1cd3a12 : SUCCESS in 11m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c369ebb764c1419489889a7984dd1f88 : SUCCESS in 6m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0550810ab5cf4492adc4106f5a1b500a : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c395671e7e7746678c04517e27b6b9e8 : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b894dbabc3004ab8b7daf082d7d798a8 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f80079c40d84201a11cd274e79ed619 : SUCCESS in 8m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0b4b1b8b325b41dba0b65fd132184bab : SUCCESS in 7m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/039fcf422105445faedc4b234ce86bac : SUCCESS in 11m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9d20ab6d764141e19e9944ceeccee568 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8e0f830fc4f4477f96cfa3cae074dc9a : SUCCESS in 11m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e28d8b6bde37448ba132bb00d9a4f1ff : SUCCESS in 4m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d54584faeb7e4fcbb852d5d0051c74f9 : SUCCESS in 12m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f3e77aa4d8a8459da2049570d978c9a3 : SUCCESS in 25m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3c140b33effe4f5ba4e6b71ad53ece4d : SUCCESS in 10m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/71fd3f1635e240c09cb9e0f19a844c7f : SUCCESS in 14m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/efafe30c1b9340a8b65c72dd35e3648a : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2800906ba82943caa117121675ac93bb : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6dcbe8058cb4d5aba2f979ab4eec761 : SUCCESS in 25m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c4349e71a29044e7aba6f0a1dd9a1655 : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6cbf416c1ed745279b5fa5b6b41d418e : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1396f724eb3c490e99596345b2f89eba : SUCCESS in 6m 17s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/fb1e91bf902b4a2cbeb31d0db0e8728d : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/365f4636be384d38aa24ceff662581ab : SUCCESS in 58m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5beb5bb3e666446b9198c574ef0a3cf8 : SUCCESS in 1h 40m 06s","accounts_in_message":[],"_revision_number":2},{"id":"63a330daaf3d950c332620a513de54fe85ef7f4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 15:39:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5448884768575efaf18cf52e1204eddb9cb74ced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:40:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f270f3f2a20b2832d34217bf8f01414293f9af4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45810e902d144154a6fca02e71b181fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de913edff46b401293b76a833d06464d : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c756788e37d4931a2e4dec69b951e6b : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e1dc51bd197413291e7014b4ec818dd : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c49a533d107f401a95d7ef69bcef5db6 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e597c205bbd34d3b89edadf2969bdf6e : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13760d4a0cd042729a1420a4843c2a9b : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2b3006594e046ddaac0d671c7c1fa33 : SUCCESS in 4m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89dcc20ce1b14f08920d38d7cc13a50c : SUCCESS in 10m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17b827dab257420ab0a0b22d70225b37 : SUCCESS in 13m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dae21f69130a4cca8ee761db31cd6a63 : SUCCESS in 10m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/dcdd8194d3e94385ba27838eb96d10de : SUCCESS in 11m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/62533e7b48844ba3a1893a0f1c44ecc2 : SUCCESS in 5m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/69d82cbdf2644e1ca3c0ad16353ff2ef : SUCCESS in 6m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f8ca69e575714899b64c1a13f4c1e153 : SUCCESS in 10m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1d9b3bff51c34490ad6c9471b50abb99 : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5fd098ee9fbd4efb98397f7df7bdc0b2 : SUCCESS in 9m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3a97363472e04d7b84075e5d77fb1a43 : SUCCESS in 8m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/61883aeb82504f2a91f9199112eb9979 : SUCCESS in 12m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/10328daa4d4341fea55ff1a30230c915 : SUCCESS in 4m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5450f998f92f4daaaed9ce2ccab93906 : SUCCESS in 8m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a8f2b38d9e6d4ccdb4697758457883a9 : SUCCESS in 5m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c625bbde64244557b6ee172bd8d0cb03 : SUCCESS in 24m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1b67bade9a9f47fdad89836ed16a82d0 : SUCCESS in 11m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/68430e3a00f54eefa8676c4a5ca24f5a : SUCCESS in 14m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/699a51589f6a4c1fbef76020594cfeb5 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2c81c1e5abd44e58b285c57a1d886ae : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fd60d07cc564d30a912e95fb30910e8 : SUCCESS in 23m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/64847dd473394ad2b0a7bdb35d855b2f : SUCCESS in 6m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fb40b0a38ab644df925bce901c09b674 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9678baeb3f5d48348023fa5fd819e949 : SUCCESS in 7m 17s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8c85eefa574644ebb75893fd25f02cb2 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1827cdc8fc57480e89ae51772a118439 : SUCCESS in 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/647d77aad0b040818021498845cb80b4 : SUCCESS in 1h 29m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8a5cedcbd90816f07932c70b867119bbfeb03333","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f3f2e646f377d75c869dea68083f4b3aec2dbd02","revisions":{"16063530b7d520f1ca38e58dbf53107bd66a8dc9":{"kind":"REWORK","_number":1,"created":"2023-08-31 13:50:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/893372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/893372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/893372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/893372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/893372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/893372/1"}}},"commit":{"parents":[{"commit":"0341312176f39d2139baa6f7f829d4bbc5455489","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0341312176f39d2139baa6f7f829d4bbc5455489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 13:50:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 13:50:11.000000000","tz":0},"subject":"update constraint for osc-placement to new release 4.2.0","message":"update constraint for osc-placement to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: If3d6894a0ddf168fbdc645870c347777c27062b4\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2badf0740fe8ca2575835912202b523649ec91fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16063530b7d520f1ca38e58dbf53107bd66a8dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16063530b7d520f1ca38e58dbf53107bd66a8dc9"}]},"branch":"refs/heads/master"},"f3f2e646f377d75c869dea68083f4b3aec2dbd02":{"kind":"REWORK","_number":2,"created":"2023-09-07 12:17:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/72/893372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/893372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/893372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/893372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/893372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/893372/2"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-31 13:50:11.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-07 12:17:29.000000000","tz":120},"subject":"update constraint for osc-placement to new release 4.2.0","message":"update constraint for osc-placement to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: If3d6894a0ddf168fbdc645870c347777c27062b4\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2badf0740fe8ca2575835912202b523649ec91fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3f2e646f377d75c869dea68083f4b3aec2dbd02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3f2e646f377d75c869dea68083f4b3aec2dbd02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
