)]}'
{"id":"openstack%2Frequirements~893388","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I23ba892e6dd7f2566e5af6626c072c2dcc5a4382","project":"openstack/requirements","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-24 08:32:49.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-24 08:32:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I23ba892e6dd7f2566e5af6626c072c2dcc5a4382","subject":"[DNM] Fix gate by bumping tox version and removing ryu","status":"ABANDONED","created":"2023-08-31 20:28:20.000000000","updated":"2024-01-24 08:32:49.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":893403,"cherry_pick_of_patch_set":2,"meta_rev_id":"328ed7e9aabfb021458f70a71513f5d4e2bf9262","_number":893388,"virtual_id_number":893388,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-15 00:52:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 20:28:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-08-31 20:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-31 22:22:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 06:58:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-01-24 08:03:36.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"c802d93c71c5d36c039bddd1a06d71e2b3c29cf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:28:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"efe2db13d53e37b58ee0f7dd09b12ad2561690b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 20:33:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f91c78bd4e764f35b0c1f07c252e93cc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5ced98410a204bb0a02d23698399415e : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e1d5879b5bdd5a09f03983e75f7eac3fe3f8e3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:22:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d71731ef67d4820b41354ebecfd17be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b6c3d68722437f807aabd45a4a8d09 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c227fa07b5c8430c9257311d7673b082 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/922e146c1a1944bdab7ecac75f62a05b : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/754a9484a0644fc0a4f8862665754d3e : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b4d880a333d4bbabe4401d1f07db60d : SUCCESS in 4m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aeaa23825172487daa8a5256bcba4707 : FAILURE in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4db7d7e5aa594f9fa33ab2d118c80fe9 : FAILURE in 13m 56s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c5471ceb44294cc3a5412793a89b4108 : SUCCESS in 7m 57s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f359dacdedda4bb68e9d7329b114bf63 : SUCCESS in 6m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fd8aaf4a33ac484a9e438aedba1bf14c : SUCCESS in 7m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/981567cd04cf4adc80d5e5221fdc95d9 : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/de6dd8e45ffd45cea76f50ed8845dcd3 : SUCCESS in 5m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c4934cd871c046ac843146e3e8bb1041 : SUCCESS in 6m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b15c573f6d44453abfd2b6c004eeee1f : SUCCESS in 8m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0f2ec67cefe4477d8347c9c32ce7d3d5 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d03fd84dcace45429f1bb1ba9f3bd37d : SUCCESS in 8m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fb046965bbf4462e88e94c8b91118b57 : SUCCESS in 9m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5c153eb4fdb14673821fffc5daef486d : SUCCESS in 13m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8eb6a7cb06af408683bb28ad9e1c11fc : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a7de7374c9f84824a7dc0992e92f0bdb : SUCCESS in 11m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/85ded51526544c5e9a8841b889420303 : SUCCESS in 5m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2bbc4bb207274438a3c41183e81b1e32 : SUCCESS in 9m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9dbe9e1d1caf4bd6a9e6a042ff9e9ac9 : SUCCESS in 23m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fcfd369565f74d32b4ec0ae3ba4f9953 : SUCCESS in 13m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1ba257d8374a46808fcb2b6cd399aa76 : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b18dbdcccb94aad92aab54be05a1c4d : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c055d8658b04b4b84a9b97ba04c21ce : SUCCESS in 27m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/105eb77aff514b98bd6ac81d94081e10 : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f37a1da37439460fb66d555b998f87a1 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c950469ffd74e06923919dae8789ab5 : SUCCESS in 9m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7e462d59736b4defadac421fc5fb33f1 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ecdf43e68954cb1b6284c1d73a7fb1b : SUCCESS in 59m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf2a5622c88b4d8fb6e9ff59366230fc : SUCCESS in 1h 48m 28s","accounts_in_message":[],"_revision_number":1},{"id":"dba384061fe069009f11e94da500855f39058528","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-01 06:58:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19f1e87438fd7366ef043b70d71f8749a8ad79bb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-01 07:02:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee374732f7a5950579a2af670c2ee4e860bc6778","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-04 08:36:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b362db90e1df266a921e7175bc8c198a099348b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 22:19:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0163ec021b758e74b1603e973cd35221d7bf17c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:25:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88af5cc421a44cd29737e95517eee4dd\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c1a016055bd14155a8d537d0c7c53a69 : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b950d649ef3514d994af14cdf54938b8d72d5768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:54:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5e302e068a04a328830c29b0e7ec6f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b3d03cc623640c6bf38181d0da2034a : FAILURE in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fd0b82344ba4e1fb2ad40c3d5ef481d : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36869e089a3c40f29794673464b706f2 : FAILURE in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60b1ff9791054d7bb98eac887987c1cf : FAILURE in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9445832140794d46bba78ed76defa03d : FAILURE in 7m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/92a6784d27154e0099bb94604596daa8 : FAILURE in 13m 50s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/171166841002436a994592af45404130 : SUCCESS in 7m 23s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/71817999a5ff49aa9126d38b8464f011 : SUCCESS in 4m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7e0b095a9e574059acd209a73ee49346 : SUCCESS in 5m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d410b2eba3644008b4a693b25fcee3f5 : SUCCESS in 10m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/675e29787e024c649b77e85748dc6640 : SUCCESS in 4m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/39baf83e56b64cad83dc5ee67c98be54 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f170d8bd304e4abb982868e912e5dad9 : SUCCESS in 6m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fa3a0d1f626c451aaebc22e31133ad8e : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/762ab1486d88475599a6866bfe9d104f : SUCCESS in 6m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e9fb5faa38f643059747f4190fbceffe : SUCCESS in 9m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/350a7307473543e394df233d3ca33fcb : SUCCESS in 10m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/623b77649846434f822e5ef725e946d1 : SUCCESS in 3m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/460e5208cb9c4ca78814e4a4a8e66125 : SUCCESS in 11m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5418e53508814e18b73a0115bc2768e2 : SUCCESS in 4m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8fba82e5fe2d4273b0e59a5152a0e37b : SUCCESS in 9m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/76d35258a5324503a8955a4f95da3cce : SUCCESS in 17m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6a40287cd14c48fd91fc1b34ea210729 : SUCCESS in 13m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/46fe95f4c5e645fd926f84fe9574f170 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/217df8695c634cb4a7b5d927546261a9 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/878ad694fdfa419283f30bd8c5920385 : SUCCESS in 19m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4b1c63003ab949a7a7544bcef68039b0 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5770711dface4ebcb8bd3f3f645277e3 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf89d224e11c4bc2834d5f05d37a5c0b : SUCCESS in 6m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0b9789990f0144f1996b2ecea03cf1a6 : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f48fb0e02f7b407f93621c1ea27abb54 : SUCCESS in 39m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/745f17377b964e6e83b8bc67c0c0edc4 : SUCCESS in 1h 29m 32s","accounts_in_message":[],"_revision_number":2},{"id":"49b3ce691cd1933d203defb4d4e7080116415817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-14 23:18:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"34ab6c7401a33e1ddf4c7b9b45c3335ae3cc30dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 23:35:27.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d8822cb5e184b66a54e1027bf4b3b1f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/83a4db860e7b42679147991c2ac5a8bb : FAILURE in 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"abd3b5f39643c437788aa1377a123cab2ecf1666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:52:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a4a5590916b42c9ba7b9307f6809e4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ad1617acbdc453c87df1f37468c1dce : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75b4644475784d7097467a76f947ae27 : FAILURE in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c9187549e4549948e66efc7b84b418d : FAILURE in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d0a46ac8a804805b42cf3c5d8995647 : FAILURE in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/76141aa9cbaf43969ae6086f3c966421 : FAILURE in 8m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5aa344a6081a40f48975f5555dc19687 : FAILURE in 12m 00s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/66782765c2604bf4bac9216991ef8c58 : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/62da114c144c4667a48a0386f1c3ef86 : SUCCESS in 5m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/94199722025c4c7b8a08bc849e1be302 : SUCCESS in 5m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c6337262f1104e1ca7d8d8a7fe327c5c : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/618e4e7f2c134e308ba645adc165a69a : SUCCESS in 4m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/086984ed74be4eafa5d1fb6272f44f7a : SUCCESS in 5m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e2baa36f2070427a80b03a01be2f20f7 : SUCCESS in 6m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7b47adf141e14b0b98ce6015d67119ea : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d707547decd4b0290c56a87721cabfc : SUCCESS in 6m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b9c502b2829245dd8bf4d8df49e30fd4 : SUCCESS in 10m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a54ac6a13bf64e06872775d9c4863ca5 : SUCCESS in 11m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/947411513be3429e9c7e6a30d486d031 : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b176dc5a822e4356af7183defb2a60f9 : SUCCESS in 7m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ef419591ef484733a13fde4f569b4ec3 : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b1b735151baa45bebed1001296c0fb82 : SUCCESS in 8m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1c82dd81bfdf4026bd15f85cf52001da : SUCCESS in 17m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ecb20893a12d41ac9e53fa9d87728069 : SUCCESS in 12m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d014635ed2ea4410ad3482a439d14dbd : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c4c32e2c62a84933af2052370d08cd60 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3ebfb4d75d942aa9c987a66cd334891 : SUCCESS in 29m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9e081406eae24529a786d3c13f5e03a2 : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a0550c982cf54f72bdb85f17beebc9a3 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7bc101979dd645589559593661a5cce6 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ceb72b3f10b44f69905d608384931084 : SUCCESS in 9m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15bf60c3758e4f9ca634fdb271429987 : SUCCESS in 1h 12m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6607dc8758b1462bbff07ad1398f7873 : SUCCESS in 1h 25m 36s","accounts_in_message":[],"_revision_number":3},{"id":"c12b461e8ead56ad30798932271a2e5ee4436aa1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-01-24 08:03:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b77dda408fdd597398c6fea82f8bc8c74951ae80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 08:09:42.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92b24a5fecf4e0db185bc8e2cc4c30c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/62ca09f1e44742b48a4cb7bff4ff0d0d : FAILURE in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"328ed7e9aabfb021458f70a71513f5d4e2bf9262","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-24 08:32:49.000000000","message":"Abandoned\n\nthat doesn\u0027t work, needs deeper investigation","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e55aa3068b0bc20b334dc2398e96dec44cdb65b7","revisions":{"a1c80c13080514918bbeffa35d2f9e4426d8926f":{"kind":"REWORK","_number":1,"created":"2023-08-31 20:28:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/88/893388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/893388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/893388/1"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-31 14:57:26.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-08-31 20:28:20.000000000","tz":0},"subject":"Fix gate by bumping tox version and removing ryu","message":"Fix gate by bumping tox version and removing ryu\n\nGating jobs started to fail with latest pyproject-api (1.6.1) release.\nThis release would need latest tox (4.11.0) to not fail, but we are\npinning tox via pinning virtualenv in tox.ini.\nTo resolve this we have to remove the virtualenv pinning, which was a\n\"temporary\" hack to avoid py3*-check-uc jobs to break. This job is\nfailing because of ryu package, which is unmaintained and projects\nstarted to use os_ken instead, which \"is a fork of the Ryu library\ntailored for OpenStack Neutron\" [1]. Hence, this patch removes ryu\nfrom requirements files to eliminate the need of pinning virtualenv.\n\n[1] https://opendev.org/openstack/os-ken\n\nChange-Id: I23ba892e6dd7f2566e5af6626c072c2dcc5a4382\n(cherry picked from commit b3ccca8516377c16663217efe20efa1593d74727)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1c80c13080514918bbeffa35d2f9e4426d8926f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1c80c13080514918bbeffa35d2f9e4426d8926f"}]},"branch":"refs/heads/stable/2023.1"},"0bbdf7c59378fcbc57be27bb1f8398849d36ab10":{"kind":"REWORK","_number":2,"created":"2023-11-09 22:19:23.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/88/893388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/893388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/893388/2"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-31 14:57:26.000000000","tz":120},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-09 22:18:26.000000000","tz":-360},"subject":"[DNM] Fix gate by bumping tox version and removing ryu","message":"[DNM] Fix gate by bumping tox version and removing ryu\n\nGating jobs started to fail with latest pyproject-api (1.6.1) release.\nThis release would need latest tox (4.11.0) to not fail, but we are\npinning tox via pinning virtualenv in tox.ini.\nTo resolve this we have to remove the virtualenv pinning, which was a\n\"temporary\" hack to avoid py3*-check-uc jobs to break. This job is\nfailing because of ryu package, which is unmaintained and projects\nstarted to use os_ken instead, which \"is a fork of the Ryu library\ntailored for OpenStack Neutron\" [1]. Hence, this patch removes ryu\nfrom requirements files to eliminate the need of pinning virtualenv.\n\n[1] https://opendev.org/openstack/os-ken\n\nFIXME\n\nChange-Id: I23ba892e6dd7f2566e5af6626c072c2dcc5a4382\n(cherry picked from commit b3ccca8516377c16663217efe20efa1593d74727)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0bbdf7c59378fcbc57be27bb1f8398849d36ab10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0bbdf7c59378fcbc57be27bb1f8398849d36ab10"}]},"branch":"refs/heads/stable/2023.1"},"e55aa3068b0bc20b334dc2398e96dec44cdb65b7":{"kind":"REWORK","_number":3,"created":"2023-11-14 23:18:32.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/88/893388/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/893388/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/893388/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/893388/3"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-31 14:57:26.000000000","tz":120},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-14 23:18:11.000000000","tz":-360},"subject":"[DNM] Fix gate by bumping tox version and removing ryu","message":"[DNM] Fix gate by bumping tox version and removing ryu\n\nGating jobs started to fail with latest pyproject-api (1.6.1) release.\nThis release would need latest tox (4.11.0) to not fail, but we are\npinning tox via pinning virtualenv in tox.ini.\nTo resolve this we have to remove the virtualenv pinning, which was a\n\"temporary\" hack to avoid py3*-check-uc jobs to break. This job is\nfailing because of ryu package, which is unmaintained and projects\nstarted to use os_ken instead, which \"is a fork of the Ryu library\ntailored for OpenStack Neutron\" [1]. Hence, this patch removes ryu\nfrom requirements files to eliminate the need of pinning virtualenv.\n\n[1] https://opendev.org/openstack/os-ken\n\nFIXME\n\nChange-Id: I23ba892e6dd7f2566e5af6626c072c2dcc5a4382\n(cherry picked from commit b3ccca8516377c16663217efe20efa1593d74727)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e55aa3068b0bc20b334dc2398e96dec44cdb65b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e55aa3068b0bc20b334dc2398e96dec44cdb65b7"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
