)]}'
{"id":"openstack%2Frequirements~893522","triplet_id":"openstack%2Frequirements~master~I931a9aae8caacc119f9016d240cc91430856f709","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-09-07 15:39:53.000000000","reason":"removed on reply"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-07 18:43:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I931a9aae8caacc119f9016d240cc91430856f709","subject":"update constraint for python-cinderclient to new release 9.4.0","status":"MERGED","created":"2023-09-01 14:31:21.000000000","updated":"2023-09-07 18:43:17.000000000","submitted":"2023-09-07 18:43:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893522-new-release","meta_rev_id":"396ddde85b36c7cc822ba8a7bdb7921f1bc2ae3d","_number":893522,"virtual_id_number":893522,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 18:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-07 15:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-07 15:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 16:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-02 17:56:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-03 01:17:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cd098801573b72f692b9d92032aac42ccdd19cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-01 14:31:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"111fcbd607e9626179176caa87b2453b121be3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:30:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/191f215f05154ef29bd837b916a9abfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86b4913633542a085cb18d61e156ba1 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e1647f688d4e7388d64b9c9c366491 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b9d2ccd094c416fa19e15b99518efe4 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91eda5fc60054340b0d59056c7f63d21 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf6825e2bc5449bfa6c62b199d688b64 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bbec9a47e534c2b8832c5ea501e70b0 : RETRY_LIMIT in 2m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9e89b9884ff4f31af400824ebfafe4b : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0aa336137c11476e86dba6f8779a333d : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4943b7f6285d439593571652a89d4eb8 : SUCCESS in 12m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8e2a08a0cc8d4f208b2b058324e474b2 : SUCCESS in 12m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cc48c65152154b6c9917f7808535cdd9 : SUCCESS in 8m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/33e16436033d4c0996ef64c2e385682c : SUCCESS in 14m 07s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c3f68ce9c0d747058b07ac01abd668d1 : SUCCESS in 9m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bab5a21395544638adf946b9de47b88f : SUCCESS in 5m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2c80651aa56842088c06c3fbf7264dbb : SUCCESS in 5m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c763cdde9ec74ebca67906e0ae74aec3 : SUCCESS in 13m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/db4178e0ff0e4926968bba951745706c : SUCCESS in 5m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a803948a458d43809ced7130603332b8 : SUCCESS in 5m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/10b8ea9fc11f460695560866f0db773f : SUCCESS in 9m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ca9367822a324978800263565d4744eb : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eae03fdebdb34fd9b613cc1dbd8d0bea : SUCCESS in 9m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2225edf69bb349629a29cb54e8059484 : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/287212ecc6ce4982aec2eb24550c2418 : SUCCESS in 12m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/31b803e36eb84bea9b88bc6f60e5353a : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ed591641d82049cab7957be23ec038e9 : SUCCESS in 10m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/baf866b2e6834f7ab3375c378f7a1c22 : SUCCESS in 4m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d7aa9c0cdbac428d849dad60d202a0fc : SUCCESS in 11m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cee201065c5649949a4ef02f1935c2e7 : SUCCESS in 24m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/893768949d1d4965980a9fa7c471cbfe : FAILURE in 5m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/154ab93ebd8e415a8674adc90235fcf3 : SUCCESS in 14m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fa6464659fb24d6b9ec8c39e664a8251 : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f58222ebba2c406889e1e3acbdb53488 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9698a3f1d64b48b394fc8f1d50fcdf02 : SUCCESS in 25m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4da98d58de8546fea76f78d0d4dc55c4 : SUCCESS in 5m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/76f30b3442d54619b59527519b0d2247 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/faf5bbce56b845b5be7b649dff492526 : SUCCESS in 7m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/beafa6de57be4c0eb7c30918cbabd38f : SUCCESS in 16m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cd80e8bef3e445fbf4c7873f35a0b58 : SUCCESS in 1h 06m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afb5fc96ffb14b7c8e6b887ab32b1377 : SUCCESS in 1h 49m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0f104dfdbf112bd1c9e9a8be7ec6b15fdf885c28","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d3e66222fbf549271679d53f378c040330078a7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 19:25:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/424c668c559e47228413dbded6d5b78f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece7684b0cde480b881df5235756118c : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ae1fef243f4c1bb5c3e851c50b0087 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53cfdaf879f146d8aea25dcbf19dd67f : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45cdb0cd43c74ce5a0125543149ec054 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44915e151ab54509ac95dc1603b72bb8 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94986c372df643f3b99257156fd20a1d : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/befa14964c3b4666a8c8ee909b790da0 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed8f50a50b89401193dc45497bc72041 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9d2274dd8f1e4362ac736c47db87fa93 : SUCCESS in 7m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0200d6967f1646f8b5def560e1de218f : SUCCESS in 11m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f871376928f428f922c4c9d5a400fdc : SUCCESS in 8m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/511874ee6fc042bf9c5602546c97496f : SUCCESS in 12m 10s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d55dfb4fec744087b5e2a3572ee20843 : SUCCESS in 7m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c456c849fb0240c392eccc075a007d1b : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a22886ff515242ad9a2022397e7cd258 : SUCCESS in 4m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3deb4cea84674488897b3493c4c10f2d : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bfe80398fc4a48e881950957e4014483 : SUCCESS in 3m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c8985908444d4854b3e17ba2ad0ec9d8 : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/97c92bab9add41428312a03743ef4ad8 : SUCCESS in 8m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/341fd5860819422491a3cb7fc0f0f436 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7658a12324a42f28adf42aa74088fc0 : SUCCESS in 6m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a271f80f2e18493ca792d1ccaf226a30 : SUCCESS in 7m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c200081f9c934b088978dc7b66a260ee : SUCCESS in 9m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4632d692151c444190ec8cb3e62a0b9b : SUCCESS in 4m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3154479dc7964a04ace984bdd8a841d3 : SUCCESS in 8m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d6b9dda225d34669b75baecb87dd9f68 : SUCCESS in 4m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5a725fb8639747a6bf093601800a3049 : SUCCESS in 11m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ae034df67ef3443496d5fb96ffeb63c6 : SUCCESS in 29m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4f40559a6dbc45ceabc2a87a80c01806 : SUCCESS in 11m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3f2f44fead084a2bb506c415ebef95cd : SUCCESS in 14m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f21937f96c2b47ab9af2d431a2908284 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cd09ac8e7354da4b58cb7161fcf4dd9 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb6e041f13f44e058b7cbb04bc84c17a : SUCCESS in 28m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6aa30f90d8504151b3ab3f892ea356ad : SUCCESS in 4m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e481a41b96f44ba4bd77aa10ef69ef5e : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3082b72a76624dba9a0051a1b5e094f3 : SUCCESS in 6m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/726788ed6c3b4ddcb34c9fc8b9c6b31c : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79c230eb29844ea19178053562adae7c : SUCCESS in 56m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcb69d35b3014600ae1803e8ad44bbd5 : SUCCESS in 1h 17m 55s","accounts_in_message":[],"_revision_number":1},{"id":"83ac1fc917661b61406eabd24a94048bbad6ba32","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-03 01:17:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75fd0185b1eeb02628b10908d4b10adcc7b196ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-03 03:11:38.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 893522,1","accounts_in_message":[],"_revision_number":1},{"id":"a610c2c467bbee0eae76c68d0290461ceaca928d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-07 12:17:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eeb71a034ab7a906bf963ef8c64ef268bf0ffb18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 13:51:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03863df4231f4d40b31b3cba82ff731d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95d57977b82437c9dedae16bb9a57b1 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124cf54f07874c2f98a0bd97942fc150 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a769338b8ff40a3854d08be6cb072ea : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66b977ee45034431aa3d59d2b8361654 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12f17029d002492db1af8099b1f46d7e : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/060f520035fb40e1acf21d96240a96de : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa601d646cf64e9e9543dcee05487d20 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a4700f2b6ed4b9abea1fc415ec5da17 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e930ea661d364b6181264687edf7ba49 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3de705ce7bcb46e8a9666ed535c26b0d : SUCCESS in 12m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e54e1565960342e4aa529df456aea77c : SUCCESS in 10m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0664c74201d0414da714819fee5e2827 : SUCCESS in 12m 22s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d39b4261889545f2879a16665a2c9cc3 : SUCCESS in 6m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0269856c2a2a4aaaa2f260ea1fe4f17c : SUCCESS in 5m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/686391ca58954698848aafeb8edfc3c0 : SUCCESS in 5m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a19eaebb7c1a42cdab0e41786fa19540 : SUCCESS in 11m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3988554305df4994adb767179269432a : SUCCESS in 6m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1547b0157ecf4469be65e03ce00f585c : SUCCESS in 6m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4868b163215245bfb4ce389c1a67e8f3 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ba7dda3c15b04417ba74319d87bab951 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37b96b2ebacd455f965ff6b5f247082e : SUCCESS in 6m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ae93c9485a32417b8cdd330b14566a7d : SUCCESS in 9m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/610ef70eab4c4052b9bb92c77ded8de0 : SUCCESS in 11m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/12e76648b92e468bb1474be8c2f96c20 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7fe001bdfe7d4f01830f9dee0b9c8e95 : SUCCESS in 10m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/734c4b6e94a64e70869b1d07333312b5 : SUCCESS in 4m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6ecc7ca352cd42f7aa6f30404015668d : SUCCESS in 8m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/af8968b743ef4a8aacdf87713f6d856b : SUCCESS in 34m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a39dd04bd8a849c2a31bac8566cafda9 : SUCCESS in 12m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1152376e605c42279509dbfbf9f57d33 : SUCCESS in 13m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/74a582448eff4c19aa29c60dfd1e6249 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88c0fc8ca58546db850838fb4c545b85 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9853cb007d34884b5caae4ecae01b6a : SUCCESS in 25m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/10be908cc8f94462b35eb3c963b7fe5b : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a5401af64f514975be85b111dbd71ddd : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c0776ad2d894a5db39ff410babae3cf : SUCCESS in 10m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5f0b85892c8e4790b2e9c4fa583319ac : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35c93d18e3c54d71b95b8823586b31a1 : SUCCESS in 1h 06m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06fb50e8affc48dc9721fa89029a6fdf : SUCCESS in 1h 29m 08s","accounts_in_message":[],"_revision_number":2},{"id":"202c9222dc3eabf75f9c9bcbf4ee90a33d866f65","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 15:39:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2338f7afd39b62ab1fee1725dfdd0d2ef1b5bccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:40:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4965d9ccfc1565a0cb29f141e457c42278440a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef4af1698974dfc9887ed858bc594e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954b4dd185584da7b6808d1c6b8fa3ef : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c16c0d3e6046098bcd7ce34af2f9df : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1da13dd595f347bf990e9c9464b6e755 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/652de307ff9747a8a46927a603baa012 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/693c4382b1094fe689f41add14f414af : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f2e5f959f3442bf850128932d920405 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b9fa7e7ec3e4874ac407596e4185b48 : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/023b7c622cea4193842a15f9d2aba8ca : SUCCESS in 9m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f36ba73eb7214cd182fdb9ba1a31d060 : SUCCESS in 11m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/652474dbf24f4eef900f8406ffdb5f02 : SUCCESS in 8m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b4e48a28604f433ab678ebb4b46c48d2 : SUCCESS in 11m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7f5cc8ae9bc841069b089e81c215b6bd : SUCCESS in 5m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/67d34b01a55941408c4e0af4fdaf380a : SUCCESS in 5m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c883b9919e1f4bb78d6461f58693bc9f : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b005460e7c0b478baaf8213bf3f417ae : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b702c1d9096847f1abade8eb492e0991 : SUCCESS in 8m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a2430ed4f4ec4f69b1d5c2e0e997aa0d : SUCCESS in 8m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bcdad41dd4194d4390da139e71fbaa4e : SUCCESS in 12m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0392c820e5004506a6f767856bac6ef7 : SUCCESS in 4m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e57238f176354b41aa109ae889a7e49e : SUCCESS in 9m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d45ee316376e484abb9e9f91ed128720 : SUCCESS in 7m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dca178cbc5954a00b80383bd8d9d2922 : SUCCESS in 26m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/37819d386f2e4b93b9f4bd3698773039 : SUCCESS in 12m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2073ba498be14f6bb63182d1423ba3e5 : SUCCESS in 14m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6188691fcfb640f8ab1779671f91afc1 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a295bb5ac584031b5a65fc63d9daa74 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9ae137f7f4f4555afb006cc68fb0429 : SUCCESS in 23m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ac553b3779204c09878ade94930742c7 : SUCCESS in 6m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c8877373ee1049d684d020a427eefd9a : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38e6930a564b4e8fb5b38c32296304d4 : SUCCESS in 7m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a5c567f1af7249e2a49e0132bd9b08b1 : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/019b082155c744e7a5a0bdafb95cb81c : SUCCESS in 47m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f63608d87c65423ebd308106acf17048 : SUCCESS in 1h 25m 17s","accounts_in_message":[],"_revision_number":2},{"id":"396ddde85b36c7cc822ba8a7bdb7921f1bc2ae3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff9d3a243652671945b6ca2da5d18703cd89d30a","revisions":{"c5c72bd7d4a51b3752ad056573a1e611db909584":{"kind":"REWORK","_number":1,"created":"2023-09-01 14:31:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/893522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/893522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/893522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/893522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/893522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/893522/1"}}},"commit":{"parents":[{"commit":"08ce437e5875c695913e03331191eb080d60c7fa","subject":"Merge \"update constraint for metalsmith to new release 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08ce437e5875c695913e03331191eb080d60c7fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-01 14:31:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-01 14:31:20.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 9.4.0","message":"update constraint for python-cinderclient to new release 9.4.0\n\nmeta: version: 9.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: whoami-rajat \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: I45708ae2b6d8d29aae98f70b5f5e1a20b1f3ae4e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I931a9aae8caacc119f9016d240cc91430856f709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5c72bd7d4a51b3752ad056573a1e611db909584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5c72bd7d4a51b3752ad056573a1e611db909584"}]},"branch":"refs/heads/master"},"ff9d3a243652671945b6ca2da5d18703cd89d30a":{"kind":"REWORK","_number":2,"created":"2023-09-07 12:17:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/22/893522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/893522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/893522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/893522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/893522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/893522/2"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-01 14:31:20.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-07 12:17:05.000000000","tz":120},"subject":"update constraint for python-cinderclient to new release 9.4.0","message":"update constraint for python-cinderclient to new release 9.4.0\n\nmeta: version: 9.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: whoami-rajat \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: I45708ae2b6d8d29aae98f70b5f5e1a20b1f3ae4e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I931a9aae8caacc119f9016d240cc91430856f709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff9d3a243652671945b6ca2da5d18703cd89d30a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff9d3a243652671945b6ca2da5d18703cd89d30a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
