)]}'
{"id":"openstack%2Frequirements~893527","triplet_id":"openstack%2Frequirements~master~Ibf14e8d1d663efbf167c950b8b7b966fe6e688eb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibf14e8d1d663efbf167c950b8b7b966fe6e688eb","subject":"update constraint for python-ironicclient to new release 5.4.0","status":"MERGED","created":"2023-09-01 14:47:06.000000000","updated":"2023-09-07 18:43:21.000000000","submitted":"2023-09-07 18:43:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893527-new-release","meta_rev_id":"c0d774414b84b048d859596749502a5bb4f1c052","_number":893527,"virtual_id_number":893527,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 18:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-07 15:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-07 15:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 16:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 23:56:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-06 12:39:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"a67539ab60ca78f0ff546efa341971580642f4b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-01 14:47:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c49149890536dc474c1cacd01889ea8d6e7aa22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:45:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5fea031e6b34cf292ebfbc2d1528679\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74e3d6833dd466ab4585cc9feb62fa1 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c1224a5ef124429bd396d8a3883c3f3 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e8ce37b4253437595582dc8c3624215 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e0ad25021f0481bbcdc2d9498d6568a : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7a8e8d97b2f46c59775eb9a47026268 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87eda742d8554293abff0a5f935210c4 : RETRY_LIMIT in 1m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/594cb0977997468d8c7a0548d6004e7f : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23d2804b19f0498da4501c1e3b03bb7c : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3d9c227d81546b2af13544891370ade : SUCCESS in 9m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/599c3660502b45d6b4d834a4471ed82f : SUCCESS in 15m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c7468bca0e244c3eab49ecad6383c5bf : SUCCESS in 12m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4d54e8c0a3d14d709fd00923d1df9916 : SUCCESS in 14m 57s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8a6d4ae3836b4cd6ba2e8c51d70ecdf3 : SUCCESS in 11m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/888d9c78b6f94dba82d0c9057c2a9856 : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f67ac59fa607437697d4ad2b26efcb95 : SUCCESS in 5m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8cc90ddb635d4b38a7f90772d87fa3be : SUCCESS in 12m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5a03d41f97ea43a18706015dfbb01938 : SUCCESS in 5m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6918a35e39634b8ab0cb9c73dccf1c31 : SUCCESS in 5m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cc187ceb246049fa92ffb76c865622d2 : SUCCESS in 8m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fb0af60652c9467986ac4a186e310149 : SUCCESS in 12m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/946b03bd07974db785a559f5357ba3fc : SUCCESS in 8m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e8cb482f329047228d9537bf27e4c712 : SUCCESS in 7m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e8193f973b144fd1971a61658ebfe2d1 : SUCCESS in 12m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d86a3b4878b14d8ab4362b35b3cf5a9f : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7870d493d610410eaf0798eaeb136f4f : SUCCESS in 11m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/83219981413145c9876f5cd8d25932d4 : SUCCESS in 4m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/105fb4c404f54b5194b82b6f05315454 : SUCCESS in 10m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/798dcba971914c3a940c1542aeea66cb : SUCCESS in 24m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a0342c563b404f9b8f5129598c53f945 : SUCCESS in 12m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ca976afd3e364d0f8b4009708e6ba8b3 : SUCCESS in 16m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9ce7d9ab73c3414397dba925a62aa52f : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ca6b8694b414aa8a61160fd8e20511f : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a247e1501a84552bd17d2a41154e1f2 : SUCCESS in 23m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/51edb6cbde3c4a1e9245d7c9440c4382 : SUCCESS in 6m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4b6df2a10eeb46ac9f6ef101a8b737df : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a6f940cce2943a3a5fc69049de4945d : SUCCESS in 7m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8869a24a02604cc6b1f7cfd9732ec870 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5dcc5d7c55546c790664d2d7c243595 : SUCCESS in 55m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ba79e70874640ef8a23b4312af602d9 : SUCCESS in 1h 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f407939b3dea7b68b0719da550628add46a7c574","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-01 23:56:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b38afca8bd5806404c5ee9285c2f6281eac42fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 23:57:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"847c4b43a1085926128b2f2c25b1847179e849ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 01:40:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e0bc324224046c79626684360a369b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a0636dab8b4e56baa8026299016e65 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44013ad23304bc889dfa5627622933f : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7a4cfdc01cf4b97a3f7693b6a251954 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a87dcabc357747078d264344d42678bc : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4b45717046c4a8386eff9cfd446e033 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f1ae38d39824173a00fc781f3d67b13 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79b7be484ca742c8b86aaa9b656a4ce0 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4b223566a94e43f294b2fd1d3c611232 : SUCCESS in 10m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9b4b83e0130d447aac375b3062f87dff : SUCCESS in 14m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/10b208f74f6c41448ebe7b348fbb4e86 : SUCCESS in 5m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a8845f2f209a41a59b230fdc69229cf1 : SUCCESS in 9m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fde726295d7948f4864703c56f92e105 : FAILURE in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/02e41e685d7d431d9ca05addd893c6f8 : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a579adc1be614e30a73bd36e009b8c71 : SUCCESS in 7m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ea9a5c1c2c63438aa82053cddccaecb0 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f03f5e0b01d4c5d927497923afabbe1 : SUCCESS in 6m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/49eb3d51dd364c6b8beabad9e9e6a5cf : SUCCESS in 7m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f64db7f273d84250ba0e45b9fd69d5fd : SUCCESS in 10m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3ff761378a9f407cbcfd1f22f153a0db : SUCCESS in 3m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/816450fac31d4ee49f4e9f73aa754fd3 : SUCCESS in 9m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/16a4614f223240f48ad3d85ad3e59b96 : SUCCESS in 4m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8b83e2ffc7f647c4badfb1a531ba8cf4 : SUCCESS in 20m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/49f6c1862e2e4615a9687f228599a351 : SUCCESS in 8m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9f6798a53c0746ad8f3b1bce8b6fb093 : SUCCESS in 12m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6f2fffd6475b48d3b8ce172ba82ba77c : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a14425779f1d4fde8dbe3fc581eb3106 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a51c0d9ab2f84f2fbc8f9db8588f7f9e : SUCCESS in 25m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/844f5be9016f4bb2b115c72d6e3469ec : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c6df81f0c2f849f39e7103283764a59c : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/877afc35dd6c4cb697aaf5820b076a13 : SUCCESS in 7m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4f0d17d5003a4d66813455beb87f0781 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00b6df2a55e34fe191ff7ca5cdfa9980 : SUCCESS in 1h 02m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7d29784f9f3415cb25a0623ab10c3c8 : SUCCESS in 1h 35m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6ea9feeee1d43d16c844b3e83cac72a9f3687719","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e786c9e5e187e3ebc0073cb276df6040b61c9715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:38:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d9e47f1cdbc4be6b61816d6179d17fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d017fb86704d1b974ea52c78bdfaef : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6edcf8ee4fc4546a61cc7fe832e9307 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e79a78d810ac46c9a13106c664a41891 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f4d58ad2447433fa76155027cba85d7 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/235497706dfe411397179805ef629bbb : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a6fbaab481c466a862d10e56d00d46a : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd8321da0b9240afb539ad0c8605fd61 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dedd647f395341b4ae1e2c7201ee3172 : SUCCESS in 4m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd8c472611b943fea09f3d09177af3c0 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7d395e1e822f47aaae92bc6bb0b24a9c : SUCCESS in 15m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54e53e3d2b3145049beab1896284b1bd : SUCCESS in 8m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1316458983434350b8f0878cace18272 : SUCCESS in 11m 35s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7d2879840ed349b2917da32c68926c6b : FAILURE in 5m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fe16f5cc68ca4c5cb1a8dc4bbbb22c0a : SUCCESS in 6m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d076f1f87d9d484fa22fbd692113d5ea : SUCCESS in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/05d6f9b6b8664a5e81b509da6905d5e2 : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ffdfc1e5659b499391a71d3c53dc875d : SUCCESS in 5m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8d95ddcb071f4f48ac2804a77fdbfacb : SUCCESS in 7m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c97f74c4d67443e6a5eaf8aaf5346f41 : SUCCESS in 7m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/26bb0be8e4984fd29d460b7d8e52068d : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58dba617e7624a4b8c42a75e987ef72b : SUCCESS in 7m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/43c4b0cbc8ed4b2ab54996c07f9d168f : SUCCESS in 8m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fbdddecc353b4dbaa29d8e66dd846b2b : SUCCESS in 11m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5c827fea0876486e8016c6fd309cd214 : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/54716fb6ce754ee2855a7c696b520678 : SUCCESS in 11m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8acf04c84334449d93e057651740b17c : SUCCESS in 4m 39s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/28d92e5854e7422b845571ecf3e56ebc : SUCCESS in 10m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f4948319e4fd4e3fb4ed2786b689ed3a : SUCCESS in 25m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/151094f8f49843b2a0d9449cb33650be : SUCCESS in 11m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/62a51d5ae0164ceeaca2eeca902e9b68 : SUCCESS in 13m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b51cf9e5b24140d9b77da608859d7a77 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f1fe494fbfb742af8efbaa762d59116b : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d86b5112071b42a986ba09e7f9442f74 : SUCCESS in 25m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7aa3755645954efb9d7f3cb967acea77 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/724166480dea48b3bc9e7a809955c831 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/996f28ca2e6c47b796d6a92885f3bc28 : SUCCESS in 7m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6604a1e60f354e758735d65b0cc851c8 : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a49c9a54a8a414393233c766ecaaecb : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d65e99928654d0d8af1e98768995a46 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6a7f2dcee052ba125a2d8da0abb8c3f0fdbcab4e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:13.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2af2dd118b91e995f051fca931f11f63c30661d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 19:42:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20f6fdd129654498b6894f11acaf7337\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35743789fdda4c1ea49fa4f48f29e201 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d884fd6a8113409baa51a0c94df0e9c7 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbf146ebe55d48d088d08ed94aedfa08 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c725a9348d4843dbb06f9b72ddc46955 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f2a1046eefb43eab1cb46277da2d978 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20a4014e0e37440b9106591141c21e47 : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e4b9ae25cd14e0cbbb1d29239d505c7 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7105895d14d40f18a51fbadd80ecc5e : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2b0f160c3fd2465f9bae070f5ad34958 : SUCCESS in 8m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f367d176c2e8444a93fca38648c59f9b : SUCCESS in 12m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/30da81239a764bce8c81230c36513aba : SUCCESS in 9m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/69580388aeeb44a8bca8387375774935 : SUCCESS in 10m 57s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/cd55dab01ed34b3db0ad12fd87f73c27 : SUCCESS in 8m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/09d0550ff8c5482aa27e075f8e7ba3dc : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c21509ccd3ac4de5a4572233f0814675 : SUCCESS in 4m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aba1bac39e304c54975b7ba4f23c99e3 : SUCCESS in 10m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/955be355999b45a1bed5213ad7b7620a : FAILURE in 5m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0e7264cbccc7492d8ed873d0be4ef3bc : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/18864d8e884a4d29844e24a4a30ef27c : SUCCESS in 8m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a67709daeebb4c6598ff0445a200cb2b : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f93da872300d4f00ac0480574128a42f : SUCCESS in 8m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7a901238429548768466ea6a544c76f9 : SUCCESS in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fbe3b3258aec4ba190b724136570b8e2 : FAILURE in 12m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6d384fe755b643b7b31ba71ddb107fee : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3368898498544c1f9c0246a42eb26e4f : SUCCESS in 9m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4035fb15c27b462bafa5043061d0f8dd : SUCCESS in 4m 15s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d167efb891604d9fa0601b843595b559 : SUCCESS in 10m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3af2d709d9c44cb88ad7f882e5d41c49 : SUCCESS in 32m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/88570ef7afd34842b2e764c3d50afd48 : SUCCESS in 11m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/44af45b722664ce7986a6c8c23a7d2fe : SUCCESS in 13m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1a3990cae9364b549f5e6198f0ef62da : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c089b40951f84c419972e74cc7980710 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f57f8494c0e440f59e3b4e81b1f4d71a : SUCCESS in 18m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c755f42208b140028c3fe567ef6b054a : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e0fe8956b4c14aaebe57078eb0fbd157 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6dd44ad073a4429e9f07d1b68e7784f7 : SUCCESS in 8m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d793b9a9737c4ef2a80d1188198e09a2 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9a7a17862b8450d803d9f6e26830d44 : SUCCESS in 58m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05ef11470043468a972044d5109be2ae : SUCCESS in 1h 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ec9cfef9eda922400e05be0114dd8122ab01f2c6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-06 12:39:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f77ab7899d5c8981ea4aeae45fa5dacbee9fb40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:59:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc34ec932b754b4ba78bc96d6ed3af62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbda3f22d2c840199a2e1424ef29f3ce : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ad7beebf91406e9b77ba5c916569bb : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b77a9520d4fe4ff1a5a007d0542d537a : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7049a930da574590bac4b6157f96420b : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d82a5562e344ce9bcb2b9fc18a8f0a5 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/840386e6971f4b1699790b053931a55b : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79f67b9d04ec4eb08487df8822173571 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c92e10d4f5f94f3c881858dcf1fed67f : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9d886e8a0fe4a349a6b9c2fb3c2a429 : SUCCESS in 7m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7ed6f8db985c46a391075e95555e82f8 : SUCCESS in 15m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7b515abac92a4572b104bb499df9af7b : SUCCESS in 8m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/03fe9dac348e43f4a9e896b60531035e : SUCCESS in 11m 04s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/06112d5a2d534a35b724d389b8bea836 : SUCCESS in 6m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/895c3a38e2914cd9b310e43bb6b1ed16 : SUCCESS in 6m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/998f8a0d369a4f258401cae0311bc4be : SUCCESS in 4m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9b512f77f81b45cdafb26640496682dc : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/08238d2521c04d9ca4146852d5306cb0 : SUCCESS in 5m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/be8c9829c3f944c980eb712d7a791290 : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/75912ef926324da3be60ed9f180af519 : SUCCESS in 10m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/614c37f13b6c437fb8536cff96af7777 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ded4fe88b6984792b3713afe9bc8a5b8 : SUCCESS in 10m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0b747315282543b5b99ba32e132946ef : SUCCESS in 8m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/10aeba2c2d534fc09a9140f80c187ff5 : SUCCESS in 11m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c982ac4e28d949f19d1cd089f664df70 : SUCCESS in 3m 54s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2c2c69b76a5445099bce3bef9dc6f339 : SUCCESS in 11m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fe6dc49641494539ae50fd38ba4cfbba : SUCCESS in 4m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9c231dc3f7fe46ec837224d0b6ae8e71 : SUCCESS in 10m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d9e2765a7e674dc1b37b4789c176e5e9 : SUCCESS in 24m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2ed9b061b8d6424c9afeae4ced57df18 : SUCCESS in 12m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/45ff6b108e1148c993e7c52a393faf2c : SUCCESS in 14m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/78ab8016e4b44d64aa6bddffdc8ff165 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/11bfa3e216cb48c7bcde7997b956285e : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ba69234d2f3461c98e1106dba72142b : SUCCESS in 23m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b48b6fe5cef04bc2ac230915af2485b4 : SUCCESS in 6m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a453063ba3c245678300ab99c65ffeac : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4113a3596daa49b6b7e3558d7ce9b285 : SUCCESS in 5m 56s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/92d8e422500e4ebf8162dd4828a79ccb : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0f2bfb6bde148cf9a11dc9ee41acec0 : SUCCESS in 1h 01m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7054e330630c44ec8178eae1ff31a024 : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"51398a02775395f4c1c8eb3fd97d0e35646a49d9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 15:39:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"040dfa33a48568a00011722fa0d32b496c018623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:41:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3123ae201dff2312b0277f0ef658655df7560df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bf3611af6054cb3ad251b80aef2a9ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4e8d79c5d14a4f81782cc2ece2f54f : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/386017fddc764716967c0690a8f57334 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/521b0e15545e4160a4e1066e621d40a7 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6909dc7afa44f97a78b7b6fe5de1ea3 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0de8ec09609d44dba65d6c051b82aed1 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7e2ae6010c44b9fad1b053dce16a54c : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b8530fdaf81402ba71045de95fe303b : SUCCESS in 4m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fb76936f5c04f359f1e9c5d87dd849c : SUCCESS in 9m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c5ac073405524678b7bfd46227f2ebdf : SUCCESS in 19m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7ef525cff33a4ce3bd559cc2c6077087 : SUCCESS in 13m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/787cdf9bd755489aabad02676e75bb6f : SUCCESS in 13m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7c1085a930984bbdbb31a578b334d9af : SUCCESS in 6m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1b45b31a00794ec89e8f7426cc08f535 : SUCCESS in 6m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/161cf37d2491468ca81ebdf48f34c474 : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/44650f8bd4fe4e7db6b22584445b1849 : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/452d591af658432b8a7e4bd5f1b46578 : SUCCESS in 9m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/985bae4a2b3c48b1aa9dbaf0c00ffd55 : SUCCESS in 8m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/07f05a94d3cf4d439e7e8b61bef36b20 : SUCCESS in 15m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a71c136ed9d446b89fb03c0e46ab9ac6 : SUCCESS in 4m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/246be272089f461faab3e3fe0c8a3ff4 : SUCCESS in 11m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3efd46973405487785821dd9b7659c0d : SUCCESS in 5m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/95175a9192794d7faa5ab15c72a0dab1 : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/243f220bf8914dd0a172540a03154776 : SUCCESS in 15m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/016ba0d66aef467f87cb549e1b2b3663 : SUCCESS in 12m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/66dcd49a5e24467c83c1185f435ec94b : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a0e38a9824aa4468ad709749551d04c5 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52e8496364e74c0aba3a6f1a4889f8d3 : SUCCESS in 24m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6bf768727b4c489084374a1178dc3c16 : SUCCESS in 7m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/196bfc66f3224a66bf53d858e263682f : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ea99077b38c453489ceb5cbca3b5008 : SUCCESS in 7m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/08967c656783486698a42e1a40fd3d8d : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62b28337ef0b4c94b83fbeb9653ae2b1 : SUCCESS in 1h 06m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8f180b7776a419180c798f0cb7cf641 : SUCCESS in 1h 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c0d774414b84b048d859596749502a5bb4f1c052","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82e5f4a8bd80a5bedf9f2ba906394d8a29423d81","revisions":{"82e5f4a8bd80a5bedf9f2ba906394d8a29423d81":{"kind":"REWORK","_number":1,"created":"2023-09-01 14:47:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/893527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/893527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/893527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/893527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/893527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/893527/1"}}},"commit":{"parents":[{"commit":"08ce437e5875c695913e03331191eb080d60c7fa","subject":"Merge \"update constraint for metalsmith to new release 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08ce437e5875c695913e03331191eb080d60c7fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-01 14:47:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-01 14:47:03.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.4.0","message":"update constraint for python-ironicclient to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\nmeta: release:Change-Id: I49bdfe06e18dd0c8cc46c2456152e686f88ac025\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ibf14e8d1d663efbf167c950b8b7b966fe6e688eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82e5f4a8bd80a5bedf9f2ba906394d8a29423d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82e5f4a8bd80a5bedf9f2ba906394d8a29423d81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
