)]}'
{"id":"openstack%2Frequirements~893531","triplet_id":"openstack%2Frequirements~master~I4e2f997f763ed3251a7a4ac5fd1a70fa4bbf3d11","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4e2f997f763ed3251a7a4ac5fd1a70fa4bbf3d11","subject":"update constraint for python-manilaclient to new release 4.6.0","status":"MERGED","created":"2023-09-01 14:50:41.000000000","updated":"2023-09-02 04:42:44.000000000","submitted":"2023-09-02 04:41:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893531-new-release","meta_rev_id":"51e2c0dfa236d24ff8f4f818a268e29b3bd0f407","_number":893531,"virtual_id_number":893531,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-02 04:41:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-02 01:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-02 01:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 16:10:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-02 01:55:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5b80c520c59862ad936ba25e51f8c5f9f7bef579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-01 14:50:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa4fee44b8bee7168112c295da0647b7f0922144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:10:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e659ac6dc9e34e8c87a06b6452b7e1f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1caffee1f148a083724176601d592d : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a85c3c49aa394e90ba234848b374bba7 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12f0941419c24acd81636c990cb958d2 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/016a8afacb734d61bb516d264ac0b3d6 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/780dd631855d43438721f47520dddcee : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46523c4847bd4e7d8303313748fb6291 : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c60677b4f62e479f8b7738a096b44a40 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8023edee5f8442199d15a737a51d1b40 : SUCCESS in 4m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/784f8095f5314eee91ad0a66ca17f3bf : SUCCESS in 11m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e5c1a1916b1f48edba261c1ad4d28002 : SUCCESS in 15m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dbd381e7011a465d96941e99fcfa76d1 : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e49108fc11234cfb89e001e29c4cb758 : SUCCESS in 13m 12s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a33e0d2d6ecc499d92ddb7f7d492d5f6 : SUCCESS in 7m 13s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8b15ae0dd4874ef196ec9c55731038ae : SUCCESS in 4m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/10300df9133b47faaedf91e3337d0898 : SUCCESS in 6m 31s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/20e745aa524649f2b1ef773bbf087bed : SUCCESS in 12m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/59f0435e5cd64a1c945e64f77b7dcb36 : SUCCESS in 5m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/64e02a362ed04b9aaf240405459c6071 : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/710a59f154b1424785cbca10247627c2 : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f326afecd61946ae942cc5aeeb832a47 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43ea64536ce24b8e80d716f853f057d5 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5a12e2d7d4fd460ea1fe695a038a9ba0 : SUCCESS in 10m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b993362f28e0476a89a768f9c1c9dee9 : SUCCESS in 10m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c6f09deac1dd4af4bd33d2168fa61fa4 : SUCCESS in 4m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2859d95240464b86acf5499f631bb1b9 : SUCCESS in 9m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3711fafaa2544d0294a8be1525a15329 : SUCCESS in 5m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/eb6f4b4b781c4bb98a5a140bd2735231 : SUCCESS in 9m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/666575c244104a41bcd99b7941721922 : SUCCESS in 27m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1bdc89af63084c25af157ebf3109184a : SUCCESS in 13m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/28a12e94d9bb4b0abb2543378cc8bddc : SUCCESS in 16m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d3f7b739b7cc42b68e59663c7ccde87b : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f38a6398ab5a461c9b6ef2d64e16993c : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eacdb76978e840b0ae34acb272c81375 : SUCCESS in 24m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f1ca3b26f7f34ad5bdeaa50e5a53e6e3 : SUCCESS in 6m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c35e3ddc244e438095ff06a29f4d1646 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b73f4cfadf9742f5bb2d0c572310b6b0 : SUCCESS in 7m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/db00270c22034bd6a44bdd1db6520010 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1047e24727ec474e82415176cd9c7f7d : SUCCESS in 58m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02fba5b730534095acc0036bdbdfc41f : FAILURE in 1h 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"23c22205e84d95bcf9d5200974c0985b5f0433b3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7596abeb76f40cbb5c863c6f2810bc5d94dc2bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:23:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4fa34f3d55c4c53bd3afb482337cabd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d527881a4dd435789a15366686703b2 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06a8b16e7be4b5797b2461a639bb1e1 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c921b8fe1ce4eb3856e0fc07e59086b : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcba8bef1c0346aba838f2dc77191b15 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efdc4ebf459d4232a8963dba5f991c13 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6340a341023a4c97b9166518c87f4615 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ccedd4162fcb4b1e9e870b87e2a60cd1 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9796f187239645649ec2645f5ee334e8 : SUCCESS in 4m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a88a3a7d7533410e930891c068322c5b : SUCCESS in 14m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5015fbaec3464ccab45cb1457eb3dbda : SUCCESS in 12m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d637570c0c2d49e6885eee9484dee8b0 : SUCCESS in 8m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e05207e0bfc346368c70f7c90ba1d75c : SUCCESS in 13m 08s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/95051462a42a42fe96537d92fc9f0bc4 : SUCCESS in 6m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bc790bf6132f46ac95d672b8fb89ea35 : SUCCESS in 5m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fac8a71809ae46608f79fde87ead009c : SUCCESS in 5m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a4a5690dca524109a35cb0f437b9f41c : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/587ac84a6c924829a0ef457ba30b22c0 : SUCCESS in 6m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/88b209fddb894c6cb225273d972a3e86 : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7812c00c64c6489ca646ac04b9de37c4 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/dfe2f595db97490daaacf25d2774d87d : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4fb09cefc09e4d2981f13a7b8fde6dfb : SUCCESS in 6m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/14f0e793782d4746a5dc4af0e8abfd48 : SUCCESS in 10m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7b1ac913c4ef4a1088cd905b25d7d8eb : SUCCESS in 13m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/80b703d8dfd44fcd834f5cb37d55594b : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/91744db977d84bf7a8aafc7f6577b6f2 : SUCCESS in 11m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b66e0c8e48714b698a7dee9af6a113f4 : SUCCESS in 7m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a5f6c1b1d4a04ca6af537112eb008b2b : SUCCESS in 13m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4bf218e6646243258a701c493c4f76ba : SUCCESS in 23m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8c1caf32f404475999a9624bdba018e0 : SUCCESS in 12m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/78f2b29756244e149a4a4b53614c9a28 : SUCCESS in 13m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/84663bc04300431a9f76fe348ea27d31 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c84a522791546858081862a6e2c920e : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b90c2c08a1446e1b5538b32b981ad84 : SUCCESS in 25m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/587b65552d0d46fb9e227288180f3576 : SUCCESS in 4m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/368fe9fcc5a74d779616beaeb0eb9112 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23489ecdf26d481e99e475548131761c : SUCCESS in 6m 16s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/96363e8c62a44ffca1869afa599148d0 : SUCCESS in 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6e5fb799145448bb264e0d9e95b1ea0 : SUCCESS in 1h 04m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c276eee5dd164ef1b22d57fd346a056f : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"66c4d14c7622c01bf53ac60355fd2b5363bb0aa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:24:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f514ba6fcd32fb593c0f63ded968c0f8ac688475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 04:41:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3276dcf304f14b09adc686ac7760e1ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1e6fa7c42c44fba92341c7efec1fe6 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9d257ac4d84992acd3262bafe27f10 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a518c0cff1740a49335ac2a01da08ec : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2e19a4d71024ab8837b8107dc4c8d0d : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/991fd99554f445548b15e0e60eb45f3e : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2e9945886584d079c77b31dc7092273 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5f6d8171a444f999fd5bf67841393de : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d71c408093a4614a8aae9104b0ef4bb : SUCCESS in 10m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c1c35ac385994d82957f5ffb0c616bb9 : SUCCESS in 14m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4ee7b0be3e614c368ed783e82ee517b4 : SUCCESS in 11m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/406c237da612423fbeca5f19b63cb454 : SUCCESS in 10m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ab159a3d88a145eb83a43a5bfa790d53 : SUCCESS in 5m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/de897aa255ea4efdb96ebc08911159ae : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/70e5dcbf2ade4ae8b89e80d5b0b5e85f : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/48cae6d4208b4fcaab8e96711801e07e : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22b2e641ce694566bbb65d49dfaf7898 : SUCCESS in 7m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9011e93cc63641b597f35782691b3587 : SUCCESS in 7m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b89e392341014972adb185300f631ad3 : SUCCESS in 12m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f370521fb67343b0825c3ee53ecd29c0 : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4d9d0e40319c4651ba701f5c5f28f2ed : SUCCESS in 10m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/595be23c6d40493383842bbb6f0a118b : SUCCESS in 4m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/32ce5450db0c4726b5f0641461374d31 : SUCCESS in 27m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b762a861d4644b5acbb84955d127772 : SUCCESS in 9m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/feecc179de10412c9018c2a26d5eeffa : SUCCESS in 10m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6e29ecc5181540b593346355a46172a6 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f00704628714f0ab52ee09ef5a4ad79 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e158f66049604948b83eb0f9bbcb05bc : SUCCESS in 18m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3dff06639bef4045b2d2ac603c4f8a7a : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/60301fc84e904dbdb14554ac6916f718 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ebc2007a0ef4d8396b0ac2feac0da90 : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5f132bccab4744b9aa14292ebe0f5ca3 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c758024b5794eaf836efb59eb70fd6a : SUCCESS in 42m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d01dcad5eb94907844bc7c7b5461bcc : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"09a1f0aea3cd4b44f691e6eefd0063e53f28145f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 04:41:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"51e2c0dfa236d24ff8f4f818a268e29b3bd0f407","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 04:42:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ace5ff884a04000b8c050ca902de410\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ee3e5738fd4587bd3852b6ea87d16f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c639d5fc4c8dce4f0fba130bd9603f0fb7fff12","revisions":{"8c639d5fc4c8dce4f0fba130bd9603f0fb7fff12":{"kind":"REWORK","_number":1,"created":"2023-09-01 14:50:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/893531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/893531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/893531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/893531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/893531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/893531/1"}}},"commit":{"parents":[{"commit":"08ce437e5875c695913e03331191eb080d60c7fa","subject":"Merge \"update constraint for metalsmith to new release 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08ce437e5875c695913e03331191eb080d60c7fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-01 14:50:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-01 14:50:39.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.6.0","message":"update constraint for python-manilaclient to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I67e146bd5a72dc529fb19297b592a636807e812f\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4e2f997f763ed3251a7a4ac5fd1a70fa4bbf3d11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c639d5fc4c8dce4f0fba130bd9603f0fb7fff12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c639d5fc4c8dce4f0fba130bd9603f0fb7fff12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
