)]}'
{"id":"openstack%2Frequirements~893665","triplet_id":"openstack%2Frequirements~master~I9697bf8722f7fe24d928f627c6080f48aacf97a0","project":"openstack/requirements","branch":"master","topic":"remove-sphinx-rtd-theme","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-06 02:44:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9697bf8722f7fe24d928f627c6080f48aacf97a0","subject":"Revert \"Add Spinx RTD theme\"","status":"MERGED","created":"2023-09-04 14:59:46.000000000","updated":"2023-09-06 02:45:51.000000000","submitted":"2023-09-06 02:44:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893665-remove-sphinx-rtd-theme","meta_rev_id":"4de906c0be9126d3bd1544d1227865144fc97a93","_number":893665,"virtual_id_number":893665,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 02:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-09-05 20:47:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-09-06 01:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-09-06 01:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 16:35:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-05 20:47:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-06 01:33:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"beb8cb9f864c47dd888c886fda8f3dbd74c7ccc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-04 14:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c54c5888a4c5d826582591e808b2e1ecf8beb5cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 16:35:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c114e913dec2457eadf3020e821ea4cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce55664ada3d4946ab237bc430adc091 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f305430df7514902a0e6c9fe7febd626 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21e1e1e307c24fb08ea4318dbe27ff78 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/737f0c88283b4854922e0b733f27b600 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08702c50463f42a8a949a77aef398094 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26ef4a4aee2f40ecb53a3c26255f46bf : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/384cc74191d9467997bab19e0021289b : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f6a472decbb493fafa92505df476341 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee0141a08e984364a480d4c8ed61b75f : SUCCESS in 8m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2b4d23aa44914a81a0e1c14c42c96d43 : SUCCESS in 11m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/068cdaa073d54b30a3df821544d454e7 : SUCCESS in 8m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/606e764e20994fdcb2706aebfd81017f : SUCCESS in 9m 43s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/16cb2ae8db8b41b2bbb873b89e238ed4 : SUCCESS in 6m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cef837e622fc40618833ff9ce8e59dda : SUCCESS in 4m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/72deccc991b14ddbba350b8eace3350b : SUCCESS in 5m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/53b22da88dea4b8ca7eddf662afb6e6c : SUCCESS in 10m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/149521cf739f47bda106c9acec294d7e : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/24ac2a9f795b49f784df7bdd61e03a00 : SUCCESS in 4m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f595fc22c7ce4da4aa233025564a3714 : SUCCESS in 6m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b41b4a63b5b14f9a98f4920b71f25baa : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bbd925d79f94795a772ddef8be40f4b : SUCCESS in 6m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/afc05f3a158d459d9d003f752eecacbb : SUCCESS in 8m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b8976a9128b24dc09c5bc9215753df4f : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e16153e410564a48aa78d551e0e6d839 : SUCCESS in 3m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f0334ed4dc244ce8a9640f2e7e1fd284 : SUCCESS in 10m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cf5868575f7246f7b34cb031c0215c48 : SUCCESS in 4m 55s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0de62e55cf8545dfaf7dc2d2f1cc6413 : SUCCESS in 11m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/01d849ff400c4b8b9a19d116d4f8b4fe : SUCCESS in 25m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac5ad2f0defc40fa940d8cc9e6a132db : SUCCESS in 13m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c1053af96a354aee91c480963ec0064d : SUCCESS in 13m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/452f2cebc9ed4398b398bccf371172da : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d24d0452cd1041b38bcc66fd55f003f5 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92faf4a0d56847fc8e7cf75d068ff268 : SUCCESS in 24m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/778a8fa7b00848928de56d8ed4b78022 : SUCCESS in 5m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1f71a8eda42a4789812f1bb26b9dc3a5 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0090c76865054a2e8890057ec4855005 : SUCCESS in 6m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2d5550065be1448d8452efba29ac678f : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af1f7d37a9e846c5a6b87250383a1f64 : SUCCESS in 46m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ba12d9d1d0b4fa192946afb48baa86c : SUCCESS in 1h 33m 58s","accounts_in_message":[],"_revision_number":1},{"id":"52fe22091fe886e066fcb2ea5305d3e677f23c4e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-05 20:47:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c1d9a322cf2ec237964a082cd490be2ebbacc69","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-06 01:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79adeec6f974313b85a9e74a2fc00291393a13ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 01:34:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b52cee565428ed59e1e31eb58cb32518ec3c13fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 02:44:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/787ca61e4d154a5bbdced337ffaabe39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6136862235134067ac46d18e61ad3615 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae2a1a1bc324310966011f856503c31 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe4a5eda1464702b7008e19951de000 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4af3ca4cd8924a7a9ef94a6044ed14c3 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c95fd1e7414c4990a605e25a9d237b9c : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6da24d98287245128ccbc357ea6879e2 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c97d86859555483e8a08bbe8ea567a09 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0e620050f8524ebda5744ee35c4c86ee : SUCCESS in 7m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d65c1c26090643a6b56669b50a5dba5a : SUCCESS in 10m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2dbbaf8f5c0f42c1b31f6192aedb0d57 : SUCCESS in 7m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2bad6d0abc2d4b5bbca3057d1cd996f1 : SUCCESS in 12m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8d702c9b26e94532a8575153d5b6d123 : SUCCESS in 4m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/27bf95dd9cde4ac4aff4758d45cdead2 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c427a1f57f174511bd9236244218ac65 : SUCCESS in 6m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6abbc0fb65b64c3f82a6e02ad390fbcd : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1142433827a3410d8e6afe46c8dcf4a0 : SUCCESS in 7m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/eb2e4f08e2bb4646b002756698679b21 : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c01765c504054e8db947fda9bbef681a : SUCCESS in 11m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f88979f7bd7f4a279e74d1b00ab3b21d : SUCCESS in 3m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6343d410fc1d4261a861c1d75526c9e5 : SUCCESS in 10m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b59347f9e0bf40e485323c591d5217e7 : SUCCESS in 4m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6f1ef4f9e5a84a7ea7de7dd3b4ef64e8 : SUCCESS in 24m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e98044087a7746ab9f3e486386fea188 : SUCCESS in 11m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/723d116f371646ee8204ce4da2c9fbe9 : SUCCESS in 13m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/785503a29569470a8237ef7ef056c661 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6850d8aff3749e7b524840fd9cbbbf1 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c83435e4ff2141f282123c001ad31353 : SUCCESS in 19m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b5f076d932544542be646b8419e292cb : SUCCESS in 4m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/272846ea8d1f43ab8c1cd92508337597 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/426b4b9885004a028f7c0e6f5204b42f : SUCCESS in 5m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0f41240bc8fc45a1bd4ba416df36ea41 : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0af57b8576b84291bd2bf0690aef8d69 : SUCCESS in 43m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c7051070dfe43ea877a55f345530b19 : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bdebe7dd1c267c8d8757101aba5a7340d0677ec0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 02:44:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4de906c0be9126d3bd1544d1227865144fc97a93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 02:45:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f32f68c0dcf94c3caef3c50996ffcf89\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b07f08699f1c4527b28d1706b1b9d0eb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"237d49d5ba63b817eece6376a284e990654f4ea5","revisions":{"237d49d5ba63b817eece6376a284e990654f4ea5":{"kind":"REWORK","_number":1,"created":"2023-09-04 14:59:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/893665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/893665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/893665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/893665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/893665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/893665/1"}}},"commit":{"parents":[{"commit":"9463129a82a49fec61ad2ba536ac904cf50fa48d","subject":"Merge \"update constraint for oslo.middleware to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9463129a82a49fec61ad2ba536ac904cf50fa48d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-04 14:57:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-04 14:59:35.000000000","tz":60},"subject":"Revert \"Add Spinx RTD theme\"","message":"Revert \"Add Spinx RTD theme\"\n\nThis reverts commit 45eeb10dd45d282109cd8f34e251a0449c3534ab. OpenStack\nprojects should be using openstackdocstheme. In addition, the current\nrelease is not compatible with Sphinx 7.x.\n\nChange-Id: I9697bf8722f7fe24d928f627c6080f48aacf97a0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
