)]}'
{"id":"openstack%2Frequirements~893864","triplet_id":"openstack%2Frequirements~master~I3596993880af2ad61274af30503f639199f76b9d","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-08 17:09:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3596993880af2ad61274af30503f639199f76b9d","subject":"Update OVS to the latest version of 2.17","status":"ABANDONED","created":"2023-09-06 11:38:04.000000000","updated":"2023-11-08 17:09:12.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5ea5a23f42e4f464822c8973cd48af201f72917c","_number":893864,"virtual_id_number":893864,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-06 13:22:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2023-09-06 12:21:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 12:21:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-06 13:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9097e1b2cf91aa360685979629d0c4acc15a25c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-06 11:38:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"213d8971a5e471165feece20e521d9106e011d37","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-06 12:21:29.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4ecc86e404a90c8f04ef4547cf798e5e9279202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 13:22:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddbe0922cadc45ea98c6dc4ebb4774de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4b4c1407aa41f9959aae5902defe4b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0bcdaded785449aa9552821804bd7aa : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd0e87dc1d104fa58307902341c518d5 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14db53b66d4f422f83643734db7157b8 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ed2a011e78e4fd99700eb15e09feda8 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6410a3e913dd49f08a481039e55fcb45 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a452eac2408f467392c578139763ab6b : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd541869a4bd480d9a495067fb5a996b : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f88d7518a55e41339c694e36d991d56e : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8617c4333cfa4833972768daef5165e7 : SUCCESS in 10m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/83f364df437542edac87fcdc9166d186 : SUCCESS in 10m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84bd812d3033400e967d2ee4d02b9a09 : SUCCESS in 11m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8fbc27bbdd974f98b95b31ffc4a8e43d : SUCCESS in 7m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d22bb0961af64fe2ab850baaded88df9 : SUCCESS in 5m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/802d3cde7f604f90a5ddce2602610f73 : SUCCESS in 5m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/29b04ce8c4c14c9eb95b48e527d958e5 : SUCCESS in 13m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aa5f3c59c3fe45de863f4a92b064e24f : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0b8e256bab9b45129fac0e142d9ef2eb : SUCCESS in 6m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a3b4e5c839624a8dbcb33508372024c8 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/144a42421a7f46828e70515762d025fd : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71461adeb0034f089b1086d9176bc9a5 : SUCCESS in 6m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bceb97e6c7f54221a0e327acf00e37eb : SUCCESS in 8m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/583036346bfd4fc0bd93b6a2e452992f : SUCCESS in 10m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d8fbfa614dbf456bb7dd6783c5c291af : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b3947e9b63d6425ca4246b52861dd2df : SUCCESS in 15m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2617f897a04b4dfc9a07d9b861c7ad1f : SUCCESS in 4m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1052bb25fd454b3fb1026d2c2db395d4 : SUCCESS in 10m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/070a7da8315f46fcb6e30dddeb18b7bb : SUCCESS in 21m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/50655093ec16425a954a5c74f01274b5 : SUCCESS in 12m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/66107008b1514385928612bb553c53c1 : SUCCESS in 15m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ef8119584c034664bdbe62fa93dbecb7 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a1365fcdbbe45a2845f668ea63cfbc9 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4142e35faa734ddcbb361d8f227bcfba : SUCCESS in 20m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/50855d15a9e841598d7fcd94ed9a8a49 : SUCCESS in 4m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/19dca94e4662473a922e82f21098c453 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88b0972da098457cbb4dfd362f7ff8df : SUCCESS in 6m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/90578439dd174415b6d384148348e7f1 : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b48d82f69e154826ae39232ed0b3abe5 : SUCCESS in 1h 03m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/525eb74fd2b441718e07c71310f253c6 : SUCCESS in 1h 36m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5ea5a23f42e4f464822c8973cd48af201f72917c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:09:12.000000000","message":"Abandoned\n\nNo longer needed upper-constraints currently has 3.1.2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b89fe3edbb4d583511c81488bce09ec944e48e4f","revisions":{"b89fe3edbb4d583511c81488bce09ec944e48e4f":{"kind":"REWORK","_number":1,"created":"2023-09-06 11:38:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/64/893864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/893864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/893864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/893864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/893864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/893864/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-09-06 11:37:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-09-06 11:37:55.000000000","tz":120},"subject":"Update OVS to the latest version of 2.17","message":"Update OVS to the latest version of 2.17\n\nLatest 2.17 OVS python binding contain various bugfixes and\nvaluable improvements. It\u0027s worth to bump version of OVS to the latest\nminor release in series.\n\nChange-Id: I3596993880af2ad61274af30503f639199f76b9d\nRelated-Bug: #2027854\nRelated-Bug: #1985062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b89fe3edbb4d583511c81488bce09ec944e48e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b89fe3edbb4d583511c81488bce09ec944e48e4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
