)]}'
{"id":"openstack%2Frequirements~894156","triplet_id":"openstack%2Frequirements~master~Ibc99040d5ecc7da31721ccb22705a65cf30ecf23","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibc99040d5ecc7da31721ccb22705a65cf30ecf23","subject":"update constraint for mistral-lib to new release 2.9.0","status":"MERGED","created":"2023-09-07 10:05:07.000000000","updated":"2023-09-07 18:28:14.000000000","submitted":"2023-09-07 18:27:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894156-new-release","meta_rev_id":"02d31b4c7dc38e432cb936a00f195d93c5475e51","_number":894156,"virtual_id_number":894156,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 18:27:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-07 15:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-07 13:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-07 15:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 11:41:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 13:23:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-07 15:39:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8618126d5c163e82231e8108da8b9678d7ad6171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-07 10:05:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed7c473b6dc70e538407d516affd8435d3e76179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:41:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cdd35e08bf24d73b6168df432d360e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71af6aed4324c6391bc198fd298fff5 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe33f671a0d44fcea73ce3eba96d4614 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d82282973af24f26a6c93f93f9b80a5a : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3376805a34ea4a8fa132518689200bec : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6284f92236f34824909340d43b2ab255 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/867157d55e57491f80a44ecd063f7efe : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3aea0e9c5b2e40b78dd2a596a0c57307 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1bc69ea0a41447e931464191729b5bd : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/47e6242a36cf4e59b6a76a9b9d0979b1 : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f6aaf37f17c3483381044543e27b9e33 : SUCCESS in 15m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7a240ba9106f4eb692c7141b8582a36a : SUCCESS in 8m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/63f28f7b4a0349db9cf7a08a9d7c6e80 : SUCCESS in 11m 55s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/12e97bb266424bd8b5893e4bb37cb906 : SUCCESS in 6m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e032d77563dc4d2e9fb39c343504349b : SUCCESS in 4m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f785ba5d8f97449693b31e2fceec5107 : SUCCESS in 4m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c60f3a9a49854caa886884609f5f3d63 : SUCCESS in 13m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f93c6f5533e04943a81303f82a7c9d96 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2b8da746988642299a1f3bbb3755fe04 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/aeaf138c56f74af9846507eebf97c8b3 : SUCCESS in 7m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/54f0030e5d7044e08d6b78722397753c : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/839738dfc0fe497dbf57422b35ee007f : SUCCESS in 7m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fc0099bc7ade4774935f4d0b7b97e9a9 : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e0e46233826d4dfcbc08fac93cb3439e : SUCCESS in 10m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/22d192cfd0234c9c85d5cd0aa50ba81d : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/efc9af4cbff545b8b1baded9343a168b : SUCCESS in 12m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8b83d0239b954740af4a16b6ed8a935c : SUCCESS in 4m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6206a2919abd4ab8a109f4e06dfec7fb : SUCCESS in 10m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c4a8e0c22e8e49abb4515e658b879569 : SUCCESS in 22m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b3897e7ebf74273878f1cdc2531ff4a : SUCCESS in 10m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/efe721253db24f9f9975897017f9be14 : SUCCESS in 10m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f425e59f50b042e49d2b1b3aaefe3205 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/befd68d8c84e4e1a9bd5ae09753786dd : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57b65e3658844fd9aa7e4f6b04a86ad0 : SUCCESS in 23m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0e3dfd4d333b4386b6ce848363714128 : SUCCESS in 5m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cbd5b2ad39bc470199aa9363b554fcd1 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/184ae09fe7784c5daf74d4917eb5ec59 : SUCCESS in 7m 08s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ff4fb7e7068548539999f157d6b182d5 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d5323db49a446b583a13203f0e2236d : SUCCESS in 49m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22f5cc85d5d849b4955d6d09325ad119 : SUCCESS in 1h 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ede6db6d719966f9712f2cce4ac2537b838a0aa3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-07 13:23:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19f6a929f230735a793b9e37d3929b434a9a2dae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 15:39:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02294f3f6926ad1096f067b71e683cf63a356048","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:40:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"484711833d5a8c691ac6bec6288b79414b29de01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:27:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b01575e723db43cea85fea1f2840d5f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f860ab311ef44c50aac0517bb4120add : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e1b5f057ee4d5c8e0bdaac3db198c6 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6430e529dab4d76b0004503ab244381 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df1e196c43f048f08f3c9f5eb2ce1c2d : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/601505f73eec4cdd937efb2890cde4cc : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9ca2d5be4db436e857ddab8c8f082cf : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dcccf6bc6d07489aa1eec0eab63eae3e : SUCCESS in 4m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1d968f4082f643c189b149b2b104f8d6 : SUCCESS in 14m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/170ea2e12fe34ac6b5926a3275337a5d : SUCCESS in 16m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f8a7334a00a484fa2e512793d561648 : SUCCESS in 10m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a838812e1675419ab70eab9f258019a6 : SUCCESS in 12m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4344099e52744deeb737de7117e1bc64 : SUCCESS in 5m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ffa51a9615204a5a860cc5a5247223e5 : SUCCESS in 8m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dbdae790230548c1845a2fb9e2c8b191 : SUCCESS in 8m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7ad2a7b8f82844b9b945fdebe4a55bf4 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/545a06a0e16a4f81859f93b64b8edcc6 : SUCCESS in 9m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f98cec751a634499b2c7cbfdacf24193 : SUCCESS in 10m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dead3abc5bb14f57ae286863b2bb1ed8 : SUCCESS in 17m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1af4124dff8d4f8394c1c70fc024234e : SUCCESS in 5m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/73b52ad7e3e84b6abab560ed4ffece00 : SUCCESS in 15m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/50542d65e5f346668b2d27ee5ead46c4 : SUCCESS in 5m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/418c8b917fba4c15bad6ae5620f417d8 : SUCCESS in 31m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1d997c899e274511a857cd778de98175 : SUCCESS in 14m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6cfd7f4f49df4d4aad1195640d612b81 : SUCCESS in 14m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/05a8af39f3ba4c77afc680f5ff66b647 : SUCCESS in 6m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c69e2093efc4bfd822f24d9727623ad : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08d3ad9ccaea405d9bf9b552c9beec4d : SUCCESS in 25m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4d91cf41f0d841509710938eec5d6f04 : SUCCESS in 6m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e1642fd05d4d457e8f7490cb9f6a2b33 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/192040a7a203477c800c0728eb9d70f1 : SUCCESS in 6m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/be7669118cb44236a022ebed799fbab6 : SUCCESS in 16m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/253cbf6e20fa469d93a8ea06ab53b32c : SUCCESS in 1h 00m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4811c2aa7e464197bb545221a8b9e715 : SUCCESS in 1h 41m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f5263b8cca954b7cde3b36768f4ce3c2de7b7fd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:27:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"02d31b4c7dc38e432cb936a00f195d93c5475e51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:28:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc2efd653de4a80887d16dbaf951228\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5feae6e5cfb04a58a1e3f3e157da4ec4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d2ab8959fa1003f12e2b071445185c58f3792f1","revisions":{"3d2ab8959fa1003f12e2b071445185c58f3792f1":{"kind":"REWORK","_number":1,"created":"2023-09-07 10:05:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/894156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/894156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/894156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/894156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/894156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/894156/1"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:05:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:05:04.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 2.9.0","message":"update constraint for mistral-lib to new release 2.9.0\n\nmeta: version: 2.9.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7d4a6dbf0964ac9d85ce62a4208a38ee68f65e31\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ibc99040d5ecc7da31721ccb22705a65cf30ecf23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d2ab8959fa1003f12e2b071445185c58f3792f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d2ab8959fa1003f12e2b071445185c58f3792f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
