)]}'
{"id":"openstack%2Frequirements~894157","triplet_id":"openstack%2Frequirements~master~I0cf39d390954f52b91eac7503f4f76c99498ec04","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0cf39d390954f52b91eac7503f4f76c99498ec04","subject":"update constraint for monasca-statsd to new release 2.7.0","status":"MERGED","created":"2023-09-07 10:06:46.000000000","updated":"2023-09-07 18:44:13.000000000","submitted":"2023-09-07 18:43:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894157-new-release","meta_rev_id":"11dffc10f39d6c6296cf74b6124ade8564ee7060","_number":894157,"virtual_id_number":894157,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 18:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-07 15:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-07 13:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-07 15:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 11:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 13:23:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-07 15:39:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2b303c23a8c91378879ce361b28ce09a79e65f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-07 10:06:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5efabdb7b7c4b68bbe6960430fe65160e38dddf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:41:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1761cd6c191e4c62a0682b6ccfb11c61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9106fa24d40e4a9587929e8b1d48639f : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62323b3f46f640f0879d2f05864a899b : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/672cdb112290429caaceba2774a1ae25 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49f1aa9ded5f4e5a988c64c93e52429e : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8617d18ff0a4f5a81989a6fcf564233 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78fe039d406341f687f7dd406753e0bd : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfb10677a20241709a04fb856a75e266 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/499a254304c0489db40df327d879daf7 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f7e4db2f2654176b455dd2bd2cf2549 : SUCCESS in 13m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9f3f337f30e2457b82f776610ef20bf1 : SUCCESS in 12m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fe8d0aaa98a4410aa987323725fad433 : SUCCESS in 9m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b0aa1ae3beef4d49ab0b97b402e81bc7 : SUCCESS in 9m 15s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5037ec5984d44555962bf412db9ee7d3 : SUCCESS in 7m 23s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/abb4146a11a24424b725be051700aa70 : SUCCESS in 4m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a6b27ce7b0744e28956ad85e365ff8be : SUCCESS in 5m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/eb4e11638acd446d99ff5a44f61c61f3 : SUCCESS in 12m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7700cb70387844f8847b2f7046b732d5 : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/24d2449427094cdc8545e409fbecf3eb : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6b1cfb737dd54d4982b77a814a18f7d3 : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9f3726dfda3d4e54bee7a5c675c5a2f4 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7854a627af344a26b62ff0abf8c85238 : SUCCESS in 7m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f0ff4989f3034536904e279aed4e8ede : SUCCESS in 8m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f6e24d96305d47589fed13e09e0051d0 : SUCCESS in 12m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/025bf943659c47e3a4b1d019863497d5 : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/63323585e94b4a179e064cad185f9bb2 : SUCCESS in 10m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fb2c925d68a04ae18ab5fc602af85511 : SUCCESS in 4m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/85f73cbebe2f4818b0a23d7f983bd5b5 : SUCCESS in 11m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/aa92ca27735c4077a659844b88a19ff1 : SUCCESS in 25m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3af2c2a11d8c42148af8922a5f383c00 : SUCCESS in 11m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bdf0d8351ab0452ba742a7443fbe3ac9 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dc8c2d6bd8c24b8f82661d6848625471 : SUCCESS in 6m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfc3ed00f964453393e699d8c1108b47 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ed39bb4d7294e39a47db68298324347 : SUCCESS in 21m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3834ab71715c42dd8f19da9284d956c4 : SUCCESS in 5m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6b1c683473a94274bbe5a0cdd5e0ad2c : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/840bad5febcb475aacccc157ea2aa9e2 : SUCCESS in 6m 19s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/760371a2d7d6490ba00390f183c4c6c9 : SUCCESS in 9m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa01c1b36cde4517a2afff323ff4604c : SUCCESS in 49m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3ccc57dfdde40cd9a783da8b7476134 : SUCCESS in 1h 28m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a376de3e04d4124870f0abcd1c5bd4587bf9034e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-07 13:23:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c7d919a40f0e5cd87b560b9db8bdcc0bd33a8e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 15:39:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0972507edd82a5f16062094b659641ba36f21034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:40:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3dd3523dfde7707abf643efb96019956b109f243","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1487a989bb0f45668551642ccb85dc3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371a45a12995402b956c606735e4c622 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c117c705612142e9af9cc1520884299c : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d9ef35558a7428fa5e68cef05b27675 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a328a40fb454ecabd1e89ce078836cc : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f475e40609e64869b41610f949a89c29 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c96f100aad24a4988a4ef5f0035be21 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01e4a9b351454653be213a7aebd24735 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8dc97abdd96b4915ad19e2dce7103253 : SUCCESS in 8m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/02456c81cca64d6fa405c8d248b6c384 : SUCCESS in 12m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6c40898969bf465eb495666a99d8ce99 : SUCCESS in 9m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/af4a199cf1d1459b8f47351b56a9f433 : SUCCESS in 10m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ff64cdc7e7294e43a8012c3361afce5b : SUCCESS in 5m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9235a35593584586b42fc12fb72a5f5d : SUCCESS in 5m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fa9a370599c449c08c3a949584de8ee6 : SUCCESS in 8m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f73f562c23f243dcb28e2303e0cf10bb : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2124b25c56de40139fd5af0ef273db45 : SUCCESS in 8m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9fd16b0c325648d0b35bc937a5e69a1e : SUCCESS in 9m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8ffd10bab9434537aeee7231a7a89e83 : SUCCESS in 12m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/05943efea38f4d65ba8799c7e5aebe81 : SUCCESS in 4m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/65644a49aef848c28f717ab183db46b8 : SUCCESS in 13m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/290b55139e244c25a14656a6699257e3 : SUCCESS in 7m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9fda2ddba6d24435a44fb3d8fbc5b2ce : SUCCESS in 32m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9345739f57fb435d9d529cf4e352a280 : SUCCESS in 11m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/74f793a1337a4f9185a285d7048d30db : SUCCESS in 13m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5eba9d8ecfa446a69074e177748a2297 : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d652401464864e4d8a87d815623c3577 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04a1d193396843008bea44ba96b0a125 : SUCCESS in 21m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ab6be323f8624d97a8001acc0aed7410 : SUCCESS in 6m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/da157a2b45ba42c699ea7e56b40e6e6a : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/726dbbfa86f4473b8cb3d391958f2e9e : SUCCESS in 9m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/85abdd4384e04359baf6e076f1cc12b5 : SUCCESS in 16m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41880f92f41141e2aeb5be617ee11388 : SUCCESS in 56m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09b0c5671d634279b77e4b1f420d005b : SUCCESS in 1h 51m 31s","accounts_in_message":[],"_revision_number":1},{"id":"71ef7058819bb6430581b7945c56604fea0ed490","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"11dffc10f39d6c6296cf74b6124ade8564ee7060","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:44:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bdac982cc224f3ea78769840db23289\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5088488999204f4b9454846d9a3a9e5c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ded27428303d5b9c095bdfe3df2ce4e4cb0730d1","revisions":{"ded27428303d5b9c095bdfe3df2ce4e4cb0730d1":{"kind":"REWORK","_number":1,"created":"2023-09-07 10:06:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/894157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/894157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/894157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/894157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/894157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/894157/1"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:06:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:06:45.000000000","tz":0},"subject":"update constraint for monasca-statsd to new release 2.7.0","message":"update constraint for monasca-statsd to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I935303f159e165d5d19a699ac7cfd983ec9ae59d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I0cf39d390954f52b91eac7503f4f76c99498ec04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ded27428303d5b9c095bdfe3df2ce4e4cb0730d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ded27428303d5b9c095bdfe3df2ce4e4cb0730d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
