)]}'
{"id":"openstack%2Frequirements~894159","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ie7c0cdb1d82fa623b094e53553b2f3b148127685","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"Ie7c0cdb1d82fa623b094e53553b2f3b148127685","subject":"update constraint for python-openstackclient to new release 5.8.1","status":"MERGED","created":"2023-09-07 10:11:43.000000000","updated":"2023-09-07 18:43:09.000000000","submitted":"2023-09-07 18:43:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894159-new-release","meta_rev_id":"ef37b5c708548ce43b51d40411a8db352a668129","_number":894159,"virtual_id_number":894159,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 18:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-07 15:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-07 13:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-07 15:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 10:28:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-07 11:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 13:23:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-07 15:39:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6e6222b8df2fa07917826a599468cb4225e4c60b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-07 10:11:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f6dd286ea1796834b112072760aac31e60a4eeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 10:28:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a21a0cab6dad414d960c608ca4a0ab0a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7dfb7bd709dd4d8bbabb4576f2903741 : FAILURE in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7ae70924afed3cf125745783a25c6d960bf19edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:40:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a17e7971084ed2b05da6704d1d564a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73cf6851c8dc45689a8f82ac30a1d003 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c71e50ca73274b2080a8207c005dd729 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e549390a67475fb354f3eb71539f8b : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e140a8c76c24d52a4b6d182f6aefebe : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae620e317bae48e187a42d6c6633bc5e : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01efc749bba64188bb6ba5b12b8ba8e0 : SUCCESS in 11m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8ebb3f286e449308a3b09ddde5a3cb1 : SUCCESS in 10m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fa6483471f7d4db288bc80c395f28a61 : SUCCESS in 5m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ec7bf30defbd463b8cccdfcd7706873a : SUCCESS in 11m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/745625831c2142f0951b18399bc8824c : SUCCESS in 5m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c46384e083dd4714a732edd11f2c518d : SUCCESS in 5m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/286bdacd68244eeb8e5d709ac9643126 : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3527007489c849a6bd7424c38886d1d9 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/36584a17eba948b2ae3b701f8e326bdb : SUCCESS in 8m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d8f6adf66b0740a5821d7e46a412bb38 : SUCCESS in 13m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1c2b45657c764cb8b3c7406dc6c4e770 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c5d54ab857ad43acb0d8190ad5ce1479 : SUCCESS in 8m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4cfb2bf8c7a34d83b112b8d42c064395 : SUCCESS in 4m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/efae3312b6394d74a914573f40f3b2b0 : SUCCESS in 24m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3557c71b78484e67a90baadc996cced0 : SUCCESS in 12m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8ae03058adf645aabcbe6e28849c83d8 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd13362b3ff44c3c96b2bac0160f5f6b : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4aff1de5addc4abb8d9ba6be6acda661 : SUCCESS in 20m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4f201d2ee83849768294101b3bc37ab6 : SUCCESS in 5m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/64c8bcf55257432bb27df596d8812a6d : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7a854e3f3684a6fad72284e5c79d1a6 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aa62583848c9454b91c31db9031ea6eb : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9694849e013b43658316c1cb4285d60e : SUCCESS in 1h 06m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaaa41365ac447acacf3756a86b5191a : SUCCESS in 1h 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"cef723b5fefd97f4a5798fde3b9abc7604c7893a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-07 13:23:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f1a347de4e2523d2f14ca216c9351b2d927b48e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 15:39:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe48a1f52517dd7fd14d39012ce4fa8c9d3b82ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:40:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60b4c53b9b4e7940e562d7c207789ba83d11e5ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d7daaaeaea4b3984259dff2cdd90af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8907cf7db90e4524b081118aa1e0aa9e : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a981e0a9bac4a97ab97f9b1cda9b23a : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/badefdd8a7aa4109a25ea77f873826be : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cb5f84d3afd4fee87e6169ba72426cd : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9113aeb5ad124b9c9768abd3f7e448ad : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/43059453a525401e97c054a0d45970d1 : SUCCESS in 9m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ba0eaa28233143b6ad93f84e0da43850 : SUCCESS in 8m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b95df696e2e84426b79f031fefd45dd9 : SUCCESS in 10m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9ebb9a4b447f44eebf0164fd5f0f5062 : SUCCESS in 6m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/60897f60627f4af6b847c3899047cc89 : SUCCESS in 5m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ae614b0bde7c4abea1832bcffd885808 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4fd2c44afc24bd199ebee321013fd99 : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/817f8b94da974224b5392703da2f9f10 : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/64393c6907bb4a2fa09a37473caf28e0 : SUCCESS in 10m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d7679818da0844f4aaa7fdd5b786900d : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bd912856ee3043c48ba23b16430006be : SUCCESS in 10m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e98c1956e6394de391c6b5d3287e44be : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ea1e0677df5b4bceaf92ff207fcc1121 : SUCCESS in 30m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7d8aa9d51ce84127838806b0cc5d76dc : SUCCESS in 11m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/589c6ca12b37411ca5f9993b96a9abb4 : SUCCESS in 6m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5b42cdce15c4838b583678ffefcc816 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f10bf788b7d4efa910cbf14e89ba5b5 : SUCCESS in 25m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3ca38944ddb64c63be4f18f270371ede : SUCCESS in 7m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cf53178406d74a55b055892a3806aff2 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fd085e72abe4019973fcfc072ff0adb : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/687cc2e792fe4fa685f99dba0255796e : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c83f2f5a61f24167ac8095d154abfe12 : SUCCESS in 55m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20c6f33ae2f043eab45b44109e0d016e : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ef37b5c708548ce43b51d40411a8db352a668129","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89d1dbd2c182603ddb7c454752a6fd38bc6646fe","revisions":{"89d1dbd2c182603ddb7c454752a6fd38bc6646fe":{"kind":"REWORK","_number":1,"created":"2023-09-07 10:11:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/894159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/894159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/894159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/894159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/894159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/894159/1"}}},"commit":{"parents":[{"commit":"66091379c628e95356d79a2586fa9ff4a04f107b","subject":"Merge \"update constraint for sushy to new release 4.1.6\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66091379c628e95356d79a2586fa9ff4a04f107b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:11:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:11:41.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 5.8.1","message":"update constraint for python-openstackclient to new release 5.8.1\n\nmeta: version: 5.8.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: Ibf402614296c8b74e079d97b66c750fc794c3146\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie7c0cdb1d82fa623b094e53553b2f3b148127685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/89d1dbd2c182603ddb7c454752a6fd38bc6646fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/89d1dbd2c182603ddb7c454752a6fd38bc6646fe"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
