)]}'
{"id":"openstack%2Frequirements~894160","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ide7f0dd31d6398fcc3bff31b881072d8bcb57dce","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"Ide7f0dd31d6398fcc3bff31b881072d8bcb57dce","subject":"update constraint for os_vif to new release 2.7.2","status":"MERGED","created":"2023-09-07 10:14:25.000000000","updated":"2023-09-07 18:43:12.000000000","submitted":"2023-09-07 18:43:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894160-new-release","meta_rev_id":"697b5bdcdd1cdf453b50b92e52830f66442a9f1f","_number":894160,"virtual_id_number":894160,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 18:43:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-07 15:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-07 13:23:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-07 15:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 10:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-07 11:25:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 13:23:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-07 15:39:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f9fb4e655ac39e23a0ded4458c7e078b76dbb079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-07 10:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28bc85e9c7061db7c80a5f50a739343a5271f50b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 10:20:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a388d455884e3ca4dc6f0f3e2cd61c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/64f4b0c30cdc4ee9b72e165d18ce919e : FAILURE in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"720f3fc4632c22e91fc9caddfecf4689faf16706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:25:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5049c9fb2e9a4ed5be0648ea083b7dba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f864ab3f454d63aff2afcc14d7c9ba : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86fb5e2666e2492cba1e1e5a9327a731 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91868812a5f6440bbaf6f54378548fa2 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68f18e0668714fc488f9e769dac5cb2e : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b01ea140dd449099e5732e9f7ba17ef : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/adc405677887430088bd84da02174f59 : SUCCESS in 11m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eadcc774086c49e3ba551b3c52cce246 : SUCCESS in 8m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e048909c64b346f4aa97b6641f808fbd : SUCCESS in 6m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/83bd964c25a04e3c817668c4f28f8c44 : SUCCESS in 11m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e47a31a9ef604056b9768df85ac0f7b5 : SUCCESS in 5m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8fa0121482f54c4f87e57cb61857da9a : SUCCESS in 6m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a730424b0894416a9f8344010ffbe0d9 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6d1c918d1804e0c97b1583c902eca84 : SUCCESS in 6m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1f469eeaebda4526aad332fc1b615be8 : SUCCESS in 8m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/67848e1cb63b42279c92084ed54f2df8 : SUCCESS in 11m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dc67ffe21359445eb6ae3d6380387e43 : SUCCESS in 3m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c0c160d0fd104fa3946ed06f99cf2291 : SUCCESS in 8m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/206ed194391740f0a5fc0e41db810b37 : SUCCESS in 4m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a9020a0770574fa695a93fbe0ea239f7 : SUCCESS in 24m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2a01717b854142e9ab9e4dcba27db282 : SUCCESS in 11m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ffb158f3658d44289b9de5b8d4f51b34 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5979026e8ab642eeae848c6d9a868ce6 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f042dd28698b46dfbcbb4430a46e65df : SUCCESS in 19m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e6a6d7932be94f7988c782a0c0f48d21 : SUCCESS in 5m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fe79f464e20c4b119423d8878e987f77 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d49be2c2b66e479e992d3a66d228d65a : SUCCESS in 6m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8b87a20aea6d49be9f7a898335894d22 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb9417063658418eb77eccd44b96f8ff : SUCCESS in 52m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fdcf368841e46c48caace68462db577 : SUCCESS in 1h 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d1b13367992c6bf84b8a22fa867b66d0eb897997","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-07 13:23:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b7025eaef54f458b7244599d84368bcca5358d63","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 15:39:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c2c9d3a5c679fd964522ea18bc5a19699b044c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:40:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d06506f1e21ceb0f6b216193b218b80dca9bcbb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e80f53035a449b1933e187077fbc648\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6226e99ee74e43dcaae87a6416f94ba3 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8d4ef58dd5c432d9115748345cbbdbf : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13e15b852580429eadcb4bc281ea89d9 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5c0f2c1776f47659231c1032908a313 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d858a44cd3c4f42ad8df055c191327e : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/afcada521a7e4e648409e75243836b0b : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42fdc6181b974cd19e956bcdd0798a27 : SUCCESS in 10m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/48f93f296e0c4e068c892ae718e730e7 : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5119cd9edcf5478b9041a42e3a02853f : SUCCESS in 4m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c47560d502e44a9f8c36c4f49308f3a9 : SUCCESS in 5m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/41371591cdb248c2956047c032bb1061 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61370d5843e64d66976a2343c1cad652 : SUCCESS in 9m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fd6286e906c74eb0af6cb959a022e0f7 : SUCCESS in 8m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/caa49fc260c6448ea222149b8c31ab17 : SUCCESS in 10m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3e20609dfcab4cb1b55d44f208c5339f : SUCCESS in 5m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a4abf5833c654a86b8fd0b7998420860 : SUCCESS in 10m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b560170acde341a8b5787cbaf10be6ed : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cd943b0d9b464d6492c22b4dd1578339 : SUCCESS in 23m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6388e2f95ae649d3bdfa79d4b382657b : SUCCESS in 11m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3e2dbfab07c2427bad22208830e59d9a : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4284dd484b7c429298e6b70142901aeb : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/098e2d30ba60479d8809e2368b16d96a : SUCCESS in 21m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/879c3b4c27064e4abb74065745fda75c : SUCCESS in 6m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4b534eeea66b44ef9366ba1a2091c8d8 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6738e1a344e34362a56e48d32afc2010 : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9c91bd042eed43dfbadaf3d1ecbcd0f4 : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8488449d4e3b4723afed8541c425d27d : SUCCESS in 1h 01m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d7e0d83526c4aa4bb1fa89b8439f0aa : SUCCESS in 1h 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"697b5bdcdd1cdf453b50b92e52830f66442a9f1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:43:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61ce175e9f17e39372d077e8fa2d934c60826cde","revisions":{"61ce175e9f17e39372d077e8fa2d934c60826cde":{"kind":"REWORK","_number":1,"created":"2023-09-07 10:14:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/894160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/894160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/894160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/894160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/894160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/894160/1"}}},"commit":{"parents":[{"commit":"66091379c628e95356d79a2586fa9ff4a04f107b","subject":"Merge \"update constraint for sushy to new release 4.1.6\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66091379c628e95356d79a2586fa9ff4a04f107b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:14:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 10:14:24.000000000","tz":0},"subject":"update constraint for os_vif to new release 2.7.2","message":"update constraint for os_vif to new release 2.7.2\n\nmeta: version: 2.7.2\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I97ea01f93e5aa4eb08660590be377bab64f5a2c9\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ide7f0dd31d6398fcc3bff31b881072d8bcb57dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61ce175e9f17e39372d077e8fa2d934c60826cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61ce175e9f17e39372d077e8fa2d934c60826cde"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
