)]}'
{"id":"openstack%2Frequirements~894195","triplet_id":"openstack%2Frequirements~master~If60986c0df24130cafc3670f35368d4cb29a4207","project":"openstack/requirements","branch":"master","topic":"test","hashtags":[],"change_id":"If60986c0df24130cafc3670f35368d4cb29a4207","subject":"DNM test new SQLA","status":"ABANDONED","created":"2023-09-07 14:10:51.000000000","updated":"2023-09-29 10:40:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"303da9ac0bc7c0d79e5eff255604bfe4b6aef859","_number":894195,"virtual_id_number":894195,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-07 15:39:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 15:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37d5e75c99f6c5db3fca8044a7f40037b5ec4475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-09-07 14:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1ca7010dd10a9fb8d61ede67fc520fdda210b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 15:39:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d325bc464d684c8981b655ddbf6cc56b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68df1bd4898a4f45984aec82c79a6f4a : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cc714611792448684b9cd09a4b13a14 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/687127e9cfee44b18bbfa4d9a2cc9377 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a96ec35fd3c4ad5844d76b54b5861e3 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3481d26d14574713a0080396f69192b5 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/552cdc42b79c4c1586ecef06ed4f1a91 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c3d54c6f8cc4edabb0a3216f5f278ce : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b8042a585e943749606114c4a15b776 : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db509d486bbf4953ab6a16a636374d73 : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/931a4a382d014214bc7783b88bbecd0e : SUCCESS in 12m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8e082f59778146da92bb2bbf102477d3 : SUCCESS in 8m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f1f1eca207e2463b8c2e64e661db9452 : SUCCESS in 11m 21s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3a3bcb88dca6422a910cb3fb0288d54b : SUCCESS in 9m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/63e04fbc64954f8387485e91fce5efd4 : SUCCESS in 5m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a41f266d3c01490b87eaadf6ec92a56d : SUCCESS in 4m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/15f81d3bef144b21973d6e65894b4b5f : SUCCESS in 12m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8a23557598f746e187adf96589241109 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fed4a439b18c4a9184d830df2c614e46 : SUCCESS in 4m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/479c42e04f5b44cba50b07c8b1a8e66f : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e170505181ba4ce4ab74eb78749ab19a : SUCCESS in 7m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3eba55569e6d4316844f7daf71b33a3a : SUCCESS in 6m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/330125f003864df4a776bf672e54858d : SUCCESS in 9m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/008715e2e9494c11a9f702953d2f0f59 : SUCCESS in 13m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/25fb05d2715945b99e3b6d41b33d90fa : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bcba6c264b274b57a2a45db19b71fcd7 : SUCCESS in 13m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d94beb6f9315408d8a1d3dc6e8ce6241 : SUCCESS in 4m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/11a34e55012742ffbe797ddce5a1cdb0 : SUCCESS in 11m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/66e44cc20dd844e88b352e399e782e26 : SUCCESS in 25m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/02ebe299eb0c44eaa4c1cbfbd027025f : SUCCESS in 12m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fc41bf41ff9b47d5842518e4b3d502c7 : SUCCESS in 14m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/55880b64514e4ec19287162c6b4f2e59 : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89bf68e836f84be383afbefc11a9ed64 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01edb0470cb84d098e0f08ef07041c41 : SUCCESS in 24m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/47d7b1e8a3b0413783093a1653ed922d : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5eaf3bc5179343248196b7bdfd00e890 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00b6d5ad1cde433785e6de1270f12733 : SUCCESS in 6m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7401b3a159634cafbf97d1bfa3e7ee56 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86f51a93258f4990b53503861a6efdb5 : SUCCESS in 1h 00m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ef90998362d411cbf5f7c4078fb275f : SUCCESS in 1h 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"303da9ac0bc7c0d79e5eff255604bfe4b6aef859","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-09-29 10:40:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"537b1e6f2ba1d070767e593c897f8bde9fd2e726","revisions":{"537b1e6f2ba1d070767e593c897f8bde9fd2e726":{"kind":"REWORK","_number":1,"created":"2023-09-07 14:10:51.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/95/894195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/894195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/894195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/894195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/894195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/894195/1"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-09-07 14:10:43.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-09-07 14:10:43.000000000","tz":120},"subject":"DNM test new SQLA","message":"DNM test new SQLA\n\nChange-Id: If60986c0df24130cafc3670f35368d4cb29a4207\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/537b1e6f2ba1d070767e593c897f8bde9fd2e726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/537b1e6f2ba1d070767e593c897f8bde9fd2e726"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
