)]}'
{"id":"openstack%2Frequirements~894206","triplet_id":"openstack%2Frequirements~master~I3ef818351d9ab2b54b58c91efdc777e0ebb52650","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3ef818351d9ab2b54b58c91efdc777e0ebb52650","subject":"update constraint for python-ironic-inspector-client to new release 5.0.0","status":"MERGED","created":"2023-09-07 16:00:30.000000000","updated":"2023-09-07 22:52:23.000000000","submitted":"2023-09-07 22:51:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894206-new-release","meta_rev_id":"067791d2f48f08adc9c04d7ac1ac211e0da8fdf1","_number":894206,"virtual_id_number":894206,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 22:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-09-07 16:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-09-07 18:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-09-07 18:13:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 16:02:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-07 17:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 18:13:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dee39cd3cb3bbb07fbc6e85f054d24829c3d0c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-07 16:00:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e41c3c2a13fb27317d256e9cc4abf1d4b49076","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-07 16:02:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5622fa28ae55183e6bc0d6f0dae147bad2d5b5ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 17:54:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780f94a93abc4f47819a8603b5d1f967\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39686da5ad94b02b4a5eca938c4a9c9 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93415db2a7af44e88e73e9cc2016f246 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89381570c3f1462c949100e41021f281 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23e20503424c4d7e991c0bab9bdfb5ac : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/089beff41995475ebd0e3e79915453be : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a4ef4c115c64e30a5e0663868abdc21 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9957935709814f17ad3f3d569b34a373 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85bc274cafe44deb864651b71ec38eb2 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/279726f338154e548610ac795f223f9a : SUCCESS in 7m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/47428e0379c14776b1596a4a332ccf2b : SUCCESS in 11m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eba88487034b43eeafbdfad49521528a : SUCCESS in 10m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a28e2f5203a2473db37a4117dd09eb41 : SUCCESS in 12m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7727551da5be4b72856a4305ff71c9ed : SUCCESS in 6m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5929e2ed06c94cc6a2cbbb55128d4090 : SUCCESS in 5m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/286466e2317c4c29bd3424440e77b9e0 : SUCCESS in 5m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2604a49b77354afab9c70b896352b2d7 : SUCCESS in 11m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c250a1562ec74a8cab42859aa5650e01 : SUCCESS in 6m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/df85ce18a31d44ea9c814eb74d537df3 : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/392db97da9f748d89b4ed99db9e86fd7 : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a7818a073d4b4b159fda8a9361b3746a : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e2c3eea82644832993962738f81a38c : SUCCESS in 8m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fe2965f5308e4054b5f63cf77c7a031e : SUCCESS in 9m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7a87696cfa074a34a085e9f372ddf07a : SUCCESS in 18m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c8fe65a7a2c1417ea79d9a30e3c9b658 : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6c1e441ae69d4df28738aceb742a3fac : SUCCESS in 10m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b140091f37f84537a5361af9d39e0431 : SUCCESS in 8m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f3c902ab9d2a477698498ef5529cca6f : SUCCESS in 11m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/12f115f3a1704648992d568f65b02d59 : SUCCESS in 22m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/afe2e59a62b3418490395806f8494da4 : SUCCESS in 14m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a8ac515b72bd48469f4d7246f0171c84 : SUCCESS in 18m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/78c7d021a34a4657a6b09de3f91394ab : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ce33f5ad60914eaabfede14d55b260e7 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/236963880d69436092460690ddf7ef5d : SUCCESS in 25m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/27339c3313d44dcc88f758e72e3b86bd : SUCCESS in 6m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6a70dcccc1824921b52e0b9916f4d099 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b5a5b4983024849a9da0d9ec3de82b6 : SUCCESS in 6m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a61544e171c34944a37223331cb44f9b : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de6a507277764f948a1c7a1bb063cfdf : SUCCESS in 1h 05m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6996bd1837f406b997b052a8401c5e2 : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0f5fe6baf7c3908008b0ab3b0d06e8f274f026e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-07 18:13:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"513eed7d91d58f8fe57a21c7478f21e8e3eacac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:14:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c085317cc31d38e647df156df348b02535065fd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 22:51:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63436abfb05d4d0da3a0d0c710af85b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939bb54bc6564459a70fd0005b7c4d64 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18508b269acd469089dea9a6755b4417 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fab1a6f2392541338a96d6f9045596e7 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7cdd7a78b624c8f93f0322c56a55cd0 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c82f708c26b24dee87336d602f1a5b26 : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d54e8ea469f44bb0830d5a5ded5b55b4 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6b091e495f844e3a9649acc7b5efb3e : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/145fb8a0f6364f0fb5dc6b7c08aad710 : SUCCESS in 9m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/24327b14c03f400b9624072c4f31d241 : SUCCESS in 10m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b85efc4b03fb49c18f676248149c1d66 : SUCCESS in 7m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8959b7b939ae45fe81b3267e684f2461 : SUCCESS in 10m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ed1ae4bfcea54a46bc342a7c9fec62df : SUCCESS in 4m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0e03a2eadd3a42048ee27dbe3ef2c53d : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/99662e40a9914bb78229c4f02a3767a3 : SUCCESS in 7m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/efadfae136924388a082eb7910be8c0d : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4cb12954ba94d77ae95a1052e3e3e74 : SUCCESS in 6m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f2dd3942624a4cd9a539313404ba62c5 : SUCCESS in 8m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4870010e79b0446a8af029672f965a3a : SUCCESS in 12m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0f5b2443694c454f9600682aa89154ac : SUCCESS in 4m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a1163783bc8c4484a09e70787f1baded : SUCCESS in 9m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d8339bf229c44aacb03a22bf982bb12f : SUCCESS in 4m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bd0eaa3226b740d1ba741d3d8f791e4d : SUCCESS in 24m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a1da97da56a448f3ae1aa85e3722d71d : SUCCESS in 11m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9fd60fe32bdb49ff8125ad1059951ded : SUCCESS in 13m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/11fa4976d0354373a9c608723f4d9906 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eacbaf8244834c9eaa36ba24a10e2ae2 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e124e2fabab47b1a52f6121eb8bcebc : SUCCESS in 18m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/49e245440fce4a108b1b8f2ff4d778f0 : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a5daab75ce6f4d658d2a4d4eb2a13a65 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f980cce15dbc4d418c777685f54fc7f0 : SUCCESS in 5m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/83e2c088a96c409982b6df0ad4fe10d7 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5531493b52344807ab5ae2b254fd770d : SUCCESS in 42m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0c509579b004751a6c6c94ac75bab7c : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"384391de2d7bb190de564d8afc10c18c70636c7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 22:51:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"067791d2f48f08adc9c04d7ac1ac211e0da8fdf1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 22:52:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a43b73fc7d2142f5a9c58598bd67ce83\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34d1eb582f146b1ab945cb29ffa1b3f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b03be4e6810b25a21324e45f9617af050c391920","revisions":{"b03be4e6810b25a21324e45f9617af050c391920":{"kind":"REWORK","_number":1,"created":"2023-09-07 16:00:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/894206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/894206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/894206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/894206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/894206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/894206/1"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 16:00:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-07 16:00:29.000000000","tz":0},"subject":"update constraint for python-ironic-inspector-client to new release 5.0.0","message":"update constraint for python-ironic-inspector-client to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie4ad16079fc4cf8b52e558c5009d52a4bd29f3c7\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I3ef818351d9ab2b54b58c91efdc777e0ebb52650\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b03be4e6810b25a21324e45f9617af050c391920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b03be4e6810b25a21324e45f9617af050c391920"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
