)]}'
{"id":"openstack%2Frequirements~894880","triplet_id":"openstack%2Frequirements~master~I32fda86f88034b4761536898b01ded5a51f63e30","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I32fda86f88034b4761536898b01ded5a51f63e30","subject":"update constraint for python-watcherclient to new release 4.2.0","status":"MERGED","created":"2023-09-13 09:56:47.000000000","updated":"2023-09-13 20:09:52.000000000","submitted":"2023-09-13 20:08:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894880-new-release","meta_rev_id":"83b90adfec649d6b8daf9c59adaab2140b838050","_number":894880,"virtual_id_number":894880,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-13 20:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-13 18:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-13 18:27:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 11:50:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-13 15:11:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-13 18:27:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"59f77093ddcb648b6b52af725229a914fe734dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-13 09:56:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a4d5fa0d0cb7f6fabe54b108fccd2740760bb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:50:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9e85fffc238494697e0cb4967286fa0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e94d0f9b9f4132ab4415361d28bbdb : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed90b114e4dd4ab086361a2c102d6a4d : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb2510e0d1234f57b9e99794df204397 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be65a94e2a2d46b8a5993b40e9ed6d22 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef9f9fc50ac4448c9039d5cd9e460ce7 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6e8f3190343477aab99451e582474bc : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc0688d9eb26486f855b7d708d0b33ca : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9b87978329fb4387a58b611a665cf91f : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d854c224436043a8a33112dddaa179c0 : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2db43eed423d4ce2a70fbbd9a4c5095e : SUCCESS in 15m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7766fc6c42c8490c9301220509035adf : FAILURE in 4m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/18ed3a389c794bb3bc1d082046f113b5 : SUCCESS in 10m 52s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/32e73e4b75244f739353ef10f191491f : SUCCESS in 7m 57s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f974bf258c3040a69edea736ad884643 : SUCCESS in 4m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/32dbd4e6dda14fbdba2d9c88f917848b : SUCCESS in 5m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ae645b2085b446c68d929330f60fe0f9 : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/71141d92d00a42f78f85e77b8ea715fc : SUCCESS in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/44fe787157b94a9a87622e92ef2d5436 : SUCCESS in 4m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/75393da95f9641e8acb8754e7dae94fa : SUCCESS in 8m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a7b95296b5ef43dca17d23970fe6f4bf : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f1db25dc6424a7f8525727adaaadac1 : SUCCESS in 8m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fa171ede2c80459b83fe1e458ee7d5a5 : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/21a078a5033f4fc0a8d49415de653304 : SUCCESS in 11m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/aac361a4312d4b32858565cea24c61d2 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e5b44b04e9374b9eae5190752c6cc77b : SUCCESS in 9m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5d7f3938bb084940926261ac6e44b1e2 : SUCCESS in 4m 50s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a203b0048a884aa988c790d0eaf2b230 : SUCCESS in 11m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d1496ba5a71543a1aae52d0b95fb2bb9 : SUCCESS in 18m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c20d883e27bc4ef3a9c84daa3888e264 : SUCCESS in 12m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/26f35b2d3eef450ab234fa630d2ea417 : SUCCESS in 13m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7fab15fb13234350b194f3220372c568 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7597f181ec394004a8c343fb45f2120c : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8ff5265d9ad4d6ebee1085698699e31 : SUCCESS in 23m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4a912e40d26146b3bf56aba712a3353b : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/88385fb3fdb34e77bb175893fc2bc2b7 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/abeaf5b35de84fa99abc5bc84187a2b5 : SUCCESS in 7m 23s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/59622b3a137d4ffdb44495c951dcad3f : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1945e6445f28460db386c69e656a58ef : SUCCESS in 46m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9e068f051094ba09e08713ebb551067 : SUCCESS in 1h 45m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b490845f0811932e41f41d33c85cd57af3268481","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-13 15:11:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4871aa0422b553da5bfc3a6c41d4e460e766fd3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 16:55:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f34f9a13a77e43ce9c156b80de8bcda5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e6aeabf4774017ac494855563739c6 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd993540c274edfaaa761b380b06790 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4d6e0336e314a01884605daaf5477e9 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f09ff16b0ab64da48676275c6fbaeb5b : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4dc49bc7d6bc440e8a7a093dd9d61048 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d627e99cf9c40789578252862425405 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d43581b7e0c40fdbab9c7a295c12c7c : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/818a2681d1304557a31a5284020189cc : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f10e87d58eb4c62a379753c5ed3060e : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f83a21d42b0a4a9f82c1091cb6d81948 : SUCCESS in 12m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9ad2f98e6e1e45ab92ee3cf1959bad3d : SUCCESS in 10m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a5ef34c99ad54eccb7bd0913e3cb6d38 : SUCCESS in 12m 33s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/46ea938c843143ce88413f168f4260be : SUCCESS in 7m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7fde83d90e4a4c5ea634ae4805632f73 : SUCCESS in 5m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e17a3b1e47fd46f38321a915556d693a : SUCCESS in 5m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f7bbb2df6ec24b00b99c2750284cea9c : SUCCESS in 11m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1b679b26a89b414f9f9c835eb9231787 : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7279c18bb69e4e6a8f7b08ab88470537 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/abc5c9e92b1f418d97b11864d85ee076 : SUCCESS in 7m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3af8c137cec14860894872c8aa5087c6 : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bc2ca3e3573430e870201d6254df51a : SUCCESS in 9m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/10378fa3dc254e5a9ac583f46257c9bd : SUCCESS in 6m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ec230a61cc4d45eabcc4419b9d24e9f1 : SUCCESS in 11m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4d1f4e9f645e4d379afd47eed96f64b1 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/579d3d93cd0440e5889445dc9e2f9ba0 : SUCCESS in 12m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5320dd6189144a0789a51b93f89f5616 : SUCCESS in 6m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8559b3da39414457bf1ffab8f3cb86da : SUCCESS in 9m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8bb23a7409c046e19b1acb24e387c6e9 : SUCCESS in 35m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b1cbe91bc7d345f3b242e56b25c9c88f : SUCCESS in 12m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/048abe436be145b990a1a765e2782675 : SUCCESS in 16m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1d0f87bf0b8244ce86a6c46efd7708ea : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/28366d72904d4de9ba5bd96764a89699 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbd1074a7d70486dbc494a6b2329f9d4 : SUCCESS in 26m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2419777d72264e6f8ec3d5cf31119de6 : SUCCESS in 4m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6b61812f759f4c2c8c9e5897a2827ff2 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40a3a943aa0f4169a28a0543c97b1472 : SUCCESS in 7m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/fdff7eb6892849b880d4fe53ad864020 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f56330ccb904c64bbe538fb73ec8716 : SUCCESS in 1h 00m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88522f2d6ad2437a991c507af2181123 : SUCCESS in 1h 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6b129608cc625be103dbfee003ccdc3d67206d53","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-13 18:27:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e3f7958493db433944ebb5576f3d9d5fff52257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 18:28:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9f33c9bb7824745fd0fa62ee16792272019898d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 20:08:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e70fa2dc70c4730bfadd1d1b12f71e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cb282991594effb64ecd7695851cda : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad6914f04a14f989b935d829b911cfd : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d45d4af90aec4a23a12af53d0d172877 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4599899f6a5249af9582d8a7c5c0b00c : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/183a8201d7a04141a3a0fffe2ce8a0ab : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed1d3169007d48d394ef172404259301 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5640987eaad740d880db76a3e8daf8fb : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c230b0dcb3f4cebb57000043c4d552f : SUCCESS in 12m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8aadba72b56d484fb0b5ba5fa13d77a5 : SUCCESS in 12m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c89fe26149154af3a466d45b384c157b : SUCCESS in 11m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4fef9960626f4f998b8a1a1159ead833 : SUCCESS in 11m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c291efb04d5a41f9ba91c6e8afa4889f : SUCCESS in 5m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8c19085c86334a07bd4fc4b8bf024a43 : SUCCESS in 5m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c9254f47aa934850ba6186bac224e206 : SUCCESS in 8m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/26691ceb28d647d4a23bc1bca461dbe0 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f73bcf6c6a324c23b8bdfc1e63c5c95b : SUCCESS in 9m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/11cf588fba64449fb5ef565a6d5848df : SUCCESS in 7m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c4499a5f0d7245af8ca195ca53e4c656 : SUCCESS in 11m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0c65adb044e64f57af5cc44d90a97fa2 : SUCCESS in 4m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c800b58b4c0e48468111cf4fa6f028e9 : SUCCESS in 8m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9bfa54cbe8aa499e87843a20b805a386 : SUCCESS in 5m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b16f261439044198ac25cc749556609b : SUCCESS in 21m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9fa2fbf5dff34fcfa73c88da8f5fab25 : SUCCESS in 14m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f803d09063044790a3add30ef2f46c31 : SUCCESS in 11m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9de2486971594575848abb540fb899ab : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b0cae1a5349c4eda84edeae64f05b179 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63cbda9067e247159f2b6466fdc24c2e : SUCCESS in 26m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6671476f72b043ca9259382ae4047cc7 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3848ab1f34084bc49907aae5cb90fcdf : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/326219ce259044b99536cdc696dfe762 : SUCCESS in 7m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c9b14dfa8773419ea06358aeacc50056 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70609502c47d4d22a16e94382797c045 : SUCCESS in 55m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e8e70525e8d48a69652ecb56adc3a8b : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"734c8fb556cab5fb52367594ea8652d41af462ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 20:08:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"83b90adfec649d6b8daf9c59adaab2140b838050","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 20:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b98ce027b2549cba92db4f2816889c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44e972bb6cd461a8fc4f3e75ae7b186 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3faa53782dbb7ff64224ab51800ef540666a0b74","revisions":{"3faa53782dbb7ff64224ab51800ef540666a0b74":{"kind":"REWORK","_number":1,"created":"2023-09-13 09:56:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/894880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/894880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/894880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/894880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/894880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/894880/1"}}},"commit":{"parents":[{"commit":"8cf4e053498c7e17752e482cd164c47306424175","subject":"Merge \"update constraint for python-saharaclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cf4e053498c7e17752e482cd164c47306424175"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-13 09:56:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-13 09:56:45.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 4.2.0","message":"update constraint for python-watcherclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic2a3b39ab24fb59ebd427260ab920dff01d491bf\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: chenker \u003cchen.ke14@zte.com.cn\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I32fda86f88034b4761536898b01ded5a51f63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3faa53782dbb7ff64224ab51800ef540666a0b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3faa53782dbb7ff64224ab51800ef540666a0b74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
