)]}'
{"id":"openstack%2Frequirements~895084","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ia6b8ca382cff1080c5d03b3df7dfd486cd6eaa2e","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"Ia6b8ca382cff1080c5d03b3df7dfd486cd6eaa2e","subject":"update constraint for os-brick to new release 5.2.4","status":"MERGED","created":"2023-09-14 01:32:22.000000000","updated":"2023-09-14 20:24:21.000000000","submitted":"2023-09-14 20:24:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895084-new-release","meta_rev_id":"63ccf42637c2e122f78b58233310ec75327e5ccd","_number":895084,"virtual_id_number":895084,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 20:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-14 18:57:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-14 18:57:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 02:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-14 02:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-14 18:57:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6b54843ffa4e83b9a4b4c085f983b184bf13e1ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-14 01:32:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f152720f96c1d29feb197b13c3ba2591f11867b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 02:07:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9b894a8d5154071ae099ce23c1913f0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b3006a7d7eb547109fa340360d742f91 : FAILURE in 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c5df813c3d000e22f04018fcd039ee1b6405c4c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 02:59:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ba7edc0924e4b62a00e112cf313b6d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6247b4ce1a5f4dd6af4523e93d6a4776 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4343b7a4fc564b219b43d07e7f283bf2 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a871d95c771448b2a2324fd0fac56c87 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d5e8a0cd4a841a1982fda8260a480a1 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/983d7ab0eeb347999674081363be535b : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f474ed48b9a84fc9b7ea3041ebcecd08 : SUCCESS in 10m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1d4858b6f0804f98b49ee2831a7eca3f : SUCCESS in 18m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4219a137ba9c4e33bb8d62b3bc6f0bd8 : SUCCESS in 4m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9f42e729c3834414b22455e68d1ecbac : SUCCESS in 11m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c623e17a0be44c49975e0b5f39aded24 : SUCCESS in 4m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/59592b4aaf334a2b87287ec36bee61ed : SUCCESS in 6m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3b8d1221d50943eea0f6ffa4cf4bd31f : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca067e60029a448f9a14e1e4cb1b76b1 : SUCCESS in 6m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/26438cd744ea4f20bceb7455684a4b2a : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5dd1a9ecb459455f9181d10c8d115fc3 : SUCCESS in 12m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5bfada59372b46bd9ea3309f3090c2d2 : SUCCESS in 3m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/be87e405d22c4706a68971ee1d8bbd5f : SUCCESS in 10m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8c51c52c5ca64eb9a14b196b05321066 : SUCCESS in 4m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/319aa1073e3c446386592473e1fe046e : SUCCESS in 19m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/49746ecd5da04063aa7cc6189cafe5da : SUCCESS in 15m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9e9ae594e53c4c7ab8396982b7ad681e : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d04495eb50d40b2a3603487b3e15578 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e806b5cad13c491d98ef4ae450090f4a : SUCCESS in 21m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0e3db3f163574e5fa715d205eafee437 : SUCCESS in 5m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d7bad89a0f474201a81fa49bf0b2af4d : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89a5de571e0641c0b0fa0f459d99cc34 : SUCCESS in 5m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/324e6011744b4234a3370fc4db93ad82 : SUCCESS in 8m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0355b83646f84e62a036110082fe0dc0 : SUCCESS in 1h 09m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea2398c67d43459fae3af7c445bf0fba : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"282e36f2f59b10926aec4ae4bdea038716691f60","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-14 18:57:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f49878f8c36ed929e4bf7b43f69e0a51e7ee1c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 18:58:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43b9d5c033f480035c58b50476d3d2183b015095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 20:24:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7113dd4038a942798e8d420dcdaa7ebb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc2fee43610499f91aab396bd20c3d7 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8860c632e9694236b9dd10779f441a35 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7857e9ce07a4996966de573845fc911 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/196c5cfb3fc14c8aab4c88962c21ebee : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4cd6a8b06a24fe7acdc82520bea1db6 : SUCCESS in 3m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/45f191f230c94010a690c343695dde90 : SUCCESS in 8m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/75b1fe204b6d49bfac09092317ff79ae : SUCCESS in 8m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eab49017e02643c0bdd31fddffbebe6c : SUCCESS in 9m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ac66136836774097940ed267b13da757 : SUCCESS in 4m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9656e7fa161241f6862d9e8e7722a932 : SUCCESS in 4m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/86bc8226d7ac4a58ad71e3a716b29a7f : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97ca57658c8f4828a125bec59ccfacb8 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fadfed49917a4e3baad5b263deb8f1f2 : SUCCESS in 8m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/931a4636015c499b8343ad7c7512b885 : SUCCESS in 11m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2d001a155736475f856231fa591f43ed : SUCCESS in 3m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1dbc58b291024f8eb9c6decf481428ef : SUCCESS in 11m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f22793f0c36d4126b97b637c2f58b728 : SUCCESS in 4m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/52d21a2dcc3841f3983bdc8ca8c69f9e : SUCCESS in 29m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4a4566588c1f4d81af8839d135ae682b : SUCCESS in 12m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4b392a3ad25c4780af814bf9b06ac365 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a9d5498b8b448c0b936f5ef338d4198 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cef4b03d7cfd49ffb8fc5f3b43d132e4 : SUCCESS in 21m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9c767a5ece7548eca97e91db5d056a3f : SUCCESS in 5m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a8ddab9ff1b6448cb17ea8e9ef56f341 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7175e9b2d2ab4cf59f66994fefae582b : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bea7ec3a3a3c406b94cc05a8dab4c9ca : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/269dd12a84e2479f9f1b324af0127e44 : SUCCESS in 1h 01m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8da587c8ea9e4891b1c3248e15e6678b : SUCCESS in 1h 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"63ccf42637c2e122f78b58233310ec75327e5ccd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 20:24:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b0af2c04ee08d4a5463d9aec04c92092a03f69e","revisions":{"9b0af2c04ee08d4a5463d9aec04c92092a03f69e":{"kind":"REWORK","_number":1,"created":"2023-09-14 01:32:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/895084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/895084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/895084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/895084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/895084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/895084/1"}}},"commit":{"parents":[{"commit":"2c8af0942819557a9176448a89620969761368fe","subject":"Merge \"update constraint for os_vif to new release 2.7.2\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c8af0942819557a9176448a89620969761368fe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 01:32:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 01:32:19.000000000","tz":0},"subject":"update constraint for os-brick to new release 5.2.4","message":"update constraint for os-brick to new release 5.2.4\n\nmeta: version: 5.2.4\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Commit: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Change-Id: I5f6c2c8a2af264c84fbf872ba4166c4762b94be8\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia6b8ca382cff1080c5d03b3df7dfd486cd6eaa2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b0af2c04ee08d4a5463d9aec04c92092a03f69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b0af2c04ee08d4a5463d9aec04c92092a03f69e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
