)]}'
{"id":"openstack%2Frequirements~895121","triplet_id":"openstack%2Frequirements~stable%2Fzed~I7ebc056454f5509a988a8bbaaa63fd8e419d6a39","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"I7ebc056454f5509a988a8bbaaa63fd8e419d6a39","subject":"update constraint for os-brick to new release 6.1.2","status":"MERGED","created":"2023-09-14 01:43:34.000000000","updated":"2024-01-25 12:19:37.000000000","submitted":"2024-01-25 12:19:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895121-new-release","meta_rev_id":"39ffc837860538fc3fbbc39d66b86de0057fd449","_number":895121,"virtual_id_number":895121,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 12:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-01-24 23:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-01-25 09:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-01-25 09:59:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 02:15:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-14 02:56:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-01 18:01:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-10-19 15:15:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2023-11-09 16:54:39.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-01-24 21:31:00.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-01-24 23:51:59.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-25 09:59:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d69a276f21f1f83913f72936dff92ebb91ee7a5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-14 01:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e838e1d4a069adb4e84b35d5104457ca23beed2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 02:15:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39628d0b5bd640c093150f6d9fcad86a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/75a5c103213f4484b9b983b26ceba69b : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"398d75ecae434f78526412f427ecc4eddb715aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 02:56:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4e03cd092a842a9a3568f052285aa0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312e652369e84936aa51a850110ee01c : FAILURE in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d234676431b5423ba927364dc5811f9a : FAILURE in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d9cff513e364fc995d918d9c25472d8 : FAILURE in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8cee4a32f8d4c2ca370e3a1f0cf5b1a : FAILURE in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a8791b349f84de6beb9b34a3981e2ea : FAILURE in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74c79ded65a14f2db29193d806ae02b0 : FAILURE in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c7a7f7bb5264945a41e981c94c5437f : FAILURE in 6m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1af18ae07a644aeb8939899f2c9948b9 : FAILURE in 10m 18s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a8343943201c4ae8a22ac50559a48e7e : SUCCESS in 8m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4ff6d7ae61e94a0eb6dfd3925559db90 : SUCCESS in 4m 52s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/062b0d9be424498882e8266de36acef8 : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e301415389f14a0d917ba80d3b3ac338 : SUCCESS in 10m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/594c3711498941b2b3ea8f5c4e6efc8c : SUCCESS in 5m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/15fb74e17df94c53b866c4ed5582f26f : SUCCESS in 4m 50s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1040a6aac041412a87b0e3cfff0382f7 : SUCCESS in 8m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b5859ce4513244b1856ff7f192c271f5 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7e2a1eac8bd40e9ad408c524b2ebd06 : SUCCESS in 6m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a35085213a774e50917779d3f0d07e83 : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/30b88552556e49a1962390a1e216f338 : SUCCESS in 11m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dda6477971924e8a847015c5be55cc76 : SUCCESS in 5m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ce585925b716466abc8e0195007ec5b0 : SUCCESS in 8m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bb02b2f8546f461fb1d41687935eb66c : SUCCESS in 6m 05s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/994d7e0b1fb84cd5b20ff563d388c208 : SUCCESS in 8m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f6e14414481944abb58c3f2b22bcdf6c : SUCCESS in 25m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4c74fe635f1c4c12b103580d0d43993a : SUCCESS in 12m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4bc93c3a5a244681ba7dca2110748f95 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a68b406c082492aa03003db6b358bd3 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/becc687156a846599e5b92427283e3cc : SUCCESS in 28m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/118d35181ab749ec90ec07dc52313d78 : SUCCESS in 5m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3ac7ad4368c046919055080a4b400ae3 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe1284e421a54a81bc926e3f8b7c5a37 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d4dac91f33e94e35ba28375f7d0d2216 : SUCCESS in 8m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ca517fbd6b3466cafcab6bbce26406c : SUCCESS in 1h 02m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/624ef56c74dd4fd48b18bb6a5d35f729 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c74a698c5dba37d4193ac5f5443ab975b15c8b0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-10-01 18:01:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc54294b4072a6437acee1662905c450ede28b62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-01 18:19:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519ad176a86343d195612019bc098677\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/da29e3248f82462a8d48196ce63bced8 : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c13f34c5183542954dc319ee1f56009556091938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-01 19:18:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ece9d51c3ab74957ad1d9adc3520e070\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59bcd5e2ac394170a3ae23168b57bd5a : FAILURE in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1413674668dd4622bd29cc8ef52a12d8 : FAILURE in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41a95a2ca2e14a3f84fe4d06663364fb : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a48b286ba27644309c9e3dfa7b7c3735 : FAILURE in 2m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6aa8a14017f740dbae81af599e515605 : FAILURE in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba19caf1aee4471a96b6603ebd67c5aa : FAILURE in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f6c7ad85962b4d01900881a44a34b3a2 : FAILURE in 8m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a38b0f0733d842ecbdbf0e2fd950ba62 : FAILURE in 15m 09s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/1d9398e3231946c7a779c429835c505e : SUCCESS in 7m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/de239c3b367d463289574fddfc40c72c : SUCCESS in 4m 30s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c35a2c18413d4753a158fe531c260241 : SUCCESS in 6m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4ca6716ee5f54ed4a4a9fcc130ca834a : SUCCESS in 10m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1c184691b5064b0da743d5c4eb211173 : SUCCESS in 4m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/be6fc29c5ab84f32a1cc9a21f502801c : SUCCESS in 4m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b2c80b5776b54e3abbc8876df25c5646 : SUCCESS in 8m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e95a9ec6d2f34772ad6e5d565bfb5ccc : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e9c1bf76c2e483eb23cf7565ff563a3 : SUCCESS in 6m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6a9b9ca754e5497ab70c3b6974fdb49b : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fcd60a285ebd4cf8a54f0c6d4b2c9300 : SUCCESS in 11m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/64f362b1c6834a97891f7313f72d9e86 : SUCCESS in 3m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/06ed120e4f264f2da149146c4b3885cd : SUCCESS in 9m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e987683ed79347ea8818ce4e674e94f3 : SUCCESS in 4m 17s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/c020503a7e754d569040545e1d969e97 : SUCCESS in 8m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b614693bc65e4f71822e9577eb9c9733 : SUCCESS in 18m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6efa6bd53a5249c39541d22799e422bf : SUCCESS in 9m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ced702e9cc604372a4d25a9f58bae114 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01c01bfbb1fc4e59962a8f454bc70b0b : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6017eaf13eed450e95832e18a25cfe71 : FAILURE in 20m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b9fc815074934d5b83b25796637bbdd7 : SUCCESS in 5m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c8d4cf8e5b2e40cfb4d957f97cf895e2 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77cfd076321847eba8785db053fa0d22 : SUCCESS in 5m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be38c8be34b24ed98f3835d8cad3425d : SUCCESS in 8m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f417e7b700c45d397d001352362f726 : SUCCESS in 40m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92e01858b8ad4e27aa489dc54d2bea5c : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6a40c0d450398c959c62611297424bb66c9f9896","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-10-19 15:15:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"391d998cc204c7ea7d174a8b0045aa6d20cb2724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 16:57:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/615db44642b64cacaeba8ad6cb4e756c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd161a5d260c4e78b28f1af9eac767c2 : FAILURE in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf015f1292c24c65bc272b12ea55cd49 : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb8b39a142ec4fec851f9466ff7a662f : FAILURE in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfaa1e46e93a4a8388bc6c3a1ed963c5 : FAILURE in 3m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9156bf74cfe74e11b2754cc98454387c : FAILURE in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2bca073ba804517ae480b9354410c17 : FAILURE in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdbf4a596f0f4a809ac55959dab8aff2 : FAILURE in 11m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f28ff48a5ca74601874d40461d593664 : FAILURE in 11m 55s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a58a78715cf44ae3b0120304ec45d0fa : SUCCESS in 8m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/daf03ecee6c04a689c68a7aa0c070f07 : SUCCESS in 5m 11s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/8b5e260b7ddd44cf893c250a94d64f84 : SUCCESS in 5m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cfe7afdf0250443b8295cf64e912537e : SUCCESS in 10m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d595bd4f2dfa49088f50a581ffd394ea : SUCCESS in 4m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/58e70113bcfc46c584652c3d0c1cc3b4 : SUCCESS in 6m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8c5e50b3aed74a4490311a7b574482ad : SUCCESS in 9m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3b0546f2e6ed4260a9f0c2af94b337ac : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0580c7b25a5c4417b7abbc78ea237355 : SUCCESS in 6m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca79a942054d448e9749f51f0d0ed323 : SUCCESS in 7m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/56d16f7fd04e496c96f42a50ce81f0e8 : SUCCESS in 12m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/02c1584a43ec40a4a1e131ab6e1ed976 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/78edf885c647440b9cbd6a34e5ef3189 : SUCCESS in 10m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f22d5c67e9dd467da582fb18f701cee3 : SUCCESS in 5m 57s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/8e6b5475b4894bfbbe3d54acaa8f5803 : SUCCESS in 11m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2721aa1c5f184f89b97b7a03ba0a0ef0 : SUCCESS in 22m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/09f7bbffaeb74cb2bd6f53fdb58f130f : SUCCESS in 14m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8ef20c138d8541b0aa72f89540a14802 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f966d683c32949fbb7e561be7b14f279 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41b0e558cd794e83932bf685577b8ee7 : SUCCESS in 20m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a7426cacde564f90a3c9d3ae210667e4 : SUCCESS in 6m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2469574fd4c54445a93995c92d94a666 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfff5a972c944e4786ae6abd13591da6 : SUCCESS in 6m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3d3be71d75224ac08a5e0128d937f5df : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/655e325eac6446bb9f3e494b161ef2e5 : SUCCESS in 1h 06m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5df3305d47204fcc8d19a3ade4d757d0 : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2af9e10fb3542bc57a4afee002d36a5b45fe81ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 20:16:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11fa5bd5ca074bc2927d4e9d85973ce3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0050930073014a24acb95bbe713fbc32 : FAILURE in 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e1e922d527ab187bc64943406083d5e308baa2ef","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 16:54:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8be8da8c00133a2cf69f08cf4a1e743b9a357ed6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:09:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34431647b7bd4353aba319ec42289856\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bcf983e86ddc4078a569a7db295ade67 : FAILURE in 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f626605dc287aa1553e246608da2b4c385bd6c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:20:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a459cc85202143f4950f974705433260\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c652e26aa10409b8f27e2587cf174cf : FAILURE in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8d31a4cd86f42d08dd142cc769d87af : FAILURE in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5a09273abae41b889f702a351ec88a2 : FAILURE in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c73833a3dad4432b9fb60329c6e29c4f : FAILURE in 3m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ee73dbde50c4751a679ef3e9bd0237f : FAILURE in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8ec24f11c58403abbad90fb1210bbea : FAILURE in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56e1daea20f44fbbb6515f386b6ff0b7 : FAILURE in 10m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/345bc2faaabd4d33998b70235a2d7f22 : FAILURE in 15m 13s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a64479eb3f5847878de1c6d176770318 : SUCCESS in 6m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3fc42728609d4126a932c1f236b05acf : SUCCESS in 6m 29s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/8e4baac589824d6ab07f8bb261fdb5db : SUCCESS in 5m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/efd354f2692b4bc5b4a1543db9a7e8f5 : SUCCESS in 11m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f5a64cf6c6fd4036a856fd2ce9cabbf2 : SUCCESS in 4m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5ae192828a5a418b94ae146405233d4f : SUCCESS in 5m 10s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b01ae211f4254cb991e79e5e72d869f9 : SUCCESS in 7m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7b7c4c53baa24cefbd39ebe2d4e0e259 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4c5207ce38d4ac6aad8afae4a305855 : SUCCESS in 7m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/40f191347de14cc785d8b9a84b1553bc : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7207d0cd9df64479a67c2af720e42803 : SUCCESS in 11m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/379f6055d99e41b09382d10f6b34958b : SUCCESS in 3m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4fab2a7451d44ed5b2277446d1cae04c : SUCCESS in 9m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0408ff8c0a2941148100fb2cf71e13c7 : SUCCESS in 4m 46s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/91b9447795814bba9e09c488b6a5a504 : SUCCESS in 9m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a312f6bc5bff4d918e488c13218ae207 : SUCCESS in 25m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1287e0d5ff8d48ac9a431aaa6d96af56 : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/693c3d4b8a27454f978d9a258d7b6c2e : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f717d0ec8964b39946a403eee6881d7 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/649a419219c14b76b95f6e044c03e572 : SUCCESS in 23m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0e5baf3032294980b84b784cdc495208 : SUCCESS in 5m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/19dd12f2cf564d20b4734352c1a43c70 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4105f583fde4b289519f97a23309ed9 : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bbf21fcc02294eb6a595b4b93774eab8 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78b7d6be7db640b29e88a6d0663c1786 : SUCCESS in 54m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09a5b0d78bc646a1a712e9579de7c50a : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6036bab8fe903af3291bf2d6bd83c0ad25492c38","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 10:03:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd1ab6f4e90435a42cd84c7df3e85eaddff191d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 10:23:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/389bccee52c742daa42e0c1dc6d95320\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0adce0521b484c298e27f1acdb307549 : FAILURE in 17m 25s","accounts_in_message":[],"_revision_number":1},{"id":"46a05e6f88aab6948213d6fd9da57f8e4be4fcac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 11:21:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a3929fae3b24161ab064e3e48c30df3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac54ecc9b58241088f5bbc048a27b3aa : FAILURE in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5ca6bba65c74f84b210674269f1c2c9 : FAILURE in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5275d0c1be544a7fb5c32a6fb0d2b686 : FAILURE in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d802157056b4b369ca791c82036c8a3 : FAILURE in 3m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43efd5607a2c453593dc34017b062b0a : FAILURE in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5402bfe5383a4a088bf16b6fe27db96c : FAILURE in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/028ca40dbf9b4a29bb690f398414fdf5 : FAILURE in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a083e84775a4232bccac6f598b26f82 : FAILURE in 13m 24s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/db6fe48eafea49a1bd31cfa3a716e0b9 : SUCCESS in 7m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2ee8d311b8864b53bcb8fbd054b9205e : SUCCESS in 5m 07s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/bed4afe23b014fa1901b00a6c362b22a : SUCCESS in 4m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d9c0589d7c284a288f687563002a189e : SUCCESS in 10m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/43b80f4953ed4202b5dd5afefeba6171 : SUCCESS in 4m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a47a726dd8704373935fc93dd1d237a2 : SUCCESS in 5m 50s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/afafd106e35e4d47a33201068160d4a6 : SUCCESS in 9m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8ac84333fd32412f814517cd2b2dc6e9 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e55b1e0e9a6e494aa2119e93af8ce8ff : SUCCESS in 6m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2e02b5d63ee0417b8e31e8b5c728f9c7 : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/57c9f6d548d748f5bba622aa0570f6be : SUCCESS in 10m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3badef5d5dd34c5bb2f4892a5dac8e3b : SUCCESS in 3m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2cf7c0ea372043419bee285e5ca13491 : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/467508d740c7488f8e9d8bd8c0d4cedc : SUCCESS in 5m 45s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/9b220a1c047749ee9f061aad354d7a16 : SUCCESS in 10m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9244e9afe38149fdb4c749f14bde8dc2 : SUCCESS in 31m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/15fd4c157a694abbadbe83be4c54fda6 : SUCCESS in 14m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/65d3232f1b6d437685211aba40594f85 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e378357171b4be4ad37f50368f73bbd : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48f9406fadad4e88a33fe6cd5e353334 : SUCCESS in 24m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/18663b11b9254c269915974f30e732de : SUCCESS in 5m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7cd753e23aa94daca9f41b64a6773bf4 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/106ccf5af91443a59b83411b644f48c6 : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6134453fac9d4b8298d486993f6eedcf : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fdaff5da67f440087a8b770fa6dd81c : SUCCESS in 1h 10m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91742d418bfd42a4a3e1719435898b71 : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bd318d011445498ba37c9719036cb88fc1090aac","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 21:31:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db655c53415453ee96786702669a2317bb8ad50e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:38:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e21f3c1658c4a8a9f78e9406ccfe55e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4db0e4e9860c4b2e958af81ef7f42048 : FAILURE in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ab80fd820a6ed57281c6d98d6299f7c5504b6119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:54:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0eb2b7515949af89d917eab75a83cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f163b9270f0447bbed98509893eed75 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ee2ec7be3244ab997b00f2fb8d5a00b : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/527c58122db74e2595dfaf4545dcab3c : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d84ad8366414585a2b8584d5166bb17 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e9b8043a8514cca9c6d09b2cc9de9a5 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c204d3c3bc784b8ca27d6559b8ace8c3 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f5a4f02c097b4add8a66f442b5ff9a86 : SUCCESS in 15m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/be587b6790964c4abe004167b519fee0 : SUCCESS in 14m 01s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/bfb065040a354b58a8160e0760fe9545 : SUCCESS in 7m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/312f662bb2d847e99dc5b034076968d9 : SUCCESS in 6m 01s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e746f7ac534c455cb55abca5b2278581 : SUCCESS in 5m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8c16de2f6974439e99448acee8fbe969 : SUCCESS in 12m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8fe0a6d7a0f64795bfab394df06cc455 : SUCCESS in 5m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f6dd7b8d6e194d338c21f733040b448f : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/911dc72cf5494770a7519219d2cf658a : SUCCESS in 9m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4b7f60d8116542ceb041d7e918622069 : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea710d65b75448b6914adc3678dbc27f : SUCCESS in 7m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c93143eac89f44baaabf965c5be6720f : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7ced927affac4e27a0803e3c4cebc969 : SUCCESS in 11m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/12757f33f30b45d99a509e4764989dbd : SUCCESS in 3m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/573d578451e145eabbb6b27ce4094307 : SUCCESS in 8m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/354bee8e339b4e3d86371612d3949142 : SUCCESS in 5m 18s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/4ef40b98ad95464ea36262f368cb756f : SUCCESS in 10m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/125559be2c554f2fba66550ad4fb09fa : SUCCESS in 25m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0628bace3c3647f988d7fc243ac99b00 : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5cbfca02187745aca4030adfb1b515d7 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1c38e5e9f894beabf18dba3c246f199 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b44f24305d7a49cab8464774335e0314 : SUCCESS in 22m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/31568d9735944054a817e6379cd52d02 : SUCCESS in 7m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4778721a076c4c77887ae2c99ad278c1 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64887590b05b467a8a4505d5efc6d01c : SUCCESS in 7m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7b18b85f809d4d68bfb2bf0880b6bd07 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a744d04b10cd44af9b99e65d3a4f7275 : SUCCESS in 54m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ef29f8ebefb4fcba2e5ba55728b8509 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9f1e16a2139ed5f9ab171a8823f02b2feab5c27e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 23:51:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67a0e6b1ab56387b35ac22d411a13d85d5eb9c0f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-25 09:59:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bd1b39c464812ca04ac8d1a478432ef8686f887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 09:59:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a9233d62fb99dfe8516ac34f879758d173182e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:19:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62db67e1245444f0866f22f37b72c9ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20fdc4480af42c3b05c90a7ebd83461 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef235d86953e4d219c72abe5b0e4b1b7 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf322ad4b206487395343a9858410206 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/804373e24e5149b3ace02dc06c2cacd3 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06899afadbae4797b97778518e2d9890 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f3eeebbd5c045d696381d40472fddf1 : SUCCESS in 8m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e594e6b1c5f4b11a50e1cf9e44850b9 : SUCCESS in 13m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bd321d7e9cc247a98f76c6634f3ffafa : SUCCESS in 8m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2c51c554f52e4b81bee101d9e069158e : SUCCESS in 4m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/74ba2bd1f02044d394dc586ba63ab7e8 : SUCCESS in 5m 40s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/31e62199cfe94fc988797440909447bb : SUCCESS in 6m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2c71e4289da54b7fb79acd15c496d95c : SUCCESS in 6m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51f221f132b446129f51cda43e39c7a8 : SUCCESS in 7m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf80660e17f640fe8933159831e46314 : SUCCESS in 5m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b7e47b61cc554a5d8ad4a7aca93dae14 : SUCCESS in 11m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/27e66b5aa2324cc3aaa7b994e2ad7d7b : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ab0e53439eaa407c8afcf4065d9913cb : SUCCESS in 10m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/666847543e7c4f9493be8c4ba2995c30 : SUCCESS in 3m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a5a63e0b09f9498f95833d1e17ef57af : SUCCESS in 23m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8404a8880de34e1e84c46c5d5c7d7e68 : SUCCESS in 10m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2dbc6597acee4734b916ddcf6bf904f2 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d59be5dec05496ca39c6cc327829501 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87a53c3d0e9040348cdb9bc2043666c5 : SUCCESS in 27m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f841e048d1a64eb48953646956646338 : SUCCESS in 5m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fc9bf14c7cb54a6cbe1fb6d3c9a23081 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b4c86dc360d416f9aa8f4f0a23f76cc : SUCCESS in 7m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/457c523445a2413d83039ecefec1346d : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f68ce42a0064651bd873c8cd368a45f : SUCCESS in 57m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7bc41bdaf9149f9baa5a4c79d6aab5f : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":1},{"id":"39ffc837860538fc3fbbc39d66b86de0057fd449","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:19:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0266ae57cb1cb5b32e317efad38c2c3768ee22ca","revisions":{"0266ae57cb1cb5b32e317efad38c2c3768ee22ca":{"kind":"REWORK","_number":1,"created":"2023-09-14 01:43:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/895121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/895121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/895121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/895121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/895121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/895121/1"}}},"commit":{"parents":[{"commit":"ce23ac59f84228b1747c22311992169cf827cff9","subject":"update constraint for oslo.messaging to new release 14.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce23ac59f84228b1747c22311992169cf827cff9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 01:43:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 01:43:33.000000000","tz":0},"subject":"update constraint for os-brick to new release 6.1.2","message":"update constraint for os-brick to new release 6.1.2\n\nmeta: version: 6.1.2\nmeta: diff-start: -\nmeta: series: zed\nmeta: branch: stable/zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Commit: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Change-Id: I5fcb50e407ef150b3b448b305152c294d8fe19d0\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7ebc056454f5509a988a8bbaaa63fd8e419d6a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0266ae57cb1cb5b32e317efad38c2c3768ee22ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0266ae57cb1cb5b32e317efad38c2c3768ee22ca"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
