)]}'
{"id":"openstack%2Frequirements~895187","triplet_id":"openstack%2Frequirements~master~I4249bdd50aa77cd184bbdeed25f0bba70cd85b10","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4249bdd50aa77cd184bbdeed25f0bba70cd85b10","subject":"update constraint for python-keystoneclient to new release 5.2.0","status":"MERGED","created":"2023-09-14 15:35:07.000000000","updated":"2023-09-23 01:47:27.000000000","submitted":"2023-09-23 01:46:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895187-new-release","meta_rev_id":"42253a98ed261862327b16622fc05a3f129177e9","_number":895187,"virtual_id_number":895187,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-23 01:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-22 22:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-22 22:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 17:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 13:25:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-09-15 17:00:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-22 21:57:31.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"6a58aa3a8d49672273451371f02162f7616eecf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-14 15:35:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ddacee13e1107689c1d6e9ab550cf024cb391d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 17:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f398d93d879248ccbdddc1d61cc5c14c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f05843cf3c046f78038df0933f1752e : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5daf8f4c538447998bc468672b4af5e : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbd4a1668d0d4863882a12626fcf33b0 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e5b906f4a7f4e6789a780233cd525ba : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a8ae706100643828f7e48393ce86d46 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25dd4385f0ec4209a65ad00a2fd8b74f : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb337624f03a4210bb326699d192f97c : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2aa63e6580124dcdb06b278a561b5ad2 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8fba4ec3d3204205b3fee860eba56bdf : SUCCESS in 10m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c36364cd9a454dddbd7bf887f8ac6b79 : SUCCESS in 13m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dbc96a3f67bb4423947a9fd8d533999a : SUCCESS in 10m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cbadf3676a8e47e08a7e2259df50167f : SUCCESS in 10m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f4fd5f34f3c94a7f8b36ccbd9e9e93f9 : SUCCESS in 8m 13s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/44cd5e27aaa440058e9d3d1094854766 : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2fd7ce3512e5495a949556483b1d7d87 : SUCCESS in 5m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8983ceca105248fe9e479ec194582d03 : SUCCESS in 11m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/76ff4cdc7bcc45faa3d6420d08a8925a : SUCCESS in 4m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2f451b35c95f484e8300feddd0bbafa2 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d399f918d8774d2b8b5abfc04df767dc : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e4b9531be538424a8c2c95512a8622a6 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eed36c49d3dd4d33bbf5c525e3a8d8e9 : SUCCESS in 8m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4fe77f22adf1435d81352050b76a6ebd : SUCCESS in 9m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/78f12898f9a94c209a7e907ad74ccc40 : FAILURE in 10m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c04fd5372f934fc8b1ee1b612edeffbf : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/39a0be922b3d459c9791e6e7b771f636 : SUCCESS in 15m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/875f591784ad4c2097aa9d958f194c98 : SUCCESS in 4m 55s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8780ef14be584cc394b910447fc088ea : SUCCESS in 11m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dd14d2516e42462ebadf54d676bcab9d : SUCCESS in 26m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ed5298ebb2844cd484d09a7c4c031314 : SUCCESS in 12m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0ae8db693c4c4ddf9b24d4d018037d19 : SUCCESS in 13m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9f6cc8122ff8438487e561e8f7c94946 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/76ddffa223144d47bb15eda9c20ff2f5 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a8ba48450f845e2ba4856d69780d229 : SUCCESS in 25m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b5a1de853bd84205b54a7ed108150fff : SUCCESS in 6m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5ad522127e89475694c8675efaee1c26 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c208a229f8484e5bbf4d95351e80f405 : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3c8cb873445c4daebbff40997f7dbc53 : SUCCESS in 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/740826254ff34045ae5a6a9acbf64812 : SUCCESS in 55m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9b173ca20f04b65a66cc89f80243885 : SUCCESS in 1h 45m 55s","accounts_in_message":[],"_revision_number":1},{"id":"224244eee8ea32f3ef0f3b97c1a8c71ce4b6915a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-15 13:25:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3791eaba000f6715b0fb798172b771863d532370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 15:20:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c107307b035487b9fd72afd04969ec5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d8fdf1f5ec446694343e025fd44f41 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc774a6f15cb4c019d6d01c580b1029d : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/761e570c9f2e4a379097eda3bbc32a74 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4a47fbb4fa14cfbbde2abc765321503 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aeae35b55a084f0a9397a9553e2bf137 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6387802939fa49498c7569cdb37669ab : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cbb9128a6e74207a5784acf3254b8ad : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5fc6eade1834a88bdf8492994b79bd1 : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a9e5f651763249cdbd9be36c3dc315f5 : SUCCESS in 8m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/480c414b9b21406f97fed6509ac6a9d2 : SUCCESS in 10m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0506f356ee7e4347bb9bc1189db6047f : SUCCESS in 7m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1f44726bc4ef4fc994ee6a952fde9c94 : SUCCESS in 13m 16s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e3ab78e930764e9582c301ea46eb58b9 : SUCCESS in 8m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cb7bdfd604004f0da3eb57ce3aa1e921 : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3712dba3afe94c979bbeeb1754d2e8e3 : SUCCESS in 4m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f817209f25da47aeb5d698d04eedf3ae : SUCCESS in 12m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/24dad4fa056a437ea37aca8972a9c39d : SUCCESS in 6m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6d0598eea9234256aff57836621de5fa : SUCCESS in 5m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/46e5667f86b94d6a9fdf674eacdbd39d : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0c5d41eb726a4269a7f4bb79115a4b38 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7913480af5374a26baf9a0979db86881 : SUCCESS in 8m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cc6608ae76b644b684752098da5656fa : SUCCESS in 8m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/eb4b033897194a97b26a553faee74c4a : SUCCESS in 10m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/52a003ed69354bcc942896785f6c1e09 : SUCCESS in 4m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e92c0f5ece114e59bb46fb7db81279cf : SUCCESS in 10m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/86b7b970dbbc45bbafc424190dda3b68 : SUCCESS in 5m 51s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/43a8f3279b1c4fccbfaf1ac406d070aa : SUCCESS in 10m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/beeae652fb1f4ed7a4f20774da97a9b1 : SUCCESS in 25m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7a872e33ede142a4a5a5a4e6210463e4 : SUCCESS in 11m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/054d73d181a44cdd94e152377d9eb39e : SUCCESS in 14m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0835a1c3c97c4fbda9d59179c1fc170d : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc3c477ba7b8415c98b084c19c98363c : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b171bd5191345f4812f498364e91bc2 : SUCCESS in 23m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/435da72f43784f2882232db6384c0f84 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5248ba5420814f19a5f0b8e48ab1d084 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/795b5aa2f5644882a1dc5be8d8619c42 : SUCCESS in 6m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/018ea00026b44ea59b338462f9544794 : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56e9dcf9f1fe4933a681b7d1bf8567a5 : SUCCESS in 1h 12m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34326a656fca4b28a30437a603cd7415 : SUCCESS in 1h 46m 22s","accounts_in_message":[],"_revision_number":1},{"id":"bea5ecbc44941a4792ba7cd9efdea8b057f7e7ad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-15 17:00:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de9d771d0d6134f6c36692baf7838de968568429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:20:11.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 895187,1","accounts_in_message":[],"_revision_number":1},{"id":"9376955f76ea215738d9f18caf169b02d20ae5a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 21:56:59.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbc4d759f13cb3e2469e00aec4967200f2c13683","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 21:57:16.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 895187,1","accounts_in_message":[],"_revision_number":1},{"id":"df1f8729e5a63b697683a3d3ff777076ba05c88e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 21:57:23.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 895187,1","accounts_in_message":[],"_revision_number":1},{"id":"5516d603d988e6dd80943aa5333fef5c3ddd7669","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-22 21:57:31.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 895187,1","accounts_in_message":[],"_revision_number":1},{"id":"e4ac647b7853100a7642bec48e376aa0a6188575","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 22:38:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7fcbf831e26df85bb683de8c9eb307a1a22c4fa4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 22:38:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f8ecaac6dd54147ccc1007917f0111b629df744e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 00:10:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6f9b589bcbd4b578225081b2deec435\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a6fdf9c02e4de2b3a054cdef61f8f7 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e65d3cf34e6c4c3aa1ccdec608c2c3b5 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbac88636ded41bab3dabbb4598d2f26 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c12ffda45d704ce9805594151325c9cb : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/790288e896f745f3a89160a3c857b655 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d54f04864ad74765bb741ea38c707c3a : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da4246c5999f4c0188f02c3e388a887e : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6d9ac42de9444fcba5b489ff0451454 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/669b3408ccf44b30acb6b1818a94cc3e : SUCCESS in 8m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c8a665d0bd494ec19eca127c42003cff : SUCCESS in 11m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/104fc5c7a3c144e49e14f98d25c50a28 : SUCCESS in 9m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ebd449ddc98c4e91b3d7436083f2a13b : SUCCESS in 12m 50s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/98f01a708e0e452d8841729e76d90f5d : SUCCESS in 7m 17s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1216ff123c3d4d09b110262d5560062d : SUCCESS in 4m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c0c349988d8e456597ae9eab2dad978b : SUCCESS in 4m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aad07074ceb8401dbdbdc1aa97a450fa : SUCCESS in 11m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7e77fb7b271a463d84be516aca9b1d72 : SUCCESS in 4m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fd89bbd61a1446c59f1366e8fba9a0a0 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ed0a204015db4c739819291ab9875232 : SUCCESS in 6m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/24f87c6b29d84f579367d01913966425 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41dc7ac503494cb388ba301e5866eb8a : SUCCESS in 7m 34s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f70e2a9cbd0c4fcdaed10d4ed893263f : SUCCESS in 7m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ac628b10413e4c4ab97b1943ed17dc52 : SUCCESS in 8m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/de7e9432de624121bc133844cb2d1c88 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8d6a81777b5042318df3761d68f520e2 : SUCCESS in 8m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e192002229064e28a5dea09b428e0b99 : SUCCESS in 3m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9b66991e335d49d0b1d91887907b4050 : SUCCESS in 10m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/15931247a49849a38a16ce4582e67354 : SUCCESS in 19m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/368f83679334424da19b3fa80cee1208 : SUCCESS in 8m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f8e6a3edd1884735b802313cc80e7886 : SUCCESS in 15m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e89d072dc65444eba157d116a5af55b7 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf1ee55d649c4444ae9982757885afdc : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88b94021a55e460abec27cde4c02b982 : SUCCESS in 19m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fa17778474b645a1a0886c079b01427d : SUCCESS in 5m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5748605467c843339da31645cd0b9d55 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4db25d9e7d3494e8f677cd00570985e : SUCCESS in 6m 23s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/640181353cf54b5fbe8c04f185033a71 : SUCCESS in 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9980df47a2bf405c8c9450c10eab75c4 : SUCCESS in 43m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7194500f68d408ea67d7e3f6cac6af7 : SUCCESS in 1h 25m 57s","accounts_in_message":[],"_revision_number":2},{"id":"66a369294df92a219b02722cc30e4b2710dc6a0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 00:10:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"34890022fddb5033e789bc600dae8bf40d1e4a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:46:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ec52ac335eb45709d8da5a2ea698099\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eace7427080043319dca25f3e6e0e01f : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ed8937edff414f8df63e8106d96782 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7f3999b13e644ac83944d5e6d2a9d9f : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de3e29fff4d34c5bad1485f55fe0a640 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6531b5f52e624d0190f91623d6a29481 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02c9b79a943144ab90b4d18f233fb16d : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49ec536aa74b4d8d80a5c32ae000fbf0 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b3fdd5316a7949858e19813f2a5d0c30 : SUCCESS in 7m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0cdc2b45ddeb4360b7d1c3c2775f89d1 : SUCCESS in 13m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dcb3ed4c3efb4b939bc654f73a4a6348 : SUCCESS in 9m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f6e5ae4cd4304878bc1d8412ac049da9 : SUCCESS in 12m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b038363efa4244a8a20bb936d1aded5c : SUCCESS in 4m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/29f84947e0ba440fad2bd4965ebed3ca : SUCCESS in 4m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5c10a9f5ae5343a89dffbd748b948553 : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6bebf0867a744e3994eac1dd576237bf : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1062afa8bda8436999e59f84cfeb8b21 : SUCCESS in 7m 43s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/eed1fa8e19a540a3a4bcb27086bf926e : SUCCESS in 8m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/10f2d71892284570a700c09dc3efe969 : SUCCESS in 11m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/62a6744ebdc44625a875b2ad1af3c979 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/950778db02914d67a4f4ee2fe21be185 : SUCCESS in 9m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/588925c427944fafa8c3df6dc02b0f68 : SUCCESS in 4m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7640933d4f814307851c9655c5f213e2 : SUCCESS in 28m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/66c99db92f104144adc3e1753fbd033c : SUCCESS in 9m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f8dbc7f2fb9e4f5dad81bd6933b3bcef : SUCCESS in 13m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2ede1cb43a054b2a8bdb30220e6fe8db : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eb5445c3712d4eec84616d70a5d6db27 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/590e23b13e3042bcbf437226ef72b63b : SUCCESS in 19m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/71a53e359d6740f386fdbd60875945fc : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9425b3b308784eedbe43ae2322133f6a : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32fbcc5d62cc463bb6fa620180c26b09 : SUCCESS in 5m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7b0b17381f2844be844643e9823fc61d : SUCCESS in 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17334ceedc6849dd9b1f87be34fc09b2 : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/391614b784c44896a3d75dbdfdc6d4fe : SUCCESS in 1h 30m 19s","accounts_in_message":[],"_revision_number":2},{"id":"505d29642bdd279a9bdfc70a7ecc9ba22028998a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:46:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"42253a98ed261862327b16622fc05a3f129177e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:47:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c4a21dbb7e406c8d5b04d4da996561\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de6606f45e3448d9fdc3f869424625c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d246194555c63b432f72a6cc301090d4340bcfd6","revisions":{"e7db46b6c5e06f223a2308f9e07c0508e497bc8a":{"kind":"REWORK","_number":1,"created":"2023-09-14 15:35:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/895187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/895187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/895187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/895187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/895187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/895187/1"}}},"commit":{"parents":[{"commit":"b7cf680233409b977c52ee3ba076ea67771ecf1a","subject":"Merge \"update constraint for python-watcherclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7cf680233409b977c52ee3ba076ea67771ecf1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 15:35:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 15:35:06.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 5.2.0","message":"update constraint for python-keystoneclient to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ifa041a86a45c8c8952c80c5b4b13da80936cd379\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: David Wilde \u003cdwilde@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4249bdd50aa77cd184bbdeed25f0bba70cd85b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7db46b6c5e06f223a2308f9e07c0508e497bc8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7db46b6c5e06f223a2308f9e07c0508e497bc8a"}]},"branch":"refs/heads/master"},"d246194555c63b432f72a6cc301090d4340bcfd6":{"kind":"REWORK","_number":2,"created":"2023-09-22 22:38:12.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/87/895187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/895187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/895187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/895187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/895187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/895187/2"}}},"commit":{"parents":[{"commit":"27b842f5414d8f631404812eb36b6951e8b716e3","subject":"Merge \"Retire networking-hyperv: remove from requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27b842f5414d8f631404812eb36b6951e8b716e3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 15:35:06.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-09-22 22:38:00.000000000","tz":-300},"subject":"update constraint for python-keystoneclient to new release 5.2.0","message":"update constraint for python-keystoneclient to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ifa041a86a45c8c8952c80c5b4b13da80936cd379\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: David Wilde \u003cdwilde@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4249bdd50aa77cd184bbdeed25f0bba70cd85b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d246194555c63b432f72a6cc301090d4340bcfd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d246194555c63b432f72a6cc301090d4340bcfd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
