)]}'
{"id":"openstack%2Frequirements~895210","triplet_id":"openstack%2Frequirements~master~I083673acaea554aaea1f04d2c3f3e8c94571efb6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I083673acaea554aaea1f04d2c3f3e8c94571efb6","subject":"update constraint for python-zaqarclient to new release 2.6.0","status":"MERGED","created":"2023-09-14 19:27:50.000000000","updated":"2023-09-15 13:25:13.000000000","submitted":"2023-09-15 01:44:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895210-new-release","meta_rev_id":"24a68728a9dcc0479b12249bf7732f9c58f570e8","_number":895210,"virtual_id_number":895210,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-15 01:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-15 00:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-15 13:25:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-15 00:09:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 20:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 00:09:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-15 13:25:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2b155568363fb8a015879d41583e26b44d4fd91c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-14 19:27:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da31690525f00aa59b86402d64a4b1ceb30c2080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 20:43:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d46cbd64a064c3cb8feb00b90922c06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e22ca62d4054c568f0374c822d1635c : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da852928e3741cdb4fb4478f8b46de0 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac31363821474742848487155cda72b4 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/750c7569993948adbeb9b7086caed6cc : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9979839306284fae8b74a797397a7463 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/955b0409e5604d888f1aee57bd523b56 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8f863c959014bc79a024c3c65a0d7fb : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4eb2bb5d17564abbbfa527a7cc8bf839 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/383820eb7ea444bba84d114f1b44f4c3 : SUCCESS in 12m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/23c112d642e94a22b491126999259aa2 : SUCCESS in 9m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/211d46f133bf43b3911640a5c65e5548 : SUCCESS in 7m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4af107c3edff48d48df7a0061b26a388 : SUCCESS in 13m 12s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/57f71a53c0a345d1afd6a7d7f60f106c : SUCCESS in 5m 39s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/650e7c3db10641558745956f3614c4cb : SUCCESS in 4m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4ea53ca2f1b448e3987ae7395187a304 : SUCCESS in 5m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f57b14f62b5d4165a3c629cb79ebc2f7 : SUCCESS in 11m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d903bf8579ad4e1eb423874cc12ec1e7 : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3815db68e9ff466cb49410218761d57b : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e33fdfde9838498688a3be94acac907d : SUCCESS in 7m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1660da69875449c4ada7554817cccbdc : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9a1644209714431afd04b82992a0b81 : SUCCESS in 8m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c221bb32aad34d2a8f4f39d122b4b5b7 : SUCCESS in 7m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7dcb2daa89654bb798a2b9fb3c279ed6 : SUCCESS in 10m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2f78605155ac48cea6649c5260ec2c57 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f75a864e83aa4169a75f5bec2fbe36c5 : SUCCESS in 10m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b5b75fa3ee134dbf98dcdc1eafc3a559 : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e5e371f48fe7466bafbb00cf5feeb52c : SUCCESS in 9m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3870c6a79fa047e5bbb372680167a836 : SUCCESS in 29m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/60295dbc70d34587a84a34f23256dc9e : SUCCESS in 9m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f334221b15b2425396fbc6eea99429dc : SUCCESS in 13m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/da8fc3192bdd4746846766291b435c12 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/576028f11c914ead98a8a0816df42bfa : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e62366903b604054bf1093bcb4155ad7 : SUCCESS in 21m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9a10fa225ff8481e9c06abc81fd133ab : SUCCESS in 5m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a782d066cc6e4e6a81a4ac249889f6ec : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/60f938bad0284cd28cd9a5e101a03474 : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1cae051a75af47ca9ffa3a09f2204aff : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/badf3d74732549c8ad253e1995fa80da : SUCCESS in 51m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34526a2fb2c94c379b010bcef116631b : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c70fb030e7d4b587342b4ab59ed70948deccf15e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-15 00:09:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e068252cda09bcc1048e2a99050cb53e71dcc21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 00:10:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"953ed93bfe21736f4620ecebc027dfdda20aa4d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 01:44:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a25cdc263bc4d5e9cfa5e283dfc34b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2614fb61ce54d159549b9317e8e5c0f : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6a46c197f7401b814e7894999165d6 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0bcd407e81d439580ca82afb821130a : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2c70c3f74f44a8a8115a58ffe1078ec : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ce2c9cf97c74d778cd6f548dcd96786 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f24b5284e93d4b31993e7947997c8064 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/deb63003b8a942f4b63bf3e566bdb849 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1e37a8d781484fdf885e792b8a11e7a6 : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/de4f9db995b3431d864f47a7432bc2f8 : SUCCESS in 14m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1a45bac32c5b4ffda97fc26c27a0db5a : SUCCESS in 9m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b5df3553bfa84c10bf4d4675361263f0 : SUCCESS in 10m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3185742ec7954141b15e51c23e00a357 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/56614f8861294fed9dcb9a011a925fba : SUCCESS in 4m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9b68eafde7b8444db8ddb1f569c65f40 : SUCCESS in 8m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b848d6c08255421ab220c9a2d30b20db : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a9cd264a29247018f5e02c003f2d0bb : SUCCESS in 8m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1efe179a4bca46b1817fc0ae32a994cd : SUCCESS in 7m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5b1f2fa4b2a644d5a58fae69f28cfd07 : SUCCESS in 15m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fa79b7218733479a89c3e9ad246dc65c : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e626b5ece4014d2fae2c2551e22c8add : SUCCESS in 14m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f0c8e0104f2e46d2bc2b6a25b3eeb365 : SUCCESS in 4m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/54e3f25130c240fdbc0ff347d3788872 : SUCCESS in 28m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/16cb12e9546d4d79b6432589aca1536a : SUCCESS in 11m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ca67f06f504c4be6a2cdcd79049d45ad : SUCCESS in 15m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2ec8b0c00cf54d9f92617aa22d7bbbaf : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/62b6c57be92d4780869104788f47cba5 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a31bb1c2357f4b5ebe0db4bb1178e736 : SUCCESS in 23m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b9bd2b797bf849d18436ebbba46bfa99 : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/474e9db002ba4af6879f50f76fc345d5 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61bf15ee1b3448f0962f04cc121605f4 : SUCCESS in 7m 21s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/769fce602b0343a182c77bc0c1da34c1 : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46b660bfb0e449b1aa9d5f2b6fcf1c0f : SUCCESS in 1h 01m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da22442b639a4b66941783e89229bd25 : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":1},{"id":"cd7424a253765e18b7dcebb013086e658a664acf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 01:44:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4528ec4121a2e7be2fa1bd0a350fe54ca2ff48e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 01:45:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c45d65e97a24c1699e121a476169fa1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2daba71c9f8b4e9fa79df90f76fb0b1e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"24a68728a9dcc0479b12249bf7732f9c58f570e8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-15 13:25:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809","revisions":{"c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809":{"kind":"REWORK","_number":1,"created":"2023-09-14 19:27:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/895210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/895210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/895210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/895210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/895210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/895210/1"}}},"commit":{"parents":[{"commit":"b7cf680233409b977c52ee3ba076ea67771ecf1a","subject":"Merge \"update constraint for python-watcherclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7cf680233409b977c52ee3ba076ea67771ecf1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 19:27:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-14 19:27:49.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 2.6.0","message":"update constraint for python-zaqarclient to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1245c938250bf04f91e4891a1dce325320a07c8b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Hao Wang \u003csxmatch1986@gmail.com\u003e\nChange-Id: I083673acaea554aaea1f04d2c3f3e8c94571efb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
