)]}'
{"id":"openstack%2Frequirements~895270","triplet_id":"openstack%2Frequirements~master~I00830cf9776b403226c04049c85aa9fbee8007d6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I00830cf9776b403226c04049c85aa9fbee8007d6","subject":"update constraint for python-solumclient to new release 3.8.0","status":"MERGED","created":"2023-09-15 13:01:24.000000000","updated":"2023-09-15 18:59:35.000000000","submitted":"2023-09-15 18:58:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895270-new-release","meta_rev_id":"aff7b0f5b5a76dac05ff78124841b7cebf2f7e51","_number":895270,"virtual_id_number":895270,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-15 18:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-09-15 13:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-09-15 13:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-09-15 13:25:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 13:24:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-15 13:25:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-15 14:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb945df66bb5eac51be5cb555cdfe0328f2d743a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-15 13:01:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e062b2b0b7f497c2b8727a14ee343c09fdc2f52","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-15 13:24:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"647e96b7ab326dde87af28a8a4db46c65a16d14e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-15 13:25:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c164f513b3ccd8482f93140172c53c2578732df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:47:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ddbb6b61c94349a05b3ecf13b6fc4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d1f09ccc19466799b0a696536f7a58 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4641dbe362804ffc907cf88402c70533 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a2c3ebddaa64186824d08d21c299128 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b6412a6c99a4ac29944bf3436741b00 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcf5ada46e424e4ea63924fe8ee5eedc : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b423a3f5a9843e08004bcad48cf99d3 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9723fe78277f473ab73345fc82cbe5d3 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a737e0fe743540a28bd7f22adc434620 : SUCCESS in 7m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/174707ec6f3e4dd28e4e5639798abed8 : SUCCESS in 11m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1ea6ff410c794fdb913513b2b764de48 : SUCCESS in 13m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b1b2db822516407e9db80fd8295eb0b7 : SUCCESS in 12m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f18a634d27064d628ab08b6bd7b987f8 : SUCCESS in 12m 49s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5a741f8186384655ac4d03cbacd161f3 : SUCCESS in 8m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9fd2d70978a24370be46c24864c0c9a4 : SUCCESS in 4m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9d656b4a03304445aec8a0817300e20a : SUCCESS in 5m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8edc4f64b64f48d5a5e7769253e2ad48 : SUCCESS in 12m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e17599c2c301421cac25b0eca4bae790 : SUCCESS in 8m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/184c54d55494403f9e4a421aff079d43 : SUCCESS in 5m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a2f31b83aa544a37a30ff5510fb1234d : SUCCESS in 8m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5b5a0d435b494e278b36d1d35fd1a13f : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/557c6c8289504d6d9abd45301b98b131 : SUCCESS in 12m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/91a577f3fe43488b83e955762e4b22e9 : SUCCESS in 9m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f604bbc9ecc243538e7985a027a30ff2 : SUCCESS in 11m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e07d3b79d7d8491b8044dd143b054937 : SUCCESS in 7m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3a7bdaf8e2214842a8667145c8cd68f7 : SUCCESS in 14m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f70e0db8685749668faf007c2906478d : SUCCESS in 4m 37s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/15ff9ae1d7e844a288087a3ed54d28ed : SUCCESS in 8m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f7b94337e17e41aea7fec3094603dc8a : SUCCESS in 27m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/af11bb0ace194aa188c539eac7b89aba : SUCCESS in 11m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/82d42e76c86343cf86936baf4e3129ce : SUCCESS in 15m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7161ae5a20d04b64bc253f912d58e542 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29c86d44f6d54f719a0b61716ff04791 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0314239109a749a7804b52ba862ad2fd : SUCCESS in 25m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2fa89b2f70b44ee5b292550fa4d4223d : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7fb7780b612e4ff9b64f57b769b8d65c : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c4995cc3ed744958b9a3645adad3fa2 : SUCCESS in 6m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1923b9b6b1b3437b9e01496039588c27 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcf3f535b5fa4004bad20b3830dee580 : SUCCESS in 58m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1e428c0c40f499787be9e6106f5617b : SUCCESS in 1h 14m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f4b4f2b29313870cf016211763b90d0b3ac7221d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:49:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f9b8ba4a1edae4ca8c969651032ab66ea3cb347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 18:58:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3cca4e10584475a8ba684d20bc454a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4133e65d35044170a2049358640e34b0 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/678944f4b84b40f5bcbed33d1e5c8d9b : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45b8533b1bab406783040a36013f6c5c : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aad645da1128482fb2afcda47e587a8d : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/539f140b84c54d3ab74c0f4e08fc1987 : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc78383906a24b0e8c2fea2adbb16bbd : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68d77c333d3e4430985d55c68ffad146 : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d9249eab69764f948d5231437544990b : SUCCESS in 8m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/120b3a0c17494caab7fe56f0e8345ac3 : SUCCESS in 12m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e515cee9c7e74fbbbb94adfe52c0a987 : SUCCESS in 11m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d03f9f68fd014ab68ff796285a9ac750 : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/87f04e40225d4de88fcf78eb3f48ba99 : SUCCESS in 5m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/98bf16e397854ff7b3d5e7c786294de7 : SUCCESS in 5m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/864c40255a84408e9c94b66352d56858 : SUCCESS in 8m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b7ee734f87e84730942ca449f97c9ded : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e4154710df84c42b1373a9b717e2246 : SUCCESS in 7m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5437be2873c1468896028d2ca0f3bcb0 : SUCCESS in 9m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/19b8fcc7000e4f7c9a9be97cff31b2a5 : SUCCESS in 19m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/08b5c116108740ce910534ed49a15001 : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/45b2510ee4094a028113835c6109211e : SUCCESS in 9m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2c4a94a136a745639ced9a490777880e : SUCCESS in 5m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d613a7138ea94e0290d57d0e0008fb09 : SUCCESS in 26m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a0f3a0b6580149f380db81d96e262ac2 : SUCCESS in 10m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a55e3fbd9e924ebfba5c343542066afb : SUCCESS in 13m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2707c904e78045f994ad6f98775600de : SUCCESS in 6m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9e0a913ff634d458105874b3521370e : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab4862342c9b4e90bbb435f82f1f37fb : SUCCESS in 22m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ee6973c3f17442cf95fa3001e82449a7 : SUCCESS in 6m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b8e24db13da04834a4a1baca11e7ea8e : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70706517d23c4dc595052ae9613849ab : SUCCESS in 7m 27s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c5debcc593cd43a39b89dacbe477334d : SUCCESS in 13m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d77c018e23264257bd1363c075960179 : SUCCESS in 1h 00m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54db1fd1dd3644c6a860addf09905cfd : SUCCESS in 1h 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"277583e66a131536dc8665afa3b7e0af4247371b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 18:58:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aff7b0f5b5a76dac05ff78124841b7cebf2f7e51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 18:59:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c973cb2fa4624824b07656e9de473e5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9289f0be1a904bd4809eca5cabc5e8a9 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19439f2672b6ceb7bd4ba1a334c3986f5ac4b2f4","revisions":{"19439f2672b6ceb7bd4ba1a334c3986f5ac4b2f4":{"kind":"REWORK","_number":1,"created":"2023-09-15 13:01:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/895270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/895270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/895270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/895270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/895270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/895270/1"}}},"commit":{"parents":[{"commit":"c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809","subject":"update constraint for python-zaqarclient to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 13:01:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 13:01:20.000000000","tz":0},"subject":"update constraint for python-solumclient to new release 3.8.0","message":"update constraint for python-solumclient to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id93fdc00e6e26e509084b1e2335ddfed77389d6a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I00830cf9776b403226c04049c85aa9fbee8007d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19439f2672b6ceb7bd4ba1a334c3986f5ac4b2f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19439f2672b6ceb7bd4ba1a334c3986f5ac4b2f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
