)]}'
{"id":"openstack%2Frequirements~895271","triplet_id":"openstack%2Frequirements~master~I6efdfd85897f4ef450bd491f40ca6b49e0c50e7c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6efdfd85897f4ef450bd491f40ca6b49e0c50e7c","subject":"update constraint for python-mistralclient to new release 5.1.0","status":"MERGED","created":"2023-09-15 13:02:47.000000000","updated":"2023-09-23 00:50:16.000000000","submitted":"2023-09-23 00:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895271-new-release","meta_rev_id":"50764dfc6ac1775713c95183f05ca7f7a12b1fe1","_number":895271,"virtual_id_number":895271,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-23 00:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-09-15 13:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-09-15 13:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-09-15 13:25:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 13:24:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-15 13:25:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-15 14:58:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 21:56:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"0417332eb7ea6013de75f0f4d399c559e649ec36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-15 13:02:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab1201876eeaed2c30b4b365637cd2d65ea91645","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-15 13:24:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2983d67767aa5db3a0588708b9294b4517ca49d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-15 13:25:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ba591f035ac16848727790883b078823551b3d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:58:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/660c93bca8774aa88a874574c68fbed1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c72ead713424b2bade580ec08cfe6ee : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41286eb6080f4745b1c301f554a5f091 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b5fcdbd51f4579a437a0df82e2b1fd : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f26fec571dd84b99be0ff3a39cf0e66f : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5fbc7d5d4a443c1853a0e4d8f2b7949 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cc592129c52428d8a27f263a0529f8c : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d799da271e32488e9f7cefa08477aa41 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d566549e516144d5b2739947b9bc4387 : SUCCESS in 5m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e6d3283bc8614060ad548e1350c8ddb5 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/14aa83bc3524443b932078605f762fba : SUCCESS in 13m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1e4dffd8de274de68eb78f7dc1aa4e71 : SUCCESS in 11m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f3780477a4964894a8442161191309e3 : SUCCESS in 11m 38s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/eda2516c9ebf4613b4066a03cfc739a9 : SUCCESS in 7m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b4e82f049f5e4d5e991780fdbf52e186 : SUCCESS in 6m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c75ca15f1a8e4cadac4d42db2fc0537a : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c6cf6c01657a4925ba961f4d03da81ff : SUCCESS in 11m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/90b37c6d8baf4b329831924815ccc8c3 : SUCCESS in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fd575225d25641d7a9473930b15a05e4 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/addf8f61d0fe4b1d9fc2ce5ccd008254 : SUCCESS in 7m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c91a899435324d34a54c967b0f80e9fc : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cabc26e44204920aa1cdabfb6b902c2 : SUCCESS in 6m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/28a7cb3d80534e75a00b7e794fab9280 : SUCCESS in 7m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/46ba78ca9c9a4d77ad1be1e8cdd16fed : SUCCESS in 11m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8d32a194b0c64ef6abeb5fc638b7abc3 : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/98c1e43f0d8f4080906b570ec9b8e6cb : SUCCESS in 12m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b8bda0ae2d744f1186c747e0ff09d121 : SUCCESS in 5m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fd6ff665da674f3d90bc24524f0824d5 : SUCCESS in 8m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/76afe9187fd24f138b872e0a5641551b : SUCCESS in 26m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/05d899d201cd48e48541db9cb3c5180d : SUCCESS in 13m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c8de749455fd4925b42c37ea0a472774 : SUCCESS in 11m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/06443cb52b564f059d122255ac610dec : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abdafd78b4494cb698024fe467e3ecd4 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/912fd2d394c947f78e3906302eadebac : SUCCESS in 24m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/add925eaf8874591826a3a31921dde1a : SUCCESS in 5m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6ab2b33ca15c457ab7687965d3f39292 : SUCCESS in 9m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd0ff7abc5d348d38c0965c35af3ace0 : SUCCESS in 6m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f27e92aacc53437d91ce8207b01836fb : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1154f3a13f94c729ed359c39703d51a : SUCCESS in 57m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3cef977ba994b8b8680892d852fcb47 : SUCCESS in 1h 49m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9ef72a55e0b351dabe561526c2283d6f6d7f1843","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:59:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0355e36898b2cc8c18ab5a059e7ae456ee6d61bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 18:58:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/afb7ee77c2bf42a4a34888fe46b70f32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424e46f39ec647309cb1f6ee5f2cdc8a : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f01c6d59c344df984d957f84c40790 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcf6903d3259404b9801214cc51e2ca2 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edb66d99c6a640269f50ceeed4cefb99 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dcff27ef8a941c495434983c05511da : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6735543b2ba4baa8c2ee5224e90d14c : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06f6fdcd4010445bb20d60794a6ec910 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a5eaadc0cb8e4deea3ccc0ba7c4ac503 : SUCCESS in 11m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/30927316fb264db5979dd736e9dda84e : SUCCESS in 14m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5822bbcffa074c69876d9726bb72ae3e : SUCCESS in 9m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a4c8a2cf21624f9ebcf9871779d34099 : SUCCESS in 11m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3bc5719262344c88afcff6bb1bfdd3e7 : SUCCESS in 5m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5f0ff2d464e844909a529dd26d51fd9f : SUCCESS in 5m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/712b302cc0084f5daa852094a601a36e : SUCCESS in 7m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e4b77e479f3046378a8e7940c8298de2 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/105e914dc0e945cdabc388aee7ad9be9 : SUCCESS in 9m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b058ee86ba98460bb5fb92a6f61a3aee : SUCCESS in 9m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5085311cbde4407a992fecf83ff92925 : SUCCESS in 10m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ee73ec1e7cf04b53b08126fedadfbb9c : SUCCESS in 4m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a1968eb5d06646cbb764b08c62a7884a : SUCCESS in 11m 55s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e9ad91a8ebbd499492555c2d5e0a7d67 : SUCCESS in 5m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/75f0526cf76e472a895b00dba7dc9830 : SUCCESS in 25m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/026b65928f6f48a990480ee8200a4783 : SUCCESS in 11m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6c22af7c3ab04775a2f8110fc4df70fc : SUCCESS in 13m 30s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/406d04c950924f088710678ee4a34ee6 : SUCCESS in 6m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/82716764056945879a1cb475d78efb03 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7c28ee599bf4a8d95c2434e8f55bcd3 : SUCCESS in 23m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4c6cae81f9be44b88b9dcc62d78674e8 : SUCCESS in 7m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/23073e1aa59449a09c6f4a695e3f896e : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdbce5f0250249e785c67d07320b75fd : SUCCESS in 7m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/21fde7990cb04b01a707cb89349f17fd : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42d6dd807e3341719a8a2d4376ef9352 : SUCCESS in 1h 02m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73325bbaef7c4b2abb80033ff6f98bca : FAILURE in 1h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0ea6480fab467ddb8cdb89c0c98dbbb8ede48d2e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 21:56:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bde292be77c32d3b3c89fe3dede75d72aaf14d21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:32:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed145e2f08e84b53a37b50b4c8ba5e0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff753d136b134951bf92cd694a4e9014 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488fe90ec621499db295d329e5f327ec : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5d7f313933d4f8e9aa2f916b34e34dc : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e02c170c4d1a48c4826a6867b03e5275 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99c62345d016467a8d33a783c3c64078 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec9a7a8fb57e4936b070082856c53094 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2014804477843ca922ec77572e0a96e : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c03a093c3cd94850a97c1ede9cca9726 : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c5508994b9d14d119be257961358cd76 : SUCCESS in 10m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0dcdff37bc5e461eaf7d70e251c66143 : SUCCESS in 15m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5f4b16f7aaad43acbbb972839641421b : SUCCESS in 8m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/40978366ef5f429696fdf5446c677416 : SUCCESS in 11m 00s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5b29a3225a6d41539981146e527e4838 : SUCCESS in 7m 19s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b092e2b9295742eaa1f42399f2113278 : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/42a99ea2ff8b40cda0191bcf26fe5782 : SUCCESS in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d1701abcf6ba4d3383d0513df29ccfff : SUCCESS in 12m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/378163ea17a44ab286461fe7890f6d93 : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/20350c9e206540e7b56583b152647929 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0718b1cbda234a12a25ae0fd8005987b : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4cd3e472e7ed48e998839fa804c26ea0 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2a9999de8804785a0e3cfe41a070005 : SUCCESS in 7m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1cb6f6e1f52044c5b0f65581dfb9ec07 : SUCCESS in 7m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a3d63d9543244e5fb9f263507407705d : SUCCESS in 12m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/26334f3f94094dc890adfd5fc7d80d81 : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9273009404ee419a90fb0722c0abeae6 : SUCCESS in 9m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cf36c928926e4ec4b521ca5914db2e2b : SUCCESS in 4m 47s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8896820e70854604a5bfaa621047e398 : SUCCESS in 9m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ebe0df7148904de683962119832a3e83 : SUCCESS in 25m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8b705383d684c49a27a2acf74ec64ee : SUCCESS in 12m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/acc1d3dd0eb64bbc8754dd1691fffb09 : SUCCESS in 14m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a78c46f7aa17467a87890dee6659952e : SUCCESS in 4m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff76d1e19676443b974df9cbb8bb01d5 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2538402bc09443a1b986c8ab6f531ea0 : SUCCESS in 25m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1328c0306ce84cf5840694640b7e1108 : SUCCESS in 6m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c0194ac4aa2b4c349e0ae57e2428d9be : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ce6c7888f24466eb909c758149cba99 : SUCCESS in 6m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d74157e7c9c94101b8bd46fe356e06d4 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da877eb7123545f29e2a379dca788979 : SUCCESS in 42m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cadb1dceba414cb48b8493af3d0b0615 : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b4d72fdf8e4b3fefbb334bd9b2895eea0fe92b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:33:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc11d8ead65a311c16789f41eb18d225a8dda1ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 00:48:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79e604bc36f4d78be53a904e2ed9e82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a9aad871824fe8b8b4ba98434cdf35 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3890905f9eb1473f91c8da681604f29f : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0b909af2c3945e48633f89db09ddf9d : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4957ec59787a4b82ac5a631d92de3e24 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/defad888f5b04912a7821ff57bc1f67f : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a874548be894020a2e8883bc81b3989 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/015c2113c03c4a5fb24a865593327e4c : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a47e793e53f49c3ab0f4d1c19945991 : SUCCESS in 9m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/64ffd8f50e5c40cd8fe1396a04550d0c : SUCCESS in 14m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0413944f5b51426788c32026b7b1569f : SUCCESS in 9m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2c3ebc55ce034011aad0f47851e406c8 : SUCCESS in 13m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7001686779634d1a884d9ff66cf35709 : SUCCESS in 4m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b3a14113386d4554acf0c1f8e61ca7be : SUCCESS in 5m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/724bcffef4e144dfae13144aae8b1b6c : SUCCESS in 9m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7b16b76b18534aafa3cf3e9e69376043 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ae4516ca8c648e2b9722d3c7072faf8 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a9622d73a91e4b0fbaa5a6b3cb6e6573 : SUCCESS in 6m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2501c2a50d2b481db42c663f3d3ab093 : SUCCESS in 11m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/339e8d0fcdb74579b2e3b5e8f7d2e71e : SUCCESS in 3m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/272a85df64a043a39b135645b335e87d : SUCCESS in 9m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/01f734af437f4ee7b74b6fd5b420a88b : SUCCESS in 4m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7edd678817494c37955c284137f10722 : SUCCESS in 33m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5d4bbe5a946e4465abbe645c33fb37a6 : SUCCESS in 11m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fe50cfa0c74447ffb8294ea3a499dbca : SUCCESS in 14m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2416e56c0199467fb5b7b408754fccc5 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7045ac98d9d14840a0b8f7a630cf0fe2 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/730da96dac0447c88ef225cf4d71c619 : SUCCESS in 24m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0d84160cfe944ba7914b2688c0435ce4 : SUCCESS in 5m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d6cca5d6972d4842b398c745e3b96e57 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52de014c8616493ea1c2c2badc2a6fac : SUCCESS in 5m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/855955b6e033447fb569dca7d3e4f960 : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2aabee849484491aef2af8cd733ae94 : SUCCESS in 48m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24fd0319d40244cf9b6b427ec72a214b : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":1},{"id":"afd2bbf2b06a80fb8addca2ef64dc6134c78bd77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 00:48:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50764dfc6ac1775713c95183f05ca7f7a12b1fe1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 00:50:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6acaa1945b041eb839d1af4e36730d0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e79e0867bac4194a932964864f36956 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d69fb9df209c9808a8e9e8dea181235507331ace","revisions":{"d69fb9df209c9808a8e9e8dea181235507331ace":{"kind":"REWORK","_number":1,"created":"2023-09-15 13:02:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/895271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/895271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/895271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/895271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/895271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/895271/1"}}},"commit":{"parents":[{"commit":"c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809","subject":"update constraint for python-zaqarclient to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 13:02:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 13:02:46.000000000","tz":0},"subject":"update constraint for python-mistralclient to new release 5.1.0","message":"update constraint for python-mistralclient to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I92515e826281d906ee043291f092b60f978bd9cb\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6efdfd85897f4ef450bd491f40ca6b49e0c50e7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69fb9df209c9808a8e9e8dea181235507331ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69fb9df209c9808a8e9e8dea181235507331ace"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
