)]}'
{"id":"openstack%2Frequirements~895272","triplet_id":"openstack%2Frequirements~master~Ib7f71c33d69f627b4d972b16eb5efa50f9279123","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib7f71c33d69f627b4d972b16eb5efa50f9279123","subject":"update constraint for python-monascaclient to new release 2.8.0","status":"MERGED","created":"2023-09-15 13:06:51.000000000","updated":"2023-09-15 20:21:04.000000000","submitted":"2023-09-15 20:20:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895272-new-release","meta_rev_id":"e1f4551dab2cd6354f23c86d7f47c419d05d91c4","_number":895272,"virtual_id_number":895272,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-15 20:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-09-15 13:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-09-15 16:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-15 13:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-09-15 16:37:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 13:24:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-15 13:24:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-15 14:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 16:37:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"28ce4eaee5794e26bcfd4ca8e8b1fa73b16e6631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-15 13:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27a0cefcbc84d84988116cacc545d76189a3e6f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-15 13:24:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc0cf28784e597db91f16d685f4c1c1df727b4eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-15 13:24:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c55be21449f6cf504fd25bf764a32c75f44e315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:41:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ef173937cdf4ceabbddd58dd1d290fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812800b32025455a8012a98527695c1f : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f324aebc0756470891db763ccfbf3460 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54b9ff1076f34c7886bddb7924e62b59 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48b371113e174f7eaf65aab46d408f5a : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a10989b1bdfc4154af7357a8d06ad3c4 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/608bcb35fbd94b2cb618fc37888cc533 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6f210c2530449f58a544a1d0f6cdb97 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50fe710adaa04459b651b038ebd17f98 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38f344376f694d96aec58d169e18bcae : SUCCESS in 8m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/edafad7500fb4f9c8f893b29e1c17cc2 : SUCCESS in 13m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7bedcd7759e94f6d9dfde8aee51d7b18 : SUCCESS in 10m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2217f2198c1e4bd8bc812136c6f51011 : SUCCESS in 13m 03s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/40b511aee76c49df9f821be16566072f : SUCCESS in 7m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/88244c154a7b41dca4a189543e4378e7 : SUCCESS in 4m 42s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bfee8f284ce04036aa4b5c51c16900b4 : SUCCESS in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9176f302a1874880a2afad017bd655c4 : SUCCESS in 11m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/da5fb32af04b466c86872d2fabc6a5d4 : SUCCESS in 5m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ae5949d74c6c4e19b671358e15f7131b : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9bbd5d8e3a7745dab3d93fc2d85bfdc8 : SUCCESS in 7m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8a513f91b4e740cd83ce41d961e352b6 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cdf5cf59ff04834a9349d28d20a15e5 : SUCCESS in 7m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4d416c9979d84a78b9ba44e4b47ab58b : SUCCESS in 9m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6ff20348c5774ae59a63ae7ccd5b0b2a : SUCCESS in 11m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dba0e56c2c10492ba1f709577d05d616 : SUCCESS in 4m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ee2f3f2664994749b590784f71abc28e : SUCCESS in 9m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/920c6630fc5348aa814155c1ba94b1d4 : SUCCESS in 5m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8bfb568aabc344e08b54de6eda24f75f : SUCCESS in 11m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/58b5c31e266c48e88642a198f44317f8 : SUCCESS in 23m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/db8495a77c04456d96a794a4a018ac59 : SUCCESS in 9m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e7fe5480da7d432b8c2c8f9b395c7bc0 : SUCCESS in 12m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/620263c86ed34deeb0501cc8f94acefd : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1ce0000fcde24f0c9371bd5882a1ae0f : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cfda718e8a7c48b093172fabdced4ef8 : SUCCESS in 26m 40s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6bd430c2b8c84c9ea0c452b0d2b65068 : SUCCESS in 5m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3ee6a4cb387e4eebb7f5e1825c1dd4bc : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdb9401844aa4980822ea008110f6eec : SUCCESS in 6m 16s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1e2a33f61fda4fc08e3bf3cd56988ed3 : SUCCESS in 14m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4cfa1043b234406ea17e7fdc14ebe003 : SUCCESS in 1h 01m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fc3379307e84a13b974efbbf7efb38d : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"644bf2807bd2c9919837e056161e48c5c53a32d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-15 16:37:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08e5e6f298f98efcc439662e2f9b9c3ee8e47f59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 16:38:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73872673a3b294ea28fb97da09db3693c53f6fd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:20:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9762d7a023d348faac43a516f8d03198\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f076dce32b41e89707efb79986844b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e66b62995fa48f4927b099fd9a3552d : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d592603c27d40129ccd3289671082dd : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f0f047b99994b3cbf60343b8498bc9a : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4774d16379744fdfaf5e02db0b53a491 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/294ed30ac3084a27ac6392635583b7fb : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da365ced81754246bf81a85d6eadfb83 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29cb8506aaa44ca0a059940fe3d8e05c : SUCCESS in 8m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0a5ec7b1311a4364a79e31b66e373f22 : SUCCESS in 12m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/20eade44a36b42778ec3df55082f0015 : SUCCESS in 8m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6f9239ee124d42c3b802e36bdf8916f9 : SUCCESS in 10m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b24df5d4ca9641cdbeaf5d186585f5f2 : SUCCESS in 6m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5e252ea8510847fab081eb881446c9ce : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5ea2c4b32c2d489abcecba31fafdac13 : SUCCESS in 8m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/25198f885bac49f598191ace326a3264 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d5fa539312947449351aba4da6745ac : SUCCESS in 7m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/14c68e88d1c04c05beb1cdc0b011f4d5 : SUCCESS in 10m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1632188333a84b7180c986dc87866c89 : SUCCESS in 13m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/be95bc934506451a8a783c46e1fd8503 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4d74f1b50add4d6a8205890bdb60d1c4 : SUCCESS in 11m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e05115f5916c471ca12586f793add55d : SUCCESS in 5m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/84252e7150d945de84b4cef7670f8d27 : SUCCESS in 21m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9afe7b2c520046609eddfbdf5d2dae26 : SUCCESS in 12m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0c1a7b2723c04d98ae1861b6a4077a2a : SUCCESS in 12m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/888567d9eb394e3486b632df1df6f6ac : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6478a76160dc410c92c67c12281485cb : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aecd8974587346e8a63f27705c13f6ce : SUCCESS in 23m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/15441f57fe724f14a1ceade7b97d2ec5 : SUCCESS in 6m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f57607f8a92b49c7bbc53809d09bb6f6 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0c68104aad34ab4be34fe4aed5ed8cb : SUCCESS in 6m 09s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/722184e1e063485bbce6646530d7cf6a : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c07c44fefc6f42629d71a6207359da2c : SUCCESS in 41m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8f1262a2ef14d6188b92ef9f7c662a7 : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a7b3f626e60a4dd8d5ff118663b0c02094214f14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:20:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e1f4551dab2cd6354f23c86d7f47c419d05d91c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:21:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa2a35937a784142a541d752ff96032d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d822a235c7241899ffe57bca353f2e4 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29fddf51465753cb4383858194aedfc59dec0083","revisions":{"29fddf51465753cb4383858194aedfc59dec0083":{"kind":"REWORK","_number":1,"created":"2023-09-15 13:06:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/895272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/895272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/895272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/895272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/895272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/895272/1"}}},"commit":{"parents":[{"commit":"c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809","subject":"update constraint for python-zaqarclient to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 13:06:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 13:06:49.000000000","tz":0},"subject":"update constraint for python-monascaclient to new release 2.8.0","message":"update constraint for python-monascaclient to new release 2.8.0\n\nmeta: version: 2.8.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I184041e5ca4662418d0d0d1b21e3f27ada0c4697\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib7f71c33d69f627b4d972b16eb5efa50f9279123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29fddf51465753cb4383858194aedfc59dec0083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29fddf51465753cb4383858194aedfc59dec0083"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
