)]}'
{"id":"openstack%2Frequirements~895466","triplet_id":"openstack%2Frequirements~master~I9066f70cf9dd29ef1aaa16c4e71dfb475597ba78","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9066f70cf9dd29ef1aaa16c4e71dfb475597ba78","subject":"update constraint for ceilometer to new release 21.0.0.0rc1","status":"MERGED","created":"2023-09-15 14:43:26.000000000","updated":"2023-09-15 20:21:50.000000000","submitted":"2023-09-15 20:20:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895466-new-release","meta_rev_id":"c5460f22b7d2af8ed5566267f022b619d72edadd","_number":895466,"virtual_id_number":895466,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-15 20:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-15 16:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-15 15:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-15 16:37:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 15:07:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-15 16:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 16:37:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c4be644c04f1d521332d5e2b168afec429543d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-15 14:43:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb604f4d90959c8b01f1748bc304da18561f39a8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-15 15:07:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f5b29e42706bbe7edfada72adab7dcd0d7907cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 16:11:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19393a942b354140b75a7be46ac6b4d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a84d747c63e4fc39bf51a684494e4e9 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e07a32ed6849a2843e448dede375b4 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41f68bd4a5e649239b394cdc321fc9b3 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c36c02f27844e1f94e08d04b7ce9a4b : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99b6bed28da748868beae33fd096969a : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cfb3c802b1a4c65888ff11a982c1726 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b94c3f74a34d4b1c9d34647da79f0112 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07b43ff611cd42848ae7ab5380e32a1a : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b0c2f771aa744baeafe5b1676f2c6859 : SUCCESS in 9m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f1bc17fef8a44ae4aa44d6aaa66cc745 : SUCCESS in 14m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2c1571d071ad4c3eb10aae50eec21c4a : SUCCESS in 13m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c7efa759e91242bb9562a4b3e689426c : SUCCESS in 10m 41s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8373c275457244f19e21f467454ddd84 : SUCCESS in 7m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/638737c006424dce960bbad6a45d0255 : SUCCESS in 5m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4311ca7aa2cb4ae1b4575cf6eed7c6b5 : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b13e3b2fb7664fc6b0406116ff8b375e : SUCCESS in 11m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5d042da078a34bc9bc59191b2d86cb4b : SUCCESS in 6m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/08338d51f2734566aea3397d98c88ccf : SUCCESS in 6m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dcff9432e7f84e81be0a8e1abcf08f15 : SUCCESS in 9m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b44a74a12a014dc289e752bb92a97b7e : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e522bd08672245bfa86dbcbc1525d9c8 : SUCCESS in 7m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fa88cebfaf014f42adf3f03943aeb701 : SUCCESS in 8m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/55c207fd431c4508a0de06bfcf2195d9 : SUCCESS in 13m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/af632ca9c6324467a0c80a5b1e64b89d : SUCCESS in 4m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5c8a5f91a7fc4ec0acdf822983ea4292 : SUCCESS in 10m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/273c23ee7f834c13bcd0b024fe4dcc39 : SUCCESS in 5m 05s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/362d8992802f415fb4691443ce5e23fa : SUCCESS in 11m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/08004041576c4211b11daaf0227d282f : SUCCESS in 26m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/03948048a45d4eda93048cf94bc3b009 : SUCCESS in 12m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5b84480ee71a4574b0201220e2b9de4c : SUCCESS in 16m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/61a0d5c9f9fe40ce9a3beead741f965b : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a511cae31764fd2a0d61ff4e5f241a5 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e70aed87047144adbc65cf5acd4530f7 : SUCCESS in 30m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4106c554482a4ee9a75f529d1bc1df43 : SUCCESS in 6m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/19f7808b40eb496f88cd1fe03ad192cc : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce3a3f6cf2334311a9a86ee03451efdc : SUCCESS in 7m 50s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/fcc69f6373ae4af1b94018df4a888c73 : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b166fd7a1e10491fa63f1e049a502512 : SUCCESS in 1h 03m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c43ed3ac8c94845b3cf1c760e815aae : SUCCESS in 1h 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2f2d4955e96db8a0b1435f24df05a2d6e3f0161e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-15 16:37:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71051c0af731e1b5fa33f1cdebcca2da8c2dced2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 16:38:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1e58b1f9942b07a1c7fe26149ad8001326358a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:20:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba3f694879c547fdb2cc4367a00e0127\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25edd0f99e0445758b59bc80223623f5 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc832d56f8fa422b8889a23814e4ad5b : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dfe639d7221429ca73811082be83d70 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e97a3081397045bc8990c7a7e7bb90bb : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b466b31283b45b7b0380ba18c9600ab : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/495d5efe83fd46648d16598e0d909fde : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c89e097b118346c3957ef7420ad77a80 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3f99414f56e8490c96aab5b19d5f3dd5 : SUCCESS in 11m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/70d9a05f63784fb59c0486db0c961ccc : SUCCESS in 14m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8794a074ced6415a86621a52ef01de60 : SUCCESS in 9m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b0128272cd8542ae921bc2e248fc460c : SUCCESS in 9m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2bff2ac81d1041f1b5f33ef9f843dc08 : SUCCESS in 4m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/43e88807f69f40b3af34d5180b189ceb : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d4d2ad8e1c784ee7b0900378e89c9f2b : SUCCESS in 8m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/150f95f8861840bf910a6312aab242ca : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/024a25d6ed7648b7871a17fcbc5d8bc7 : SUCCESS in 7m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7d94e823fd974a8e92fa5d244a932df8 : SUCCESS in 9m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/92d3b0e9017c419c9eeda1d14b29cff1 : SUCCESS in 11m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/37409c49fe494207bc11dde4f28daf06 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cc74a4696fb14e4daf71ee85eab2fea1 : SUCCESS in 12m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/14b5d3e4f65b4f0e837d4d9eca748bd0 : SUCCESS in 4m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ead9ee16251e4cfc9b4e6d809f162841 : SUCCESS in 27m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7735576ae1204f0dad1233a0d82f3bf4 : SUCCESS in 12m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f9e0095dddf04cb593f41ca67b86a318 : SUCCESS in 11m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f5a690a9deac4e34a31ea558d8cad989 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5982cefdf8e0442e9f62372c44a1403d : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c1cf806ab40248e496d290b127d14951 : SUCCESS in 25m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8d200143c0d24261925edb2132640059 : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c5d02e0d274d4201a02b782da9e224aa : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77db2d5d45944770a1e9b0560b2be8ac : SUCCESS in 6m 53s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c555e26a64644e5aa29d77a08a67edcd : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be50733b4a91474f8b5b3dad24fbca60 : SUCCESS in 1h 07m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83477eb3bd5742dda5f4e4ef5deadcab : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f5762ce0e852e768010aaa345c4676088660e899","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:20:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c5460f22b7d2af8ed5566267f022b619d72edadd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:21:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f427f47624dd459cbea835bcaf459af3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0c5e0549c04dae85af103fdb361525 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2f3261b3ffb8fb695926eeb4044d738beba2535","revisions":{"e2f3261b3ffb8fb695926eeb4044d738beba2535":{"kind":"REWORK","_number":1,"created":"2023-09-15 14:43:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/895466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/895466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/895466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/895466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/895466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/895466/1"}}},"commit":{"parents":[{"commit":"c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809","subject":"update constraint for python-zaqarclient to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 14:43:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 14:43:25.000000000","tz":0},"subject":"update constraint for ceilometer to new release 21.0.0.0rc1","message":"update constraint for ceilometer to new release 21.0.0.0rc1\n\nmeta: version: 21.0.0.0rc1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release candidate\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5b7c41b2e0244108ecbe2dc3d017ded2f6b5e99b\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9066f70cf9dd29ef1aaa16c4e71dfb475597ba78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2f3261b3ffb8fb695926eeb4044d738beba2535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2f3261b3ffb8fb695926eeb4044d738beba2535"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
