)]}'
{"id":"openstack%2Frequirements~895490","triplet_id":"openstack%2Frequirements~master~I22f03a91b2b61f42dfc6d60a4f00cbd53c857499","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I22f03a91b2b61f42dfc6d60a4f00cbd53c857499","subject":"update constraint for python-openstackclient to new release 6.3.0","status":"MERGED","created":"2023-09-15 16:10:27.000000000","updated":"2023-09-15 21:45:49.000000000","submitted":"2023-09-15 21:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895490-new-release","meta_rev_id":"579f3c14f3cfefec9e5cc08698cc0dee2563f431","_number":895490,"virtual_id_number":895490,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-15 21:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-15 16:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-15 16:37:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 16:37:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-15 18:02:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 19:06:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"2ab6e75918bac71a2651be9ae55522c87db83bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-15 16:10:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3007dbd8b2a2a89df104b0686d9e9e7b17e6a06","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-15 16:37:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"054ea2cd6a3359472e079371442b076e556a306a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 18:02:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccd2b8126700424f97ec6014360c2c98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32df2616c4a0483d8d491a59f11ca5ff : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731cae6912fd4ebdb19c18fc28152939 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71fca130d191411d9032d9c5a5221640 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efa8655591614f85b482cb44df3191e7 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/919a7f41dc2b4343a5af1e494eba2c19 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/061c15a17cdc49c7a5c69f6b9bd21291 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14b1c112501f4fda847e8113eab78a19 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2a1ff291be9412790beb62435b7685b : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2fa75ee3e0564b268879964c553794e4 : SUCCESS in 10m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/768189fef58b4937a9b2efc781368952 : FAILURE in 4m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8737befdf580495eb3ff49bbf36058d1 : SUCCESS in 9m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c05425722c05450994079d939cce984e : SUCCESS in 12m 46s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d3fb5aba5910457b8c6772161f838f1d : SUCCESS in 8m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/183c33f691c34a25b398ba3f3f94cc11 : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8c390347c3b948f5bbb7995a3edbda5b : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ef6ebcd2952d4a86852920c5380b4731 : SUCCESS in 12m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/49e61ad2b49141e18862f88fbc35f22b : SUCCESS in 5m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/05ed303d44864582820aedc3f2626310 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/337583f5d3ba4590a265269a2abbb28d : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e707e09060544d909b2cc8735c718839 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1149bb8bd1741e3911097a303b52789 : SUCCESS in 6m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9778458cbd8246129e9d61bbce82eb77 : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fd395e7f4ad6467d90a4fb0a17f4bf88 : SUCCESS in 11m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e13a8f8e815345c880278dfa2ef912ca : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c55def30f4c844bf95bf59e4d3751662 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5b76c7d89cac46cda938214dcff58be4 : SUCCESS in 3m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1dd6283ec4fc4ca687561badc4626298 : SUCCESS in 9m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0e9cf35d81584483a398d389d19c77e7 : SUCCESS in 31m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/46d5e0ab1e3f4a0abd2d579f369e5d5d : SUCCESS in 11m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0fdbc01b987149a695614e2d5005ab64 : SUCCESS in 13m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9876115bc8aa404fad644b1a6b05ec7a : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58aad9c0137049059728d7ad6f9ce686 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/072be96fce1640e69828caff98349633 : SUCCESS in 24m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fd2d5436dc664eb3a994518ce804c484 : SUCCESS in 5m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b3b5ab17236d481791b92b9814443ece : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af6d5cff00f44c45be2e6ecbfb32a229 : SUCCESS in 7m 10s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d6977bed21294e9f8e406bf0579a3669 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55fe377f760a4ac4a02b19a7034d000a : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0d472ccd85c4d22b183bc5a617b762f : SUCCESS in 1h 43m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ee1af00a18905a97c5c13e92095fb54f7cb8dd8c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-15 19:06:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3a02a33973abdb81e1d6a8ecc939f042e3fa099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:35:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8bdcdac3939405f9b7c4d342eee76a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88fd9c5baaf448b1bac387c1e9edcf5d : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de8b2f181b51476191a8c9bf87ef8de8 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f2406c8d14146e08f1e2406f461c085 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54835723aec94536a69404e00f0d4900 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c17ce101cc89475e97d9ad5bfe463cf5 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/793441f46f0e418d9b1153fe16802ef4 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26e87ae2a0bb4bf1aef2ee9e5936d2fb : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ea9579e823e4132ac8a705805bd0f91 : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd9a56fa7acd4aea916e03bee2a672bd : SUCCESS in 10m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/790752e5607f48f9965eb951825af056 : SUCCESS in 12m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f5d32c24e8764db0879e60d0b3131192 : SUCCESS in 7m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/19d7b7c2deb245408549d0d3fa7431bd : SUCCESS in 13m 26s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ad71949dd3b645a58ed580b47a9ccdb2 : SUCCESS in 6m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6bb13c673dd74004b81ed74a94c721bb : SUCCESS in 4m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ac9cd88254de46f390f6d46d8c4cd1bd : SUCCESS in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2c9e3b8e62184a5190d524b621ec959d : SUCCESS in 12m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/279c6fd872fb4604930f70ca3c2a6205 : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2be1b3305c514110ab286850f9a832a7 : SUCCESS in 7m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e340a8ff499d47babbc0687311de040c : SUCCESS in 8m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f99fd590dda5475baa50a80ee9d16b52 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0c21282fd864de59a2f030b9854ca51 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/705d9830725041bf8df71b23ac6c5cb0 : SUCCESS in 7m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/54934e4324fa438d863a253143b58ed8 : SUCCESS in 9m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/158813425d1e419cb1b64fd3d2d5dc22 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bcd6d5987c454b28a48fcf88041b8656 : SUCCESS in 11m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1b3eacf3ca0f43f0a2cbec1566a0b97b : SUCCESS in 4m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/05cdfa7801c048fe80f638fa9ed04d7a : SUCCESS in 10m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1458917147604dbd8814da2b2318b713 : SUCCESS in 24m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7cbee7837b0444b2928e5aa597f7bc56 : SUCCESS in 11m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5a536c1971134392a3cf6bf2351eb14e : SUCCESS in 13m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fccdee55f5b945fc90aa613f59ca0b63 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/94c3f14583924e188aeb15e81d11585e : SUCCESS in 5m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/522b4642b657423293b6a608832f692f : SUCCESS in 23m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b987f2d534074c398987a946374c359f : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6ed35d29216e4938a7f01520822c9f31 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57dfecf9edee4bbfa46ef59c76624cef : SUCCESS in 5m 21s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b74d6e2259b34d9f9f063ea3627aacd7 : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3f442f6ca49411390990ac57a51091c : SUCCESS in 44m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee4746d6c8f3481b98e389797674073b : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f1c81ad26502c1d8b0a595c859dd6ce2ed4239b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:35:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"863c6ffc56f129d93da9278419bbfbced979251d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 21:44:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c08d8fbf7d4d46d3b609c04da7dfa3f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2def8764524439aca343e21fb14488 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fb3409f7594738aeec28515f9c17ef : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14d636d4f4fd43bcaea1bf5f5ddf2a0d : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a94fb335df8e45d399da8ff47638a1ab : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/649daba1f93b41328f632bf99aa7a2bb : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea43f60774464a4491cceb71220600ac : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23646dca67f249cfabe9e7ea935346a3 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aacbfcd33fdc4c2b8eee2c4ba2021df7 : SUCCESS in 11m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/22f243c428a8445198f11ea08e5e283a : SUCCESS in 13m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/78516796478149a198145789e0adde9d : SUCCESS in 9m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/64af46073051445b99f936438967a847 : SUCCESS in 10m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b9e3565b335140ca9b0f901959467907 : SUCCESS in 5m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/77c6f478c92848e9bbb539c553068f67 : SUCCESS in 4m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/565f153314464b559496c658a2c6036f : SUCCESS in 8m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3b764a2ceb364a39927450ecf2222b49 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d67475c55644cbda2c4407508ed448e : SUCCESS in 7m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6f62fb31fd164642997742865f696afd : SUCCESS in 8m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/df7fde866ae84fb790da513a0e61041f : SUCCESS in 10m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/484f79550ffb4203abd146da499bb49b : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e7f0769b103f472c8ff769b1e0326fe6 : SUCCESS in 10m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d1cee829ca5841bb947ce1de28464ee5 : SUCCESS in 4m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3636343419804f6091ba9a4bb6830a43 : SUCCESS in 19m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f45817efba6b467ea2b0b568a516b96b : SUCCESS in 11m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d9ae264472f0434aa196cd65f16f2f05 : SUCCESS in 14m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/79863a8e28584bb6bcea549220663d29 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b68e4833ead4a2e9166682cdd0f7e70 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97090d1fb8cd4936a45a08368d72a79c : SUCCESS in 19m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/16bb8ead5b72436cbf379afd88b3a697 : SUCCESS in 7m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/274db37c0e48431abe6bd69026e050d0 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db70574605074b54bc8a4cbc0cec5461 : SUCCESS in 7m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6c37b77231ea46ca9aa19fb3eb52cd77 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c53f537b4fd64ec08c614c701eb9a2e2 : SUCCESS in 1h 02m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a1b435d6d374978abba1d0031b0fe78 : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0de5012d6ed7ffce5416342ad62e4f6150d45c36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 21:44:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"579f3c14f3cfefec9e5cc08698cc0dee2563f431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 21:45:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e7b7afd4fa40af80efd32a06ea4fed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f847ad0e0334d1593dc4a14cc8b64bc : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"073a41aa78603cb96e6c88b72ad5db570cf4a97b","revisions":{"073a41aa78603cb96e6c88b72ad5db570cf4a97b":{"kind":"REWORK","_number":1,"created":"2023-09-15 16:10:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/895490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/895490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/895490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/895490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/895490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/895490/1"}}},"commit":{"parents":[{"commit":"c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809","subject":"update constraint for python-zaqarclient to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c30bc3e37bfbf6ef0d77ad2b5516d00d50a9a809"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 16:10:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-15 16:10:26.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 6.3.0","message":"update constraint for python-openstackclient to new release 6.3.0\n\nmeta: version: 6.3.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5f9c6bbe5a509a8e3cb8cf7b74a320f663d2eb82\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I22f03a91b2b61f42dfc6d60a4f00cbd53c857499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/073a41aa78603cb96e6c88b72ad5db570cf4a97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/073a41aa78603cb96e6c88b72ad5db570cf4a97b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
