)]}'
{"id":"openstack%2Frequirements~895732","triplet_id":"openstack%2Frequirements~stable%2Fstein~Id47c97bb75ed815f3512b51a8db0e48df90b3584","project":"openstack/requirements","branch":"stable/stein","topic":"rocky-eol","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-09-19 06:11:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id47c97bb75ed815f3512b51a8db0e48df90b3584","subject":"Drop cross-osc-build-sphinx-docs from stein","status":"MERGED","created":"2023-09-18 19:17:34.000000000","updated":"2023-09-19 06:11:07.000000000","submitted":"2023-09-19 06:11:07.000000000","submitter":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895732-rocky-eol","meta_rev_id":"04127d7191e96dc7c3f9dae66f699e018701b58a","_number":895732,"virtual_id_number":895732,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-18 20:03:50.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-19 06:11:06.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-09-18 19:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-19 06:11:06.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-19 06:11:06.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 19:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-18 19:39:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-19 06:11:06.000000000","updated_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"reviewer":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"state":"REVIEWER"}],"messages":[{"id":"6e0d49823fc5154eb60b81d8741f4afd49ca86ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-18 19:17:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"654ef269c4c0e422790a21e7bb55df4ad68bf059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:17:57.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch stable/stein.  The error was:\n\n  Job cross-osc-build-sphinx-docs not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - openstack-python36-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              voting: false\n          - requirements-tox-py27-check-uc\n          - requirements-tox-py35-check-uc\n          - requirements-tox-py36-check-uc:\n              voting: false\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-cinder-py27\n          - cross-cinder-py35\n          - cross-glance-py27\n          - cross-glance-py35\n          - cross-horizon-py27\n          - cross-horizon-py35\n          - cross-horizon-npm\n          - cross-keystone-py27\n          - cross-keystone-py35\n          - cross-neutron-py27\n          - cross-neutron-py35\n          - cross-nova-py27\n          - cross-nova-py35\n          - cross-nova-functional\n          - cross-osc-build-sphinx-docs\n          - cross-swift-py27\n          - cross-swift-py35\n          - openstack-tox-py36\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - tempest-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      gate:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py27-check-uc\n          - requirements-tox-py35-check-uc\n          - cross-cinder-py27\n          - cross-cinder-py35\n          - cross-glance-py27\n          - cross-glance-py35\n          - cross-horizon-py27\n          - cross-horizon-py35\n          - cross-horizon-npm\n          - cross-keystone-py27\n          - cross-keystone-py35\n          - cross-neutron-py27\n          - cross-neutron-py35\n          - cross-nova-py27\n          - cross-nova-py35:\n              files: ^upper-constraints.txt$\n          - cross-nova-functional\n          - cross-osc-build-sphinx-docs\n          - cross-swift-py27\n          - cross-swift-py35\n          - openstack-tox-py36\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - tempest-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - tempest-full-py3-opensuse15:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@stable/stein\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"372894dfd80d2ce0454c63ed3f644e2b81d6033b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-18 19:21:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ed207e58bffa3eb0b0061bf013a9a55a9f47935b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-18 19:24:21.000000000","message":"Topic changed from heads/rocky-eol to rocky-eol","accounts_in_message":[],"_revision_number":2},{"id":"060143e268041654c6b04e6f0301050308fa1882","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-18 19:39:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e3ab28f8d11963fc891b3f4e52c6d6d5e3df0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 20:03:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2b1a2b61f6340a0a898fa1368bd76ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952829e2ce5f4543bfe73c4a92f431e7 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4bea1fd8194f4d1cb2204f1483b4b3bd : SUCCESS in 3m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4cb1e2a04a064535bc7e7cf6e414d19b : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f419fd94e27f4666be01503906346380 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47f018e4aa4f443c8aed3fe735502ddf : SUCCESS in 3m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e02c1cc911ff4a7b91132e92fd08b9cc : FAILURE in 2m 57s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/0514f945a19c4cb1a8f7637552de2a0e : FAILURE in 4m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6ce772b6932b4b6299724308156bcb92 : FAILURE in 2m 52s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/1e01520d961d40aaaf471871c93138f5 : SUCCESS in 5m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/07a87295989c44dbb1b6a47d09d1b4b4 : FAILURE in 2m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/40f2438d78514cb5913f25a1959b7dd2 : FAILURE in 2m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67d06586ee9e460b8bb42e52b79cf491 : FAILURE in 5m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f2496690ca10406f80e84b9d4481cdf7 : SUCCESS in 21m 49s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/25576c9c500b404da9b77459d3a032f4 : SUCCESS in 40m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1504eefe389e4345a5d92dc86bf5b486 : FAILURE in 3m 54s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7e574d95c81e4818bc882a29b09cbc5c : FAILURE in 3m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/95fab34a2a3b4cdebee48549312a5efd : FAILURE in 3m 54s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/91a08cbb59d24dddb48dfc892c16da30 : FAILURE in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d77002ddbb445e0988d45fc650bb116 : FAILURE in 4m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac44c02831204ad9b184daf87c5f5b5f : SUCCESS in 7m 55s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/27eee6b986524320aed145783f30dd75 : SUCCESS in 10m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/697b1dc1d1da487781a98323d68c79a5 : FAILURE in 13m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fb80d1c3a0836e6060c9ec87db36c35c776b027c","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2023-09-19 06:11:06.000000000","message":"Patch Set 2: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this and unblock rocky-eol cleanup.","accounts_in_message":[],"_revision_number":2},{"id":"04127d7191e96dc7c3f9dae66f699e018701b58a","tag":"autogenerated:gerrit:merged","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2023-09-19 06:11:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ca8e28da38bbbc9f66c683f2ad71f34cdc6337a","revisions":{"79c5a4fb67027e4de5448a27bfdf7daf655f3295":{"kind":"REWORK","_number":1,"created":"2023-09-18 19:17:34.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/32/895732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/895732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/895732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/895732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/895732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/895732/1"}}},"commit":{"parents":[{"commit":"3a3ed8a7d36c2321a519fd6b54b7b328274e4d07","subject":"[stable-only] Use bionic for requirements-check job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a3ed8a7d36c2321a519fd6b54b7b328274e4d07"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-09-18 18:44:20.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-09-18 19:15:49.000000000","tz":0},"subject":"Drop cross-osc-build-sphinx-docs from stein","message":"Drop cross-osc-build-sphinx-docs from stein\n\nThis job is parented to build-openstack-sphinx-docs which is only\ndefined in openstack-zuul-jobs as being relevant to the stable/rocky\nbranch, and that branch is now EOL so we need to remove this in\norder to be able to clean up the job definition there.\n\nChange-Id: Id47c97bb75ed815f3512b51a8db0e48df90b3584\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79c5a4fb67027e4de5448a27bfdf7daf655f3295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79c5a4fb67027e4de5448a27bfdf7daf655f3295"}]},"branch":"refs/heads/stable/stein"},"8ca8e28da38bbbc9f66c683f2ad71f34cdc6337a":{"kind":"REWORK","_number":2,"created":"2023-09-18 19:21:23.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/32/895732/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/895732/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/895732/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/895732/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/895732/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/895732/2"}}},"commit":{"parents":[{"commit":"3a3ed8a7d36c2321a519fd6b54b7b328274e4d07","subject":"[stable-only] Use bionic for requirements-check job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a3ed8a7d36c2321a519fd6b54b7b328274e4d07"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-09-18 18:44:20.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-09-18 19:21:18.000000000","tz":0},"subject":"Drop cross-osc-build-sphinx-docs from stein","message":"Drop cross-osc-build-sphinx-docs from stein\n\nThis job is parented to build-openstack-sphinx-docs which is only\ndefined in openstack-zuul-jobs as being relevant to the stable/rocky\nbranch, and that branch is now EOL so we need to remove this in\norder to be able to clean up the job definition there.\n\nChange-Id: Id47c97bb75ed815f3512b51a8db0e48df90b3584\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ca8e28da38bbbc9f66c683f2ad71f34cdc6337a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ca8e28da38bbbc9f66c683f2ad71f34cdc6337a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
