)]}'
{"id":"openstack%2Frequirements~895900","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I9ca3b2fae48f6b4dc090e464dcd9073f944af6cd","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I9ca3b2fae48f6b4dc090e464dcd9073f944af6cd","subject":"update constraint for python-octaviaclient to new release 2.6.0","status":"MERGED","created":"2023-09-20 09:20:44.000000000","updated":"2023-09-22 19:32:44.000000000","submitted":"2023-09-22 19:32:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895900-new-release","meta_rev_id":"7e87c3e793368fab06da8f97df5f5f33141def16","_number":895900,"virtual_id_number":895900,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 19:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-09-20 13:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-09-22 17:36:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-20 15:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-09-22 17:36:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-20 10:34:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-20 10:49:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 13:40:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-20 15:43:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-22 17:36:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4e53c887200a0072cc89b409366a6b97dd33d7f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-20 09:20:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f0e3ba54b0009c4994266997016b550a6aff8b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 10:34:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3afeeec546b94bb78f6baa429d59acf8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/42747b7c5c7f4617948c1ecbe0ac3392 : FAILURE in 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7d243a58b6e1f511b7d5c6d9a4a5514715476eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 10:49:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d75f2c52ba74a69b9e3a1cca69655b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47bb6fed285b43a9b0a58996740a9224 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe6b2ac3300142f583d6c04408fae38e : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4e2fd4460d4c98addef046f1ece0a5 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b4c928740124b269a1a4a55473c6eb4 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdbaa91b0ef14dd08b7cd9d5d0d4490b : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/154b2cfc2b674122aa9b91695add2027 : SUCCESS in 22m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ff0605d15dc4fcdb87a43f74579a7a3 : SUCCESS in 13m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c77bde05274841c2ab7ab3166df746fa : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/643c5479d6d2452ebf226feb82db8947 : SUCCESS in 10m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/48e51880e3f340cd9f24719a5243cfd3 : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/af550d215d6e40af83e8cc36bd26b9cf : SUCCESS in 5m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d5a9e049633a473c834f653209bda295 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e059d8a84a8c492e94e492c5165d89d8 : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3c64ef79a93d47d2b8c7c612bfd40114 : SUCCESS in 5m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6ff6100931e24003840637ae6dbe0d2b : SUCCESS in 10m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5aabbb220e4640ca938af7f50fc87138 : SUCCESS in 3m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5dd5c40cd8ff4c23a59aec47d6473f9a : FAILURE in 17m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/67f72c118a8848208f71ee0ae9224324 : SUCCESS in 5m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ad3087eac2f54cf6bce90c4feed95fb1 : SUCCESS in 23m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a641db0186ff4b70b4d15466e814e452 : SUCCESS in 13m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dfe48ef2d93f489e9dfaa4a7ed2207bf : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2fda1485bcc045b29aad1ddd92dce48d : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2bff092b21844fe85e0a962d70a36fa : SUCCESS in 23m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9f50f5eb070f48c590cbb7e744844617 : SUCCESS in 8m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c0f4219092864316b328ceb82a80d057 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78d040c48f114e6db11ad0672814d3c5 : SUCCESS in 9m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d3aa328a471b4db4a7eca6d9fd6eae2d : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc579622be1d4d9881d161418c92ec5a : SUCCESS in 56m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc8560157f7b4b6abda3a2cf8f852a88 : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"042f2fc2211b6e463527fc575d032281b7e2dc50","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-20 13:40:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bc16040479f7159dbd0aa38df4910d40929f16d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 13:58:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e0a8738df545999cca87fd40f0dffc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e402615cea7d43b6b9fdfbbb5b85e719 : FAILURE in 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a061ec24838ebd9cabc48bddac629555820525e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:27:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac4fe0d6c1354697832fe2d748c248c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a58df5cb3ff432b98e389d96c67fea1 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e2e51eb99ae4911a49fb945f8996bee : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9c9c0b9feab4e158605082e605fef8e : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca6b34019b4a4318b1f02ab388933dc6 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5abb64884d540f699c636dd7ca0659b : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4759eea29d0d45d1a25f0c7fe5bab43a : SUCCESS in 20m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35be9c0db62749c0a88290d8893210fe : SUCCESS in 11m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ebde867d4ea44991b59339ff2250c774 : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b261363b5c484a1b9b14b4bd956114bb : SUCCESS in 11m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/96844730ab1044adadc7d70e83607ea8 : SUCCESS in 4m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f2c84e08fe8a41f892e6b9e9816b39d5 : SUCCESS in 5m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cf112b29499b4310aebbf1ac1767d512 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e15d0ae763e54cc5a9666ea43c65fb7f : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6362374efd4645c1b11c03203f8baa75 : SUCCESS in 6m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/097dfa2649c846e7872ca4ebb482f372 : SUCCESS in 15m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ab9605e977e2466eb9fd32cabbf984e8 : SUCCESS in 3m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2f0c7ed71c0b482092060037880daacd : SUCCESS in 15m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/709c7faaa91746fd8035033884dca670 : SUCCESS in 5m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/665c9d9dc332422cb63cbf8ac7dd4670 : SUCCESS in 20m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0e3f1fd31f3843ed865abe772bfda503 : SUCCESS in 12m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7c484219a611422baab8f1de945e8131 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/13b7a586150e4aa6a2e81574083a6920 : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93ff5362de434ad4bb34939f006cf455 : SUCCESS in 21m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2695ce65ab424df5b5f7d5c3de2cea18 : SUCCESS in 6m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/df1cd240689a4dcaa8de77a5cacda2ea : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4bb9a80099b0406abd03a2eba05a79f3 : SUCCESS in 8m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0d9a6e2dd0c7432599f7698b6ca209b1 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf9ea7ece80a4f09b7efda9dff1a4017 : SUCCESS in 1h 03m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d95abeb6ea8a4ae792e2a3f027d0418f : SUCCESS in 1h 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7086dcf450b80e14f0b2462f43a0b0fd7b07081f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-20 15:43:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e44600cd75db3a8fdeb124275dc2daab72f4e376","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 17:36:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d36671ecbb437c2131e5c75f6e02270bd71827d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10da539642bfa6c87c1e88e7038e72f3e31edc0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:32:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07f309b45df4c6cb3ed269af41bcd0d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b43d0765f64549a5c38136221dd6c9 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/085251e825ec4c338c4ba42c1b69faf1 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be592ad06b784d058d5b0210536a2fd8 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89f4538281fc4be5afb25c381ea8f4ad : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84fb63bbbfb34090aab9cfdf00144810 : SUCCESS in 3m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a89bda9a5674291ad7b089d04fc6746 : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57b5ad6dc47943d4b5bfb4c20a1f6a36 : SUCCESS in 10m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/63eea35356ad4dc0adaf4779eb0fc6fe : SUCCESS in 11m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/858671337a5048f09f2b612f5cf440f6 : SUCCESS in 4m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/41a3087b14ed4314b319bf1b466b8f8b : SUCCESS in 5m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1863171ec917430395cef19ba92c771a : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57c8c6318fb94c58a6b2a654fad68cef : SUCCESS in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c3bed45521794f7986df335569460940 : SUCCESS in 7m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/609a6c7a19c94ba7bae7fb486a6633d4 : SUCCESS in 11m 33s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a8c91787ddf24ccfb1654abbd0d50ef9 : SUCCESS in 3m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f76c3286a73b43abba54d550181c5239 : SUCCESS in 9m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b60f18defe4f4549a359f21d31897361 : SUCCESS in 3m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/111fcae13a494a8d803e88e01a95e83e : SUCCESS in 23m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8f57d639772d43afbb297ba9d12c1c1d : SUCCESS in 12m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/11e68a43dedc4358950a9f9c49ce7ebf : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/288388fa512c4d3a8c09a96b84cef966 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c3740804ba049f0a9910c09367f59fe : SUCCESS in 20m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ab9b52084cb14039993b8b408622e8ac : SUCCESS in 5m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4383d5f3ab3249dd9b24d336d68fc3f4 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c84d5bff67694145ad9723af3b590b2d : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/61e2cb31811a4128af90d76e7d5fecb5 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/824d4dc957b746069186659a90be70ac : SUCCESS in 50m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58123ff8fc9f4957b3a5326f7e1d4205 : SUCCESS in 1h 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7e87c3e793368fab06da8f97df5f5f33141def16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:32:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f68f16da1e676621f0b74cfbd2af7e10070dcc3d","revisions":{"f68f16da1e676621f0b74cfbd2af7e10070dcc3d":{"kind":"REWORK","_number":1,"created":"2023-09-20 09:20:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/895900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/895900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/895900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/895900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/895900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/895900/1"}}},"commit":{"parents":[{"commit":"9b0af2c04ee08d4a5463d9aec04c92092a03f69e","subject":"update constraint for os-brick to new release 5.2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b0af2c04ee08d4a5463d9aec04c92092a03f69e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-20 09:20:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-20 09:20:36.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 2.6.0","message":"update constraint for python-octaviaclient to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I8c58eb2063bee5f85231b41fe6c21b6c6c403def\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I9ca3b2fae48f6b4dc090e464dcd9073f944af6cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f68f16da1e676621f0b74cfbd2af7e10070dcc3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f68f16da1e676621f0b74cfbd2af7e10070dcc3d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
